)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~736876","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I60ed1a2415b4d145c3d4fc4207c30b384291a0e0","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"I60ed1a2415b4d145c3d4fc4207c30b384291a0e0","subject":"Add Centos-8 support","status":"MERGED","created":"2020-06-19 01:56:26.000000000","updated":"2020-07-21 17:29:01.000000000","submitted":"2020-07-04 10:23:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736876-1593858202924-ea166436","meta_rev_id":"5faff7c1cc57999f50050084c67dbc382b0e3775","_number":736876,"virtual_id_number":736876,"owner":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-04 10:23:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-07-02 14:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-03 22:17:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-03 22:17:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 14:07:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-07-03 22:17:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-04 10:23:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb13c359dc0aba5080fe8625a7f3adb3fff96aaa","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-06-19 01:56:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aabe49148fa72f8bcca75f5b830b7a78d0cc3ca8","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-06-19 02:01:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc2df630565fd957cdf4d0de1d2579754fe17a3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:06:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/19cbb8ff93384094b310735ba728430b : SUCCESS in 9m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/274ca011b41d46ffbc3271b946ea7b4c : SUCCESS in 1h 24m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff8eb07949094426a5fa1b81a990b634 : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0c8a4a4e754846e19ff953f3aa51c547 : SUCCESS in 1h 25m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9235109e82ba4b9d94837715bbad7f53 : SUCCESS in 1h 33m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff82c7bb38f143b0b95d42f4032ddd19 : SUCCESS in 1h 13m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aef3904907544d22b84ca35099703c33 : FAILURE in 1h 41m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8b99b71f19044f79acf8d956a5760e5 : SUCCESS in 2h 04m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c75197b79a4a4085c696617fc1c6d2 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"692c915533e932f33526f0f3d91119e0b5ba8f66","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-19 19:48:14.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"d344fa96846f1b493daba956d1c53ea218c47f57","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-02 14:07:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4a1d4fdd183fccac516af801eb78ab03341c8f5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-02 14:07:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9e468f46abd833a2f0d35a11ca988c6924672b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 16:18:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b26d7d9152d748a782f999840deec810 : SUCCESS in 8m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fadb73ceae4144b48b64d5e371c86b80 : SUCCESS in 50m 54s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3adcb867f5014355bfc728c6c8f11b65 : SUCCESS in 1h 01m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6291908b285949caa01d3fe03f0fb152 : SUCCESS in 53m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7f063b3951cb423d840a192ea085c353 : SUCCESS in 2h 05m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6875ac080b0480aaa8b2194a23c7dfe : SUCCESS in 1h 18m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab799a0986e94ac584daf90b2a0bc0fe : SUCCESS in 1h 45m 33s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5b81c150f31344a7945a466fefa6b403 : FAILURE in 1h 36m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24ed2737bc9240a28d719e95a77e700e : SUCCESS in 2h 01m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f948c9a34b4724a4300b733068c4e0 : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b455163cdffc0ccdc64ede2219d6a7d50a545ea7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-03 22:17:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"28f273dff3120bd962155869c5ef82d08f95d5d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 22:17:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"87369da3c657768abfe2a6dd432768f4532c4efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 00:07:39.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7a2a6b83b51b4d67a45b6d3b64eb5076 : SUCCESS in 7m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3ec7783fc2d04549bd56efbfafa05b2b : SUCCESS in 1h 05m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c08f014022f041f08ddd1d67efeebd70 : SUCCESS in 1h 02m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b491db88bc85452e9648d7eb7b1a8d0f : FAILURE in 1h 24m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/01d3248467bb43f79bd640477b8a99a5 : SUCCESS in 1h 46m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19ff5be692234db0a97a896afcd8a468 : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/118aa80466ba460faa54157838ba5e27 : SUCCESS in 1h 17m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff46bdd242c74c388455012bd63c0ec6 : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":3},{"id":"cb3c5ec7fe80657bbd6e9c75196ed65fb1c204c4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-04 06:54:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3ee625ede6dd5ea5795ba60ddfc34a524ba2bbd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 09:22:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/086d5cc3187e4d9abb00d1bfd03b0dec : SUCCESS in 8m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/794ccfa87ced464eb03ae2a9553822cb : SUCCESS in 1h 13m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/742f2cc702814c8299346177c72681b9 : SUCCESS in 1h 11m 21s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c5c7600a95e844d985212c16dad81ca9 : SUCCESS in 1h 26m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8f1e4e7da7d04e43806e3eac17f05e73 : SUCCESS in 1h 36m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09bb334162a34d25a9abd4fff49aee1a : SUCCESS in 1h 13m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6e94ad10f664097b9a8ecb33f5140b2 : SUCCESS in 1h 30m 17s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f9ae5ac3e65e4f8982d26187a6ea9657 : SUCCESS in 2h 21m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90e406f777fd49a19ac25e3ca2487eb7 : SUCCESS in 1h 59m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2584a14400e442abab895429d17600f3 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":3},{"id":"dbd043cea5fd94adf666229d217d47012069d4c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 09:23:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"69144387551bd2777c94cbb9f15d8d1eab2342f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 10:23:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e7c507310c2a1732318334b3b53541f7bf671934","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 10:23:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1696e5a4eb9242eb816389103753969d : SUCCESS in 8m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a84a274ec48a40eeb22219c310354bf1 : SUCCESS in 44m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c356b61ae1214581a31cecdebbdf2b85 : SUCCESS in 43m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/702eec6cbdc7484fb95eecde80e68b7b : SUCCESS in 48m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e16cb98bcbf34485872dd790c65ea78d : SUCCESS in 55m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b014cce996614e219f7c956ecaf4ae1b : SUCCESS in 53m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9d2df6cc934440f9553f40d84290551 : SUCCESS in 47m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aefbf86d92e4017abe43485824641ca : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":3},{"id":"35a8c0e031e45edce35274267811dfcc0902a3ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 10:24:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc2351cdada4980ae2e950136bc1e3f : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3},{"id":"bffc4f355f714e63b7b59e073577eeadc8744b39","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-21 17:29:01.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 14a1e2d8225ca8047d39d13e59e4290635e0c49c","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"18c54d70648547a758395180e244f4c4b5084deb","revisions":{"03de2ab876f81d9167df02cd4c0ac7fb01dda303":{"kind":"REWORK","_number":1,"created":"2020-06-19 01:56:26.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/76/736876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/76/736876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/1"}}},"commit":{"parents":[{"commit":"dc0a9588b12bbf1e9d6427c0d6e4bd1b2d1aef47","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/dc0a9588b12bbf1e9d6427c0d6e4bd1b2d1aef47"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-06-19 01:55:07.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-06-19 01:55:07.000000000","tz":-240},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\ncreating redhat-8.yml file for c8 support.\n\nChange-Id: I60ed1a2415b4d145c3d4fc4207c30b384291a0e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/03de2ab876f81d9167df02cd4c0ac7fb01dda303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/03de2ab876f81d9167df02cd4c0ac7fb01dda303"}]},"branch":"refs/heads/master"},"3c98f7997cce2510ee5b84cf39d4352558e06f48":{"kind":"REWORK","_number":2,"created":"2020-06-19 02:01:25.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/76/736876/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/76/736876/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/2"}}},"commit":{"parents":[{"commit":"dc0a9588b12bbf1e9d6427c0d6e4bd1b2d1aef47","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/dc0a9588b12bbf1e9d6427c0d6e4bd1b2d1aef47"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-06-19 01:55:07.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-06-19 02:00:53.000000000","tz":-240},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\ncreating redhat-8.yml file for c8 support.\n\nChange-Id: I60ed1a2415b4d145c3d4fc4207c30b384291a0e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/3c98f7997cce2510ee5b84cf39d4352558e06f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/3c98f7997cce2510ee5b84cf39d4352558e06f48"}]},"branch":"refs/heads/master"},"18c54d70648547a758395180e244f4c4b5084deb":{"kind":"REWORK","_number":3,"created":"2020-07-02 14:07:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/76/736876/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/76/736876/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/76/736876/3"}}},"commit":{"parents":[{"commit":"dc0a9588b12bbf1e9d6427c0d6e4bd1b2d1aef47","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/dc0a9588b12bbf1e9d6427c0d6e4bd1b2d1aef47"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-06-19 01:55:07.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-07-02 14:06:51.000000000","tz":180},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\ncreating redhat-8.yml file for c8 support.\n\nChange-Id: I60ed1a2415b4d145c3d4fc4207c30b384291a0e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/18c54d70648547a758395180e244f4c4b5084deb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/18c54d70648547a758395180e244f4c4b5084deb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
