)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~744149","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I9490820bf344fc387a15020d0c5dedf08ab925f1","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"copy-policy-json-in-place","hashtags":[],"change_id":"I9490820bf344fc387a15020d0c5dedf08ab925f1","subject":"Adds handler for copying policy.json to the right place","status":"ABANDONED","created":"2020-07-31 12:59:07.000000000","updated":"2022-11-25 17:23:44.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d72943c16860bd57e2714aa9186f61c0cb3b0047","_number":744149,"virtual_id_number":744149,"owner":{"_account_id":30660,"name":"Jan Marquardt","email":"jm@artfiles.de","username":"janno"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-21 17:57:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 15:01:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-31 16:51:22.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"8cd3004cc524909f1764e11879671f6f277a093b","author":{"_account_id":30660,"name":"Jan Marquardt","email":"jm@artfiles.de","username":"janno"},"date":"2020-07-31 12:59:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7db77809cc15840611d250057388c709e76eec5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 15:01:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5f2c74691fdc44199122e634638a8bc2 : SUCCESS in 11m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/22d3bb5da44741fa8d43ec7451c1379e : SUCCESS in 1h 04m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f06f3d198d5b427bac99e5f098baa779 : SUCCESS in 59m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4a88bb617974388941e4b910e17e589 : SUCCESS in 1h 00m 16s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f10c5bfdc80b4b22b77b07b4f16346de : FAILURE in 56m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4108cce981ac4aedacaa7a8b5216d8a0 : FAILURE in 57m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2d0dd1ec00914834a2ce899f9ae3c1d6 : FAILURE in 1h 35m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/758e8f6a30054fc388c087410c4e927e : FAILURE in 1h 10m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1709401a4cd94705aebc4c52a4d8a0a4 : FAILURE in 1h 16m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bf9c2e0b6d844e4a149a49ca6ff106f : FAILURE in 57m 05s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ae15081d6cce4e28bc2b332069c6a189 : FAILURE in 1h 29m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53e858d806b1492e9ff2e5ab10ab30ef : FAILURE in 1h 44m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db83be2bb7a94e9ebedbb9f5eaaf1b85 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"058e8a7b882eba8dad4b08bb254b670a2c996a9e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-31 16:51:22.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThere\u0027s some unrelated to the patch issue with designate resource creation","accounts_in_message":[],"_revision_number":1},{"id":"03f3ff534af38c1517d8bf427799ec3672005f71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-21 15:36:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2f176aa069c7c2da17ba09a8856a51461be74426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 17:57:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/437a19bd6eb54a5ea93a1374689fca21 : SUCCESS in 14m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1d5f53dfa85d4f45b41fd61c3acf3023 : SUCCESS in 1h 03m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67728beec761493baf2ef174119d8934 : SUCCESS in 58m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3c86faab83b64bbf844e526573b01d19 : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ff2baf9cbd9e431291e85168c0487856 : SUCCESS in 1h 43m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d867506ea8c74f999eab29f67e3b99ab : SUCCESS in 1h 25m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/554dde8351b8415b9ef3f78aef0db852 : FAILURE in 1h 20m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0abea18c359b4af2affaf3b1b0e3710a : SUCCESS in 2h 05m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cc9bb3efef5417884b645a05b2ef106 : SUCCESS in 1h 40m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f920ff7abff4179a29597498ee87908 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d72943c16860bd57e2714aa9186f61c0cb3b0047","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-25 17:23:44.000000000","message":"Abandoned\n\nI believe this has been solved differenty","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"07e8bf0534d5058d6bf2b33d078ebd605b8a9a10","revisions":{"8b44a5b6ad416eb011ee6078b2bdc23619cab2cf":{"kind":"REWORK","_number":1,"created":"2020-07-31 12:59:07.000000000","uploader":{"_account_id":30660,"name":"Jan Marquardt","email":"jm@artfiles.de","username":"janno"},"ref":"refs/changes/49/744149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/49/744149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/49/744149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/49/744149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/49/744149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/49/744149/1"}}},"commit":{"parents":[{"commit":"18c54d70648547a758395180e244f4c4b5084deb","subject":"Add Centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/18c54d70648547a758395180e244f4c4b5084deb"}]}],"author":{"name":"Jan Marquardt","email":"jm@artfiles.de","date":"2020-07-31 12:55:02.000000000","tz":120},"committer":{"name":"Jan Marquardt","email":"jm@artfiles.de","date":"2020-07-31 12:59:04.000000000","tz":120},"subject":"Adds handler for copying policy.json to the right place","message":"Adds handler for copying policy.json to the right place\n\nIf designate_policy_overrides existed, the changes were written to\n/etc/designate/policy.json-{{ designate_venv_tag  }}, but not\ncopied to /etc/designate/policy.json.\n\nChange-Id: I9490820bf344fc387a15020d0c5dedf08ab925f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/8b44a5b6ad416eb011ee6078b2bdc23619cab2cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/8b44a5b6ad416eb011ee6078b2bdc23619cab2cf"}]},"branch":"refs/heads/master"},"07e8bf0534d5058d6bf2b33d078ebd605b8a9a10":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-21 15:36:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/744149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/49/744149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/49/744149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/49/744149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/49/744149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/49/744149/2"}}},"commit":{"parents":[{"commit":"a5d177dfc99976fb9f99916eaf15b61ccb6b09c5","subject":"Use global service variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/a5d177dfc99976fb9f99916eaf15b61ccb6b09c5"}]}],"author":{"name":"Jan Marquardt","email":"jm@artfiles.de","date":"2020-07-31 12:55:02.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-01-21 15:36:01.000000000","tz":0},"subject":"Adds handler for copying policy.json to the right place","message":"Adds handler for copying policy.json to the right place\n\nIf designate_policy_overrides existed, the changes were written to\n/etc/designate/policy.json-{{ designate_venv_tag  }}, but not\ncopied to /etc/designate/policy.json.\n\nChange-Id: I9490820bf344fc387a15020d0c5dedf08ab925f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/07e8bf0534d5058d6bf2b33d078ebd605b8a9a10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/07e8bf0534d5058d6bf2b33d078ebd605b8a9a10"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
