)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~745696","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I196c263c7fb017a1d8386e8e9fb339af70165c51","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"I196c263c7fb017a1d8386e8e9fb339af70165c51","subject":"Add CentOS-8 support","status":"MERGED","created":"2020-08-11 15:09:34.000000000","updated":"2020-08-16 11:42:01.000000000","submitted":"2020-08-16 11:40:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745696-1597578000230-bd91aa04","meta_rev_id":"491fc655a38bc5d68cf0f1e0294134bfffeebcf6","_number":745696,"virtual_id_number":745696,"owner":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-16 11:40:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-08-12 12:51:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-12 19:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-15 14:23:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 12:51:37.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-08-15 14:23:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-16 11:40:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2f4e803ff3f589a094f281f3733ba96769b73ce","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-08-11 15:09:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8dca06432fd552edc67130ec60bdf8055512bae","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-11 15:11:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c84f44a7bb0007c35b1309b38f6e62ef0aa34f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 17:13:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ffd95252549b48a6a683c72c47ba26a3 : SUCCESS in 6m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/905a9dbf19ce486480917d28eec89399 : SUCCESS in 1h 02m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8cc00eedc9014c8383fab90271e5c8f5 : SUCCESS in 1h 08m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cca4346ca64345ee843a4321f9571f71 : SUCCESS in 57m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c506203ba6747eea076521caa74d8b9 : FAILURE in 43m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e1b35e4d8a8d4d5ca31b249e28ac3942 : SUCCESS in 1h 10m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dd6fb759e67e4155ad1afce98d29ed8d : SUCCESS in 1h 31m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8130e92fd84349e28c151bb4a1e8e39c : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5f84072493d4f8b9eca6a2e5682c9bf : SUCCESS in 1h 24m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ce39e82c20f41dea645163ce253f7d1 : SUCCESS in 1h 01m 48s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2cda74e12eea4c6a8bf4fe373eb7ff71 : FAILURE in 1h 56m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48e1b3f0b86a43bc8c360f38ff4df78a : FAILURE in 1h 39m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54987d6f44f14511b98b43371b50885c : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"fdf9b86445bd9d2b77405e3472012bda39cfed8f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-12 06:28:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48dd937847a3ad232bde7616fa2308aa71f9c95c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-12 07:27:39.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"faaf4be0732fe6cb15cd2373c58c51b512badee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 10:12:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/50780beed6714b3d8e390e6cc866037d : SUCCESS in 9m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/98fa5e0f143b494cbbfe4e01b34f3f67 : SUCCESS in 1h 17m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/341c69b82b1f4bde98e3ebd4c4ce365e : SUCCESS in 1h 03m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b96c44a0f69d47938e38f30ac4e5c4f2 : SUCCESS in 1h 14m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0294890a241745ed96fb1645729642dd : SUCCESS in 59m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a2c492297dc047d28a45c182d72343d2 : SUCCESS in 1h 11m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1be8f5ec90ff4bc4ab4917203a3295a6 : SUCCESS in 1h 34m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fad13f615b23484aaa5b921ecfdd0dbb : SUCCESS in 1h 19m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/937922282e584fff8a71622b1e03509f : SUCCESS in 1h 12m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3f9de59fc6b4046bdceb7b0f844459d : SUCCESS in 1h 03m 34s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4d5ab9a36dd5414180e6dc73b04ba089 : SUCCESS in 2h 36m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a1be810614a40bda68e19f9a97ed14b : SUCCESS in 1h 50m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0bc923d80db4c9d9e82feb6266bc1ba : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":3},{"id":"7d54adf9cd0cda6e38c4fea280678e6f07fa2cb4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-12 12:51:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"aa0d0be64f54a87a0d2247c7d44de9eb174ba60f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-12 19:26:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d9e723063749fcce28142e630e0e541ec379d4f1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-13 06:32:30.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 501c5b93b0f2ce56f12a6b5b912b2763a7552efd","accounts_in_message":[],"_revision_number":3},{"id":"c85a7f5863298f298af1dae510303ea3b38df834","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-15 14:23:42.000000000","message":"Removed Workflow+1 by Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"c9d8999300df24931a20a7985b9c8b869793ca78","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-15 14:23:48.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3b1d46d2229147ff042990fec14c4baf99d60e37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 14:24:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8fc7bbbbb499be5665521d5e6b6445bcf6ecb822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 17:38:30.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/904df266f75a4f358d02ab847b92e821 : SUCCESS in 8m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8162f7263e2a47e5800c976066f9d0d4 : SUCCESS in 1h 00m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/edc4f5146ff640cb9c15df1dcf8e26bf : SUCCESS in 58m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7b3c4ede86a49068b5704d2caaf1f3b : SUCCESS in 1h 16m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d84ba8e53884b2aaea373ffed8aa2d4 : SUCCESS in 55m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/42a6375c9480467a9527293c48d70183 : TIMED_OUT in 3h 01m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dca081d4a062417ebebfbc5cb1452e99 : SUCCESS in 1h 49m 09s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9e9c8e3ab5744da79c9fa7da3fe106e8 : SUCCESS in 1h 15m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ec6880c666c4964969b9f9c59d4b28f : SUCCESS in 1h 30m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10bf7d016c084da38af36ea1dcc1bde8 : SUCCESS in 1h 39m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e260b1ed0f74b8288884cf024f2a68a : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":3},{"id":"2545dafbc47bcde14cecff6ae3df311c459f23df","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-15 19:07:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"24b6958dffab9b2066fee9288a46b898eda4ad72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 22:12:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9ad5c67ef8b741e798f720aa7a309694 : SUCCESS in 6m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/854759ecd1264bc8871952d06f2e7b1c : SUCCESS in 1h 05m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ab91e00d3a9644c692f8d81db1e1d170 : SUCCESS in 1h 00m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0dc673cdc1d74e7099b2872575658dfc : SUCCESS in 59m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8676195679f44054814fdfbaf7100cad : SUCCESS in 52m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3b049e2aab8a4261ae1f624c37b532c3 : SUCCESS in 1h 10m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/93cdb53bdf044e588dc6897be2119dce : SUCCESS in 1h 28m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/90f28c601d3d42cc86dea8e11feb6fad : SUCCESS in 1h 21m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/837bfd0e8ec646139b0523ed9eb30afa : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c072091a4254398aa5ff452ac1e58b4 : SUCCESS in 1h 03m 56s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1146cfb0755b40908ecf32c3f4ac4f04 : SUCCESS in 2h 28m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a01da2e0b3b043d4b66b1a2afb2c5012 : TIMED_OUT in 3h 01m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97344a4fac1443908a7d043dd77073d3 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":3},{"id":"2366a96727653aafaa86f7d5135f1d91db0df74d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-16 07:57:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e3b03ed2e6e2b4cb6608696ffa592bfba85f5464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 10:17:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/be7188c2ca6f475fb0063b174cb14ba9 : SUCCESS in 5m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/54e7f177ac6848b099dd3f460a85e823 : SUCCESS in 1h 01m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3b4ffc55a8ac48089c2f318a5bfc268f : SUCCESS in 1h 01m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e356b6aa1b74a79b21037532af1539b : SUCCESS in 49m 14s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/695475233dcb42eb925f575993e5824c : SUCCESS in 45m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/98321c98dda54a1dad963cef787ce8b8 : SUCCESS in 58m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d14712eb26394edba1e338a64a8bc578 : SUCCESS in 1h 25m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/39607c3c192e4286b3b4afaa911492ae : SUCCESS in 1h 18m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0754ad859a3743b8aeaf0ed4b40be1fd : SUCCESS in 1h 08m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19720163ccff4946b5a34198759eed91 : SUCCESS in 57m 15s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1fc69ca585ca4eb9bac6605d63b2d3aa : SUCCESS in 2h 13m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/990885e1284143c782af3d326c0571fe : SUCCESS in 2h 03m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343025fb17dd4071b93fe81f543df09c : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"9fc2c6118cf3df91e7f2d01472ecbb4bc935f1bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 10:18:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dab6942c04da6ab05a76483af6cd50c289389a70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 11:40:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d4135b7c5093497faedbfd11dd60c3f5 : SUCCESS in 8m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/df635c3b13804900b17236f0d638a17b : SUCCESS in 1h 08m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0d5a61ec2261457bbcd7a07b98e646e8 : SUCCESS in 59m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/725bc04d6e1a4c7bbfc09e2e45d0d50f : SUCCESS in 57m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13bb03b4907243d8a2e1396768938590 : SUCCESS in 48m 21s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9f7e66c59e0f414787771a0a690d7c3d : SUCCESS in 1h 05m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/99154554934c428d847508f3f98113fb : SUCCESS in 1h 17m 16s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1d4c8e0d0ea349188effc1026e266c1c : SUCCESS in 1h 16m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bee1d119fba648b2b317c5b35c0bd580 : SUCCESS in 1h 08m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/784254e4570f4c75809649dbdcca4544 : SUCCESS in 1h 01m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45f30d70c6045bf9b4aef4368090d7c : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":3},{"id":"50f1b8ce9c3326129968ff69415ccc8141b3ce86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 11:40:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8fc62bab9d95310f62b9c5099be4350ee94faad5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 11:42:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07e8c668179c47618a13523c446ae5e0 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d16298c411a40c307752b6a6494ee8379a654525","revisions":{"38e87732eaf5c7278a648955d75ced44bb31580e":{"kind":"REWORK","_number":1,"created":"2020-08-11 15:09:34.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/96/745696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/96/745696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/1"}}},"commit":{"parents":[{"commit":"18c54d70648547a758395180e244f4c4b5084deb","subject":"Add Centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/18c54d70648547a758395180e244f4c4b5084deb"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-08-11 15:04:40.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-08-11 15:04:40.000000000","tz":-240},"subject":"Add CentOS-8 support","message":"Add CentOS-8 support\n\nreferencing ansible_python_interpreter to main.yml file to fix it.\n\nChange-Id: I196c263c7fb017a1d8386e8e9fb339af70165c51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/38e87732eaf5c7278a648955d75ced44bb31580e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/38e87732eaf5c7278a648955d75ced44bb31580e"}]},"branch":"refs/heads/master"},"c5ab0575084029a38b36ef0455add33a240bec7b":{"kind":"REWORK","_number":2,"created":"2020-08-12 06:28:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/96/745696/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/96/745696/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/2"}}},"commit":{"parents":[{"commit":"18c54d70648547a758395180e244f4c4b5084deb","subject":"Add Centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/18c54d70648547a758395180e244f4c4b5084deb"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-08-11 15:04:40.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-12 06:28:35.000000000","tz":180},"subject":"Add CentOS-8 support","message":"Add CentOS-8 support\n\nreferencing ansible_python_interpreter to main.yml file to fix it.\n\nChange-Id: I196c263c7fb017a1d8386e8e9fb339af70165c51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c5ab0575084029a38b36ef0455add33a240bec7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c5ab0575084029a38b36ef0455add33a240bec7b"}]},"branch":"refs/heads/master"},"d16298c411a40c307752b6a6494ee8379a654525":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-12 07:27:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/96/745696/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/96/745696/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/96/745696/3"}}},"commit":{"parents":[{"commit":"18c54d70648547a758395180e244f4c4b5084deb","subject":"Add Centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/18c54d70648547a758395180e244f4c4b5084deb"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-08-11 15:04:40.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-08-12 07:27:35.000000000","tz":0},"subject":"Add CentOS-8 support","message":"Add CentOS-8 support\n\nreferencing ansible_python_interpreter to main.yml file to fix it.\n\nDepends-On: https://review.opendev.org/710849\nChange-Id: I196c263c7fb017a1d8386e8e9fb339af70165c51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/d16298c411a40c307752b6a6494ee8379a654525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/d16298c411a40c307752b6a6494ee8379a654525"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
