)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~755533","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I9d99593b03913a09fe20082c39d3bbd044f07acc","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I9d99593b03913a09fe20082c39d3bbd044f07acc","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-10-01 14:30:02.000000000","updated":"2020-10-02 16:33:13.000000000","submitted":"2020-10-02 16:31:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755533-1601656297003-72bff4cb","meta_rev_id":"3e690d3b8e10cefad7b9298625018e5bc734874a","_number":755533,"virtual_id_number":755533,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 16:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-02 12:21:14.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-02 14:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-02 14:50:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 12:21:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-02 14:50:19.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-02 16:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7ac31d6b5ef7558a0f7cbf561048eb1d948d223","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-01 14:30:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e51d1a1e769123e350fb1b4b6a704c53b2ba27b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 19:04:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/066f04f06595482396772bc1afbe5e8f : FAILURE in 8m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/035448f38d2d4efc84e021cd008b57c1 : SUCCESS in 56m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/881f4c5c14304471bc7e63824de8eb52 : SUCCESS in 59m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/56ca85eea0364903887e1fe48686b404 : SUCCESS in 1h 08m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a0113458b96442bc8b7f8d4ec07198e0 : SUCCESS in 1h 40m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/277dd418db5a44d1b25b3ef33d0cca04 : SUCCESS in 1h 26m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eb3de4c76014547881ab6ac0a52cd7e : SUCCESS in 1h 04m 02s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2f2fdf1fe4364932beac992dba274c8d : SUCCESS in 2h 16m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7fe60d0f25e4344ad52dc4a4bd07d3b : SUCCESS in 1h 18m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665aa14693d44df7bb3762fc6117fa37 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c00da14fa0844b7a810a49fcbfa6746ee91a1699","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 12:21:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"725a0c5976a16824fbadf06dc007c43990ca49e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 14:49:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/22a1983ad1ad408faf26bddf7156206b : SUCCESS in 9m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5595756e72b24596ae7892efdbaa0a56 : SUCCESS in 1h 04m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e2e36fdbc20451c923b3006284a91ad : SUCCESS in 1h 03m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e3d58dee49604918aec399a011e8103f : SUCCESS in 1h 06m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/412ebe3b79f14c58be694547213e2ed8 : SUCCESS in 1h 45m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bf6f08299264fb78a97dbb1377650f9 : SUCCESS in 1h 18m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7f60a98e66941799f20d94ddd828750 : SUCCESS in 1h 07m 08s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a4e014f86a0c4189a09fb7c476fd6ec9 : SUCCESS in 2h 07m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34b208f578594c378a8f397546571393 : SUCCESS in 1h 39m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213b98bb18ba4b329255392bf37f4d1f : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"64ec8c674939049c721f14efc8cd67ce56f6031d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-02 14:50:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a09864b8b508ae94055d462b56877eafc47b2244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 14:50:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3a863520a0b29893c3b044610df884b27128f6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 16:31:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1856399cd52d473aa4da5860b4bf3efb : SUCCESS in 10m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9e07b125a4824e1fa271480a35576e9e : SUCCESS in 1h 06m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fb603bbd8cb45efbe6ce90757af2a07 : SUCCESS in 1h 00m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/da185bffafc2496b8960874cfe9024b6 : SUCCESS in 1h 14m 50s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cd0e9f6829d647e68387ccd4059d40d9 : SUCCESS in 1h 32m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96a721baedec47e8acb32495b0b3fab8 : SUCCESS in 1h 26m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe720916f1354ab494b9406b9fdefe63 : SUCCESS in 1h 08m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f42f21dd7734a3a9d3b8720b502a4a6 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"96e17e3a792b00593590ed1e5ba9f4d0e43cbaa2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 16:31:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"be7710c10e3738b7774525e8feddfa10a2bcf032","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 16:33:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdcc5dd1d0374146a9c32c2cab1f38c3 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a06d75b7eadbbaaa73ac0cd1b5d3b390dcbe24c8","revisions":{"a06d75b7eadbbaaa73ac0cd1b5d3b390dcbe24c8":{"kind":"REWORK","_number":1,"created":"2020-10-01 14:30:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/755533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/33/755533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/33/755533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/33/755533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/33/755533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/33/755533/1"}}},"commit":{"parents":[{"commit":"0bc0f321b70e4986e87d5ef3b86ac41f6e4676c5","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/0bc0f321b70e4986e87d5ef3b86ac41f6e4676c5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-01 14:30:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-01 14:30:01.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I9d99593b03913a09fe20082c39d3bbd044f07acc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/a06d75b7eadbbaaa73ac0cd1b5d3b390dcbe24c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/a06d75b7eadbbaaa73ac0cd1b5d3b390dcbe24c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
