)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~755720","triplet_id":"openstack%2Fopenstack-ansible-os_designate~stable%2Ftrain~I3a4a1823a9c9a05164e8a1654a5c39d6c935bbb9","project":"openstack/openstack-ansible-os_designate","branch":"stable/train","hashtags":[],"change_id":"I3a4a1823a9c9a05164e8a1654a5c39d6c935bbb9","subject":"Fix python interpreter bug for designate source installs on train","status":"MERGED","created":"2020-10-02 07:35:50.000000000","updated":"2020-10-17 18:24:18.000000000","submitted":"2020-10-17 18:23:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755720-1602958991070-84541501","meta_rev_id":"6cfad3d6c17b8f7b2b1b5f8e02555fcbdeb1af07","_number":755720,"virtual_id_number":755720,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-05 14:45:50.000000000","_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-17 18:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-10-16 13:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-05 17:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2020-10-16 13:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-17 16:06:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 13:26:11.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-17 16:06:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-17 18:23:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6c9c21adcb467629f646a2de38dcba5eb99781a","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-02 07:35:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2dcc9fef68bf9411bb6f2884393ff262afea3705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 09:51:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/699bc63102ce483a922148d71b258665 : SUCCESS in 6m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d7a299919be74c3ab82c5407f4031096 : SUCCESS in 1h 09m 10s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fd0648edf2874596aed0b23793a43956 : RETRY_LIMIT in 16m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41c9e694e34a4fd9b71d04a903fe700f : SUCCESS in 57m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/60fb2855b61f467d92f69f3d1449be18 : SUCCESS in 1h 06m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d495db4dc21347029a123d5a33494db3 : FAILURE in 1h 05m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e4b0cf4cc0d246788ba807b6ebcdf695 : FAILURE in 1h 14m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d18eef7acc0f433aaaad22598ddb9543 : RETRY_LIMIT in 14m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d62f16be402c495e8df0732afa897e60 : SUCCESS in 1h 13m 18s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/26a2a9779dc84671bc1de426377e1145 : FAILURE in 2h 08m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93261d3a86924b13805a6563bdab8949 : FAILURE in 1h 24m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34988d7164fd435f87e697fc7e8280f1 : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2d1ac37a34a91ee21b621bd872211d07fc5589cf","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-02 12:30:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"864e581e5e3eaea1050dae7aab4ce5e7525b0bb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 15:04:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/88a77b3b8907480190fa94d4b3bec3b2 : SUCCESS in 7m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/818c5494a3054948877073b0fd436d85 : SUCCESS in 1h 06m 00s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/17faec716af943de9fc112f8b6395e89 : RETRY_LIMIT in 17m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0c29b632a3949699179b7d51217aa2c : SUCCESS in 58m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/72bbaa4f45244f8591028b395e11dc05 : SUCCESS in 1h 04m 52s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b457e9bf6a214f349eaa56b0ae85e56b : FAILURE in 1h 13m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/893f016f32124119bcba528173b549c8 : FAILURE in 1h 14m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/546a453e32154b3ba038550cf3b950a5 : RETRY_LIMIT in 16m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20b0aef9ad51497b83759724a0339534 : SUCCESS in 1h 09m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/117df6f3e47d4c6da651f22e62a1616c : FAILURE in 2h 11m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/feff9e70e4e54a3b9030228377b2f029 : SUCCESS in 1h 52m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62233c56641544379050a504d78dc786 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"028f9f9c3c4824426d2cda525a60f1e946c5a6f5","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-05 08:59:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"359f5fcc4f10dcc6f3aefa55bac27c85638b17c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 11:22:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b94caeae4f9549b7a73355ca0b07464a : SUCCESS in 6m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/617b8f8810454342ba1526c77c33591c : SUCCESS in 1h 04m 41s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a94558472e2d464997a6407ab7256783 : RETRY_LIMIT in 16m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ddba1bf76794ee1a5945477a0f10aba : SUCCESS in 57m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/438c320c3388422fa0627389836606b4 : SUCCESS in 56m 03s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3a2c2e0a791c4436ae978d2dc4d5c1af : FAILURE in 1h 05m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e98a48f3efcc444593c9e9411e6741d0 : FAILURE in 1h 20m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d4f8fefb09a24178882e759760956013 : RETRY_LIMIT in 16m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f9385136f624b4187dd5008f0789bdd : SUCCESS in 1h 06m 32s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/40cebef18f624081a3d62e20d025b804 : FAILURE in 2h 13m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b02d89069f324c6fbd7c1c19d88a2f4a : SUCCESS in 1h 43m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9cb366e825a49338e5eb7d294eceb22 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"36e0c8cfb5a5adef159abe7dd8b8ab92d2a57458","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-05 12:23:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"354cc52d9d112bf0a1571904dfb4ad1876f26b66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 14:43:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/aca06d501a494dc8b897add172be8f21 : SUCCESS in 8m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ba65b781dcac4d69aff4e85e0758b994 : SUCCESS in 1h 11m 09s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7bebe9fde6a04fca96a132edd45ecb95 : RETRY_LIMIT in 14m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76c7884bad01439eb2dfae6064560ed5 : SUCCESS in 51m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/680c2a0cca374ac79324b91bd9029ba9 : FAILURE in 41m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/32ee101b544447488230e6b1437607a6 : SUCCESS in 58m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/41b306008e44474b9a971e32a499915a : SUCCESS in 1h 44m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8e736464913a4badb4e5d0b0b4330fd1 : RETRY_LIMIT in 15m 24s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1db55e45787b4d0ea32f34f7668f483b : FAILURE in 35m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/20cc1625c5e84b9ca997adb9c2aa4149 : FAILURE in 2h 08m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52224c0454484ae6b8909b5e84d16da3 : SUCCESS in 1h 45m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83938e16943945829d27c0fde63069b9 : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ddfbfd28350950b29d6a88068ce4d5c89c78a33f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-05 14:45:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2fa9c8116cc61a1c811dccee638cfff713ba88a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:59:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8b1c52c8f97d4e47a1b33d08bb1fe7b7 : SUCCESS in 7m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/228c593020a44c5b93da42f4ed3f164a : SUCCESS in 1h 08m 00s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/19cfa41047ba4880885779989b075d75 : RETRY_LIMIT in 17m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff7df8b01c204644a60ca12119d92c6b : SUCCESS in 51m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/55ae1d9aad7646d2b8235fdf94d07472 : SUCCESS in 1h 07m 27s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/17b553fea9be4cdea9212e27aba6c8d9 : SUCCESS in 1h 14m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5b2889f2e3cb4ad89d71b10fc64c5248 : SUCCESS in 1h 31m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/eb423cb89dc140bfafc758a882584645 : RETRY_LIMIT in 14m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cd60885807247248d3f8d7650fa4e2e : SUCCESS in 1h 11m 15s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/71da417e177346e5ba90499776ece6c9 : FAILURE in 2h 05m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88f106311d3f423495232ca77140280e : SUCCESS in 1h 49m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68be30751b194b2c8e4e4cee60dd599e : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":2},{"id":"03b6f089d47fa47996a403a329fd1c0ed316f82c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-05 17:04:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a572ab4f48f78c99a8db6b753f4d22e27e97891b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-16 13:26:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7d7568e146ca19450da4018430716b989865fcc8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-17 16:06:16.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e4b03ef315cd53fc3ee35d008774d9c0767a7210","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 16:06:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d09d796cbba0b68f8aae67c9e970e0e50e038b39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 18:23:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/26603be18e654202a9c90514b6f77acb : SUCCESS in 7m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/56a76dcbb90046d791609031cd5889fc : SUCCESS in 58m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b46879358754fca912148c797cf1dbb : SUCCESS in 47m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6ed6ecf7223247008e0393c35042964c : SUCCESS in 1h 04m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/cff28e5064f8457b8202320e7bb70582 : SUCCESS in 1h 22m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/19fc1f1acaab477084a2c5e4514291ef : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/134d1ac996684473a03bef665e74b1dd : RETRY_LIMIT in 18m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ede9b2f78c86489494b8aff2c70134da : SUCCESS in 1h 14m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16ef42e6540442f3aa01f41fc5ddbf34 : SUCCESS in 1h 23m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/397e0bebad354e35a0299bb233fecff8 : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":2},{"id":"ee8bc1d9fab353be6c825068f03e8a070a9c06bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 18:23:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6cfad3d6c17b8f7b2b1b5f8e02555fcbdeb1af07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 18:24:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169dbfbc169e4003887da407fafbbf85 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1dba8cf97f84e8b57000b30f24e0d485d2b0f2fb","revisions":{"fa683908c7a51d42251091b191f7626689d39091":{"kind":"REWORK","_number":1,"created":"2020-10-02 07:35:50.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/20/755720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/20/755720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/20/755720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/20/755720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/20/755720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/20/755720/1"}}},"commit":{"parents":[{"commit":"60f018a90c24398fb3399d9f29d2a6a07b3c1324","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/60f018a90c24398fb3399d9f29d2a6a07b3c1324"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2020-10-02 07:27:05.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2020-10-02 07:34:59.000000000","tz":60},"subject":"Fix python interpreter bug for designate source installs on train","message":"Fix python interpreter bug for designate source installs on train\n\nWhen performing post install tasks, the openstacksdk is\nimported using a python2 interpreter in a Ubuntu Bionic utility\ncontainer (as a result of a fallback to\nansible_python[\u0027executable\u0027]). This fails as the utility container\nhas been set up using python3.\n\nThis patch corrects the python interpreter to match the variable\nalready used in service_setup.yml. This is a partial backport of\nhttps://review.opendev.org/#/c/745696/\n\nChange-Id: I3a4a1823a9c9a05164e8a1654a5c39d6c935bbb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/fa683908c7a51d42251091b191f7626689d39091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/fa683908c7a51d42251091b191f7626689d39091"}]},"branch":"refs/heads/stable/train"},"1dba8cf97f84e8b57000b30f24e0d485d2b0f2fb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-05 12:23:54.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/20/755720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/20/755720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/20/755720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/20/755720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/20/755720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/20/755720/2"}}},"commit":{"parents":[{"commit":"60f018a90c24398fb3399d9f29d2a6a07b3c1324","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/60f018a90c24398fb3399d9f29d2a6a07b3c1324"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2020-10-02 07:27:05.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2020-10-05 12:23:22.000000000","tz":60},"subject":"Fix python interpreter bug for designate source installs on train","message":"Fix python interpreter bug for designate source installs on train\n\nWhen performing post install tasks, the openstacksdk is\nimported using a python2 interpreter in a Ubuntu Bionic utility\ncontainer (as a result of a fallback to\nansible_python[\u0027executable\u0027]). This fails as the utility container\nhas been set up using python3.\n\nThis patch corrects the python interpreter to match the variable\nalready used in service_setup.yml. This is a partial backport of\nhttps://review.opendev.org/#/c/745696/\n\nDepends-On: https://review.opendev.org/756057\nChange-Id: I3a4a1823a9c9a05164e8a1654a5c39d6c935bbb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/1dba8cf97f84e8b57000b30f24e0d485d2b0f2fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/1dba8cf97f84e8b57000b30f24e0d485d2b0f2fb"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
