)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~814341","triplet_id":"openstack%2Fopenstack-ansible-os_designate~stable%2Fstein~Ie00eacabacae3f63c0d6d598694086d97c07ea8d","project":"openstack/openstack-ansible-os_designate","branch":"stable/stein","topic":"debian-stable-rm","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2021-10-20 03:36:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie00eacabacae3f63c0d6d598694086d97c07ea8d","subject":"Remove debian-stable job","status":"MERGED","created":"2021-10-18 03:35:58.000000000","updated":"2021-10-20 03:45:38.000000000","submitted":"2021-10-20 03:36:41.000000000","submitter":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814341-debian-stable-rm","meta_rev_id":"4cdbd57a13c363e3d9a32810ddb9ab774d2a99a3","_number":814341,"virtual_id_number":814341,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":2,"date":"2021-10-20 03:36:41.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":2,"date":"2021-10-20 03:36:41.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":1,"date":"2021-10-20 03:36:41.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-18 05:03:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 03:36:41.000000000","updated_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"reviewer":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"state":"REVIEWER"}],"messages":[{"id":"12541bcced8a7e743dde102e459cc7c64b385b35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-10-18 03:35:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c4028721799370c8f4e969a5a36a127da35e441","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-10-18 04:14:04.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e1385dd4cee14b8e0a15da2018ccd32f5886281f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 05:03:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c5c942167f7349a391f60c0032307a29 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7efbb18d956c41d1a737eac9ef9f2c5e : SUCCESS in 4m 56s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 https://zuul.opendev.org/t/openstack/build/2de519d4d0a7473abed493aaeb02eff2 : RETRY_LIMIT in 3m 32s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4eb9e29c01864a25b31940f947030bb7 : FAILURE in 47m 38s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 https://zuul.opendev.org/t/openstack/build/0722b0a8f77843aeb2eba41eef3c76ed : RETRY_LIMIT in 2m 47s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a55881d0e6f4d728ec4032bfcb7c24d : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ba340ebc2d5dc3a700bf9ab5006377213b05b0aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-10-19 04:59:37.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e55ef14d09bfdde266782d234dd6b0beb5c076a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 06:13:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e037fcc737a24f20aef8a64e9ae56cc8 : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6523b2850d314c6e86e5cf10763e018d : SUCCESS in 5m 12s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 https://zuul.opendev.org/t/openstack/build/f75ca152fa2f4bec8efcf575fcc65552 : FAILURE in 1h 00m 55s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12e7a27e7fc1462fa04ce6f250f30230 : FAILURE in 50m 45s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 https://zuul.opendev.org/t/openstack/build/0443059665f440529ee482518381ba86 : SUCCESS in 51m 56s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd17f15953654c77a77aafcca267ade2 : SUCCESS in 43m 23s","accounts_in_message":[],"_revision_number":3},{"id":"377744ee3323ae09633eebfa7457eeece084f9d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-10-19 07:55:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"674cf8bf180b357ceadc5232eb1cf12fc96abbdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 09:05:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3692b737f7ed4bc6a4518e612486ce1d : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4761317df4dd48f095cd1cc636b0d517 : SUCCESS in 4m 37s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 https://zuul.opendev.org/t/openstack/build/2cfe860b36bf4b3aa85d97ee33f185de : FAILURE in 1h 03m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/defedc02497a405f8309db4cfe29bff5 : FAILURE in 50m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 https://zuul.opendev.org/t/openstack/build/c23054bfd89047acb400afe035e79f65 : SUCCESS in 57m 02s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1449ab973c8b4fe7a0e8d7302e28f496 : SUCCESS in 48m 12s","accounts_in_message":[],"_revision_number":4},{"id":"4d4d0129b39a6c056b419f566feb2223321537c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-10-20 03:15:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d53f75d5ab639e50682da8b52b8711ef6b7efd81","author":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"date":"2021-10-20 03:36:41.000000000","message":"Patch Set 5: Verified+2 Workflow+1 Code-Review+2\n\nMerging to enable debian-stable removal","accounts_in_message":[],"_revision_number":5},{"id":"22dd629f32b63a631deb17ce7a90fa54b2447beb","tag":"autogenerated:gerrit:merged","author":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"date":"2021-10-20 03:36:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"4cdbd57a13c363e3d9a32810ddb9ab774d2a99a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 03:45:38.000000000","message":"Patch Set 5:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b743f76df5f3483dad87d06b536fa4c1 : FAILURE in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"13a52643452a86a0c3c80c2ce4fbcc61b26928ae","revisions":{"265ffddcb3619ff34d3e36e0b1e38791ce9c5f1a":{"kind":"REWORK","_number":1,"created":"2021-10-18 03:35:58.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/41/814341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/41/814341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/1"}}},"commit":{"parents":[{"commit":"c7ffaaafb416b697af1f67ef985f012db3078f53","subject":"Switch to opensuse-15","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c7ffaaafb416b697af1f67ef985f012db3078f53"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-18 03:35:18.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-18 03:35:55.000000000","tz":660},"subject":"The \"stable\" this originally referred to is stretch, which is now two \"stable\"\u0027s ago.  We\u0027ve since realised that giving nodesets generic names doesn\u0027t work so well for this reason and will use codenames in the future (debian-bullseye, etc.).","message":"The \"stable\" this originally referred to is stretch, which is now two\n\"stable\"\u0027s ago.  We\u0027ve since realised that giving nodesets generic\nnames doesn\u0027t work so well for this reason and will use codenames in\nthe future (debian-bullseye, etc.).\n\nSince we\u0027re not going to update this old branch to new releases,\nremove this job so infra can remove the debian-stable nodeset.\n\nChange-Id: Ie00eacabacae3f63c0d6d598694086d97c07ea8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/265ffddcb3619ff34d3e36e0b1e38791ce9c5f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/265ffddcb3619ff34d3e36e0b1e38791ce9c5f1a"}]},"branch":"refs/heads/stable/stein"},"4dd2df62de893258f1a8572384a6be5d6482a83b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-18 04:14:04.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/41/814341/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/41/814341/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/2"}}},"commit":{"parents":[{"commit":"c7ffaaafb416b697af1f67ef985f012db3078f53","subject":"Switch to opensuse-15","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c7ffaaafb416b697af1f67ef985f012db3078f53"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-18 03:35:18.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-18 04:13:49.000000000","tz":660},"subject":"Remove debian-stable job","message":"Remove debian-stable job\n\nThe \"stable\" this originally referred to is stretch, which is now two\n\"stable\"\u0027s ago.  We\u0027ve since realised that giving nodesets generic\nnames doesn\u0027t work so well for this reason and will use codenames in\nthe future (debian-bullseye, etc.).\n\nSince we\u0027re not going to update this old branch to new releases,\nremove this job so infra can remove the debian-stable nodeset.\n\nChange-Id: Ie00eacabacae3f63c0d6d598694086d97c07ea8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/4dd2df62de893258f1a8572384a6be5d6482a83b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/4dd2df62de893258f1a8572384a6be5d6482a83b"}]},"branch":"refs/heads/stable/stein"},"52d9bf85faf314512eadaa97471e9cbbdaca07f6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-10-19 04:59:37.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/41/814341/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/41/814341/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/3"}}},"commit":{"parents":[{"commit":"c7ffaaafb416b697af1f67ef985f012db3078f53","subject":"Switch to opensuse-15","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c7ffaaafb416b697af1f67ef985f012db3078f53"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-18 03:35:18.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-19 04:58:58.000000000","tz":660},"subject":"Remove debian-stable job","message":"Remove debian-stable job\n\nThe \"stable\" this originally referred to is stretch, which is now two\n\"stable\"\u0027s ago.  We\u0027ve since realised that giving nodesets generic\nnames doesn\u0027t work so well for this reason and will use codenames in\nthe future (debian-bullseye, etc.).\n\nSince we\u0027re not going to update this old branch to new releases,\nremove this job so infra can remove the debian-stable nodeset.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/814506\nChange-Id: Ie00eacabacae3f63c0d6d598694086d97c07ea8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/52d9bf85faf314512eadaa97471e9cbbdaca07f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/52d9bf85faf314512eadaa97471e9cbbdaca07f6"}]},"branch":"refs/heads/stable/stein"},"95979156f1a87f45d60febbfd4af95184fc17460":{"kind":"REWORK","_number":4,"created":"2021-10-19 07:55:48.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/41/814341/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/41/814341/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/4"}}},"commit":{"parents":[{"commit":"c7ffaaafb416b697af1f67ef985f012db3078f53","subject":"Switch to opensuse-15","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c7ffaaafb416b697af1f67ef985f012db3078f53"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-18 03:35:18.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-19 07:55:22.000000000","tz":660},"subject":"Remove debian-stable job","message":"Remove debian-stable job\n\nThe \"stable\" this originally referred to is stretch, which is now two\n\"stable\"\u0027s ago.  We\u0027ve since realised that giving nodesets generic\nnames doesn\u0027t work so well for this reason and will use codenames in\nthe future (debian-bullseye, etc.).\n\nSince we\u0027re not going to update this old branch to new releases,\nremove this job so infra can remove the debian-stable nodeset.\n\nMake the failing aio_metal jobs non-voting and remove from the gate\nfor now.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/814506\nChange-Id: Ie00eacabacae3f63c0d6d598694086d97c07ea8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/95979156f1a87f45d60febbfd4af95184fc17460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/95979156f1a87f45d60febbfd4af95184fc17460"}]},"branch":"refs/heads/stable/stein"},"13a52643452a86a0c3c80c2ce4fbcc61b26928ae":{"kind":"REWORK","_number":5,"created":"2021-10-20 03:15:59.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/41/814341/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/41/814341/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/814341/5"}}},"commit":{"parents":[{"commit":"c7ffaaafb416b697af1f67ef985f012db3078f53","subject":"Switch to opensuse-15","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c7ffaaafb416b697af1f67ef985f012db3078f53"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-18 03:35:18.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-20 03:15:45.000000000","tz":660},"subject":"Remove debian-stable job","message":"Remove debian-stable job\n\nThe \"stable\" this originally referred to is stretch, which is now two\n\"stable\"\u0027s ago.  We\u0027ve since realised that giving nodesets generic\nnames doesn\u0027t work so well for this reason and will use codenames in\nthe future (debian-bullseye, etc.).\n\nSince we\u0027re not going to update this old branch to new releases,\nremove this job so infra can remove the debian-stable nodeset.\n\nChange-Id: Ie00eacabacae3f63c0d6d598694086d97c07ea8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/13a52643452a86a0c3c80c2ce4fbcc61b26928ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/13a52643452a86a0c3c80c2ce4fbcc61b26928ae"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
