)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~820231","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I28c64b44eadfd726e07cb7159e5d3d94fde106ed","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"db-pooling","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2021-12-03 15:43:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I28c64b44eadfd726e07cb7159e5d3d94fde106ed","subject":"Database connection pooling improvements","status":"MERGED","created":"2021-12-02 18:19:43.000000000","updated":"2021-12-03 15:45:03.000000000","submitted":"2021-12-03 15:43:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820231-db-pooling","meta_rev_id":"3675091341dbb2e612dc2a29f5b64c2103181dac","_number":820231,"virtual_id_number":820231,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-03 15:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-12-03 13:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-03 13:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-03 13:57:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 20:42:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-03 13:54:35.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-12-03 13:57:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"dcd751fbce160155b9af062a896d9041ff7fc2f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-02 18:19:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29ce144d49be3362dddb7b006e961b21fbbf6346","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 20:42:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/126536af96ee42c7810c5ca790519ad6 : SUCCESS in 17m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/49ded0a3b57d496484ce99a78bc9c98a : FAILURE in 41m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c181027cd1214fc6a3d27ee0e650aadc : FAILURE in 50m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b14740edd2f84d16ad42361a3eaaa415 : FAILURE in 1h 35m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/235d5609ed284e9186e15438100a8c85 : FAILURE in 1h 04m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fb9403461e404108821b28a3dc291ed6 : FAILURE in 1h 16m 08s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5e086d6a1dd4fa0a6f386bf55a45235 : FAILURE in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9d7710996824983bd0e5b3d52f44df6 : FAILURE in 1h 22m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0582cf69e294c4a84c8723dd8df9232 : FAILURE in 1h 22m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1abcd9c7cba4a809ed9f12a69b0f549 : FAILURE in 1h 54m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c76483e0a152448b83343805e94985f1 : FAILURE in 1h 51m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27453dadd2ab4a18a0c643ff11fe22ef : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"77da1ff6d05189d8f3b3ee9299ff7ac872295a0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-03 10:40:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c55dc59c1ad60eca6cb80dbcb4bfce7a3d46c840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:21:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc0fee490ec34d13b3aa08750544664e : SUCCESS in 20m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/efe54d5fceb349e59292838da1404fc2 : FAILURE in 40m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/850986417d124a3794d296fbf6109eb3 : SUCCESS in 1h 07m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/984b964acad447919f19f847866dc86c : SUCCESS in 1h 22m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/799463c5bddd45a6af08dcae513ba4c3 : SUCCESS in 1h 26m 28s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/37b459af6b65448f92e725325b7665c6 : SUCCESS in 2h 21m 03s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e530224ca9cd4258809f15b00f56b859 : SUCCESS in 1h 42m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66d0460e07134f66a8de342f09e508ba : SUCCESS in 1h 58m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4c7c16abc134089ae976626f59cb68c : SUCCESS in 1h 40m 57s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b11b90b0dbd44daa4865363782ab48b : SUCCESS in 2h 12m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7721f0a295454ea8969be2e554ff77ff : SUCCESS in 2h 15m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e351fe51df6413bbbdac1a888dd271f : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":2},{"id":"be093e8a06cd2afd0bc328407197995c57f6211d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-03 13:54:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0a04703051ec7eb005b25f4fc7e72fab66407e3f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-03 13:57:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"047f9ed3b284e53e515f44db418076002d052d43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:59:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c528f823fcbcd0f49e6a6cfbe8bf86d71d6bf06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 15:43:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/339cf7ff571b44389eb5d4e4e9dd1d0d : SUCCESS in 21m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/865a76690d5345e7a1d5dbb919d29961 : SUCCESS in 1h 06m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0ed92efb955545119bff0d18eea9e72c : SUCCESS in 1h 34m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6025eaca6a33459dbb6a631832321475 : SUCCESS in 1h 22m 34s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/87f9d521dec948de8069532e098b8a1f : SUCCESS in 1h 31m 22s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/357bde8aa4944e7c8ec7c98b3268026d : SUCCESS in 1h 34m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7407f2de8cfb445597ab65229535ffca : SUCCESS in 1h 26m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e03507848f942009e8672a79cb68666 : SUCCESS in 1h 29m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce0efa9192f4bd79dbf716d21af0547 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ce7245927bec61fa145b953b54083566983bc97b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 15:43:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3675091341dbb2e612dc2a29f5b64c2103181dac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 15:45:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c18ed4de5814ce79e51879329dbf332 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4667c80b90ff686c6dd16ae6c109753a55b90a4e","revisions":{"f3b6c571b834f4574e3a8a4a9664900ebb5c486f":{"kind":"REWORK","_number":1,"created":"2021-12-02 18:19:43.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/31/820231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/31/820231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/31/820231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/31/820231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/31/820231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/31/820231/1"}}},"commit":{"parents":[{"commit":"4d4366ed164f298e03da7b59cdff210f1d393cf0","subject":"Use config_template as a collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/4d4366ed164f298e03da7b59cdff210f1d393cf0"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:19:33.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:19:42.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: I28c64b44eadfd726e07cb7159e5d3d94fde106ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/f3b6c571b834f4574e3a8a4a9664900ebb5c486f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/f3b6c571b834f4574e3a8a4a9664900ebb5c486f"}]},"branch":"refs/heads/master"},"4667c80b90ff686c6dd16ae6c109753a55b90a4e":{"kind":"REWORK","_number":2,"created":"2021-12-03 10:40:50.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/31/820231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/31/820231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/31/820231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/31/820231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/31/820231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/31/820231/2"}}},"commit":{"parents":[{"commit":"4d4366ed164f298e03da7b59cdff210f1d393cf0","subject":"Use config_template as a collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/4d4366ed164f298e03da7b59cdff210f1d393cf0"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 18:19:33.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-03 10:40:46.000000000","tz":60},"subject":"Database connection pooling improvements","message":"Database connection pooling improvements\n\n- Implemented new variable ``connection_recycle_time`` responsible for SQLAlchemy\u0027s connection recycling\n- Set new default values for db pooling variables which are inherited from the global ones.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\nChange-Id: I28c64b44eadfd726e07cb7159e5d3d94fde106ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/4667c80b90ff686c6dd16ae6c109753a55b90a4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/4667c80b90ff686c6dd16ae6c109753a55b90a4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
