)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~880397","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~Id5c18a7305c744a2b0252f62debb1b5654e4abd7","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"tls-backend","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-05-03 11:15:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5c18a7305c744a2b0252f62debb1b5654e4abd7","subject":"Add TLS support to designate backends","status":"MERGED","created":"2023-04-13 22:50:07.000000000","updated":"2023-05-03 11:16:15.000000000","submitted":"2023-05-03 11:15:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880397-tls-backend","meta_rev_id":"3ab48d319fced42e49368d534ef56ad9d9e9c5c3","_number":880397,"virtual_id_number":880397,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-03 11:15:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-05-02 10:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-03 09:23:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-03 09:23:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 22:50:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-02 10:46:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-05-03 08:52:04.000000000","updated_by":{"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},"reviewer":{"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},"state":"CC"},{"updated":"2023-05-03 09:23:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"1242e5868e5b59a0b0aba9493275d16b9a70d3b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-13 22:50:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b19576fa89a5011b168dcd9373e00770af309c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:50:24.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/openstack-ansible for 871189,32","accounts_in_message":[],"_revision_number":1},{"id":"86a4d17764a7bd5d381b64870f33390989d79c88","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-14 22:07:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f52b8ed4b0ffa0eff9279cdcef1a4e3ed540c1a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-15 00:48:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7af92b62bd45cbb36be0d174908928\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61c88386bc824ee481c863b32fc3c8b1 : SUCCESS in 9m 33s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/91b03393c8a64df4b596de906be6bc8e : FAILURE in 29m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02da31b32b8c4014b54b98e80c51e40b : FAILURE in 26m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/66b4c70fab3e4a8e84c7d10845197628 : SUCCESS in 1h 33m 26s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/aae80c44a847463ba383b73f726f0f84 : SUCCESS in 1h 18m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1889fd80216b4ca3b6dc29d85196e0d8 : SUCCESS in 1h 32m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a7ba1f7b3e44ecbb688cfc4718c7322 : SUCCESS in 1h 53m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd3deaf93e674e42897a5ecb84454901 : SUCCESS in 1h 32m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2acd7cfd7374ec39c410ff21631390a : SUCCESS in 2h 26m 40s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21088048c84340e188cd77625ec69173 : SUCCESS in 2h 33m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e586b936b00042d0a47e4a0d3250d0c7 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b2e36537dda48ce09d457f86eabd0c16b57c47d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-24 22:01:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fe896a59bf7bcad56dd4d94d783bc94e0ada0c96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 00:41:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ba2c9a4656345b9b8c2faa5ea1d1831\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae87c4013dfb4e3bb0f131f52085bffc : SUCCESS in 14m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/86b29c8354834563906a96b754553fef : FAILURE in 31m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a6dda32fccf4476afd83a66a5120a82 : FAILURE in 31m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f0aeba8310a94bcda8d03b332987601a : SUCCESS in 1h 32m 11s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/136bc0330bd54a43a4067e70bf336705 : POST_FAILURE in 1h 53m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b6faeab614654a63a88aa49c5cdd06d2 : SUCCESS in 1h 45m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cc898da96c04c7caf6552feea20230d : SUCCESS in 1h 37m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7c3cb2e65ac466e8b1e5618ab51477a : SUCCESS in 1h 36m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d3bfde8d7c1949489b01afefcbfaa167 : SUCCESS in 2h 23m 54s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fd8582ed5a44aa598801c35db07107e : SUCCESS in 2h 18m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe1f1f8b324144bf8b2d6ecbfdf89bea : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"b15ec3814410ae153d03f5345945d7f73e7de4f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-29 16:36:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"71a7fa0261acfcd1cc0efca42b0c02c38405c378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 18:35:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fff4173efee4957b0a07ae066acc8d7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/746a1fdd0f1c4de7920c6ce97bcb298b : SUCCESS in 12m 34s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dc1652789f244975a4047a093be26900 : FAILURE in 56m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d56812d3f29645058baa6da3c4d7ec65 : SUCCESS in 1h 06m 50s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/827f103ea74e4116a0436dca5e5acf13 : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ecc5b304412d457fbfad9e45c9435989 : SUCCESS in 1h 31m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ceafda760aa4eaeb77da263920962c7 : SUCCESS in 1h 47m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d34cdf2c4a7743d5a3b1202bccdf3e19 : SUCCESS in 1h 35m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/515a38734ab54c46b5e6c0224bcfd3c6 : SUCCESS in 1h 41m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ee909d993bd47fda4c363fbb5877111 : FAILURE in 22m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8ef5c867fde4e68ac06f0a1c088f5be : SUCCESS in 1h 51m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a6997dfad8e4a29a36baf371c4d2d0d : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":3},{"id":"73b8664c99235968d95a0b055bb69a39de8c3b1c","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-29 23:30:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f64c526cfd2f4fa02306200cba2d5088230e0953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-30 01:30:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45ffce36260c488790d26cf92c0caa6d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15fe157d77f64e99b80833041ed8171f : SUCCESS in 11m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/abddd7c4452843c592710b37c5d379f0 : FAILURE in 54m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca1f986129674e4e829f8387aa03938f : SUCCESS in 56m 49s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/009e379bb58347d5a1e250c31dedefe6 : SUCCESS in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a54e7609f9b148a284347df71fd97b3c : SUCCESS in 1h 40m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a649a8849b3846f1913a6fe596b3aedf : SUCCESS in 1h 38m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b80a4058d7054b7e87114f0b14b912c0 : SUCCESS in 1h 26m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80991011fb7b4190947ccc2e925fde00 : SUCCESS in 1h 44m 14s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4c4381f1dffa42b39a78cc82bdd704c0 : FAILURE in 19m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ffb799b09424ab39b3536f3943883fb : SUCCESS in 1h 48m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8184978fa9e948f4be6951419b9c687c : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":3},{"id":"bc201de663d3ad890c1ceb9bb04c559ea0d57303","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-30 09:31:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea7f5921f19c634718fa1096da4564e64c7d1d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-30 12:09:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43ab55f11eef4e8e8f0925075b0081a6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43f8f891041343d68c54fff3a676965d : SUCCESS in 9m 18s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48ed906b55ab4b54ae36e0100c6c2f73 : FAILURE in 50m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35f6bc51138845548c8fb8d4106cfefe : SUCCESS in 1h 10m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d61b79420d534da99695e963c23129a2 : SUCCESS in 1h 43m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7acfe94748cc4de29524bf81c43a73f2 : SUCCESS in 1h 33m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ee3a1c58f1084f7aba09b293d6e69382 : SUCCESS in 1h 36m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbf01d2eae9748428861925c7debca7e : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/650ebf04141243e5aa38a67d259d92f9 : SUCCESS in 1h 19m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5decefbe9d5a4cafab2cd5c03a54f127 : SUCCESS in 2h 19m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f097bf9f4500480493a8f893f6e1f2de : SUCCESS in 2h 09m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c052c01ffe457092b359ecced2c84d : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"8d68e83f9bb6ba21b5864858997ff54ff2a9241b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 10:46:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b2596d33d2e09f047157bc6bcf2084e8e2bb7e84","author":{"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},"date":"2023-05-03 08:52:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"352edaccc3d3e90b297e2d7dac516ad81f680c27","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-05-03 09:23:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0ea049ba8fcf3e2eb76f1a4e7ad4200d3531683f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 09:25:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0c91ddb5947f5a830f181f76ec789d537b80bd86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 11:15:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af2736ecd771410ca91227a8aea59bb0\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4717aa4992004f19856afa932d9d34d0 : SUCCESS in 10m 43s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/49e42c9a7e1d4178ba251a516b3d696c : SUCCESS in 1h 36m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/02a3acbc0275449e97444258176ea37b : SUCCESS in 1h 45m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba1d16f1d76b42f6918dabecd7b0c68c : SUCCESS in 1h 48m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/084bab9a271544a2a514a1f56f7e3f70 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"e476aa9dc2f63118e7ef9093c441394008bd5859","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 11:15:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3ab48d319fced42e49368d534ef56ad9d9e9c5c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 11:16:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c2c456d186349c69495ab8230bf7f0f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f670d326c904daaab2c14c25be32faa : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"34649668686fa19e1d11eb104094732eb23c0a25","revisions":{"bb83a23e0eceacc617c4ac60a7129294241dbb4c":{"kind":"REWORK","_number":1,"created":"2023-04-13 22:50:07.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/97/880397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/97/880397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/1"}}},"commit":{"parents":[{"commit":"b6d826f675efa50770d1339add61dfcf11a13dda","subject":"Ensure service is restarted on unit file changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/b6d826f675efa50770d1339add61dfcf11a13dda"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-13 22:49:44.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-13 22:50:05.000000000","tz":120},"subject":"Add TLS support to designate backends","message":"Add TLS support to designate backends\n\nBy overriding the variable `designate_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the designate backend api.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: Id5c18a7305c744a2b0252f62debb1b5654e4abd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/bb83a23e0eceacc617c4ac60a7129294241dbb4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/bb83a23e0eceacc617c4ac60a7129294241dbb4c"}]},"branch":"refs/heads/master"},"6d0df5ddba084009bcc448fc127942acb4ddda3d":{"kind":"REWORK","_number":2,"created":"2023-04-24 22:01:25.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/97/880397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/97/880397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/2"}}},"commit":{"parents":[{"commit":"b6d826f675efa50770d1339add61dfcf11a13dda","subject":"Ensure service is restarted on unit file changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/b6d826f675efa50770d1339add61dfcf11a13dda"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-13 22:49:44.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-24 22:01:23.000000000","tz":120},"subject":"Add TLS support to designate backends","message":"Add TLS support to designate backends\n\nBy overriding the variable `designate_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the designate backend api.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: Id5c18a7305c744a2b0252f62debb1b5654e4abd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/6d0df5ddba084009bcc448fc127942acb4ddda3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/6d0df5ddba084009bcc448fc127942acb4ddda3d"}]},"branch":"refs/heads/master"},"34649668686fa19e1d11eb104094732eb23c0a25":{"kind":"REWORK","_number":3,"created":"2023-04-29 16:36:35.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/97/880397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/97/880397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/97/880397/3"}}},"commit":{"parents":[{"commit":"b6d826f675efa50770d1339add61dfcf11a13dda","subject":"Ensure service is restarted on unit file changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/b6d826f675efa50770d1339add61dfcf11a13dda"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-13 22:49:44.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-29 16:36:33.000000000","tz":120},"subject":"Add TLS support to designate backends","message":"Add TLS support to designate backends\n\nBy overriding the variable `designate_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the designate backend api.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: Id5c18a7305c744a2b0252f62debb1b5654e4abd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/34649668686fa19e1d11eb104094732eb23c0a25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/34649668686fa19e1d11eb104094732eb23c0a25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
