)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~881975","triplet_id":"openstack%2Fopenstack-ansible-os_designate~stable%2Fzed~I3dcb02cfd8c5057411488dfefdd0b5381231a972","project":"openstack/openstack-ansible-os_designate","branch":"stable/zed","topic":"cherrypick-osa/systemd_restart_on_unit_change-hbqfsn1p95","hashtags":[],"change_id":"I3dcb02cfd8c5057411488dfefdd0b5381231a972","subject":"Ensure service is restarted on unit file changes","status":"MERGED","created":"2023-05-02 11:01:31.000000000","updated":"2023-05-02 20:55:32.000000000","submitted":"2023-05-02 20:54:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881975-cherrypick-osa/systemd_restart_on_unit_change-hbqfsn1p95","cherry_pick_of_change":879960,"cherry_pick_of_patch_set":2,"meta_rev_id":"1b84f197a866215c459e81ab2ecbc40f52a77de6","_number":881975,"virtual_id_number":881975,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-02 20:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-05-02 17:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2023-05-02 17:48:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 13:42:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89b02e9b9260f8c5e1cab3941b8a16e358525d7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 11:01:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b8c8754bfdd7ade53e83cc3b41e6aaa3a69f8607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 13:42:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b813c098c620458eb0bba501a1a815e7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c8d87b6699b4085904a59f3564db369 : SUCCESS in 11m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9c731887eba4774836275cc93d344b4 : FAILURE in 44m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9ad3edbac374890907b30819e7aaf42 : SUCCESS in 1h 09m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7a5864c41080454fb36e22ded6020c95 : SUCCESS in 1h 17m 47s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cba3badf0a2d468baf3e8cecc9c4af7c : SUCCESS in 1h 21m 59s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b5c6a4c91ba4494ea17e12103156a731 : SUCCESS in 1h 27m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95a9fe5b3bf94a50a3e402d35a0e8eb5 : SUCCESS in 1h 28m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cb5960a8e5e4f63a0667c9848ab1f80 : SUCCESS in 1h 22m 38s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/504c09f8b96140898f307f64d8f4080f : SUCCESS in 2h 28m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e8e664c0c90407ab489624bdf6d08d4 : SUCCESS in 2h 03m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36179c6cbe141c8be4acc86b355462c : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ef4d35e5cd629a2d9f003173780e3a8710c79d7a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 17:48:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"002fc6fc0b025a2d29fa746e4af9603355ba7e86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 17:50:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7de5d54aa3b63294221740e406e4d3606bac286","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 20:54:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a10bcbc9c8b407bbe1ec1c17a2b55cf\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cab5cbae9f68435bbd31b1bc84506af4 : SUCCESS in 11m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6ef9d9f6a2534daf9a28fa55747090dd : SUCCESS in 1h 35m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/276341e203f1401090db14e17bb3e03b : SUCCESS in 1h 35m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d871ec4fb99e4fe28e2c1b3963e99dde : SUCCESS in 1h 32m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00747ede240041038915d4025d7c4782 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"dc06434ff646d0d0bc2bbbe8c9e5fe2a9455aadd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 20:54:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1b84f197a866215c459e81ab2ecbc40f52a77de6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 20:55:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7047eb9ae1024cda8fc7b9169f05671f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb676214b0c4480cb8fb8e19a57b7534 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2d95f99585442ac6a9a31818ae4ce8a41546394","revisions":{"f2d95f99585442ac6a9a31818ae4ce8a41546394":{"kind":"REWORK","_number":1,"created":"2023-05-02 11:01:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/75/881975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/75/881975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/75/881975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/75/881975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/75/881975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/75/881975/1"}}},"commit":{"parents":[{"commit":"b40ec599fa882f9650177df01f8d47c86731933f","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/b40ec599fa882f9650177df01f8d47c86731933f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-10 13:54:50.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-02 11:01:31.000000000","tz":0},"subject":"Ensure service is restarted on unit file changes","message":"Ensure service is restarted on unit file changes\n\nAt the moment we don\u0027t restart services if systemd unit file is changed.\n\nWe knowingly prevent systemd_service role handlers to execute\nby providing `state: started` as otherwise service will be restarted twice.\nWith that now  we ensure that role handlers will also listen for systemd\nunit changes.\n\nChange-Id: I3dcb02cfd8c5057411488dfefdd0b5381231a972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/f2d95f99585442ac6a9a31818ae4ce8a41546394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/f2d95f99585442ac6a9a31818ae4ce8a41546394"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
