)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~898818","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I8c418906b75edb633948f2c074170454a8f3e2d0","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"osa/quorum_queues","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-11-02 10:03:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c418906b75edb633948f2c074170454a8f3e2d0","subject":"Add quorum queues support for service","status":"MERGED","created":"2023-10-19 12:31:00.000000000","updated":"2023-11-02 10:04:33.000000000","submitted":"2023-11-02 10:03:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898818-osa/quorum_queues","meta_rev_id":"a64c3d3624a7539fd3eb6a3065fbdf7c85ba1203","_number":898818,"virtual_id_number":898818,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-02 10:03:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-27 10:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-26 20:41:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-11-02 08:07:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-19 14:35:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-22 15:43:28.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2023-10-26 20:41:34.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-10-27 10:57:30.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"2cdf4ef58f6b0b1dec2600a233a0e29b5a68c74c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-19 12:31:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"583e8ddf39bb50d58a4de7677eb7bbffb4b85b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 14:35:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efea66755bc541af88ef683eb985a68f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57419fa3026b4e11a07e7d1978292628 : FAILURE in 12m 53s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2690d72a5d9543e09e8e5aa614ed96a7 : SUCCESS in 1h 24m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbafc3b65a0e4c5cbf143714aad569c7 : SUCCESS in 1h 02m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/485476a57a3b47008a145217fd2ea4ce : SUCCESS in 1h 50m 13s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/07d3300bb8554866af7d82a20930312a : SUCCESS in 1h 57m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b101a31c0f44b1286ad161546dd8bb6 : SUCCESS in 1h 52m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/275f9a781cfd4ce1b31366ea0fb94fef : SUCCESS in 1h 43m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cbe338904b2f4d58975af4c9225fec4e : SUCCESS in 1h 40m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f4dd3540a46409ea8b0ca1a44e38b3d : SUCCESS in 1h 30m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad581410cf1d45919045792d41fbf8c5 : FAILURE in 32m 53s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cda87a457d1e4722947eed61f464aa3f : FAILURE in 28m 56s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac0e5b240e404bf39110ef79963a8131 : FAILURE in 34m 44s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1db114f9eb4497fb8025d9d25502747 : FAILURE in 30m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96908af5d2c84be49e926a96b19dc227 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d110b8ca1209a21d7f3c1f104be6cb4ea0f682ba","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-22 15:43:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc6879a29cd062b7e5c6d0e0faa1048a217a65aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-22 18:06:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eca9954aff0b48af9916c5188b3c102b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd79984cf8054b879f6a2db0f3c44768 : FAILURE in 10m 57s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb62273371304014b9271184203ce9b3 : SUCCESS in 1h 15m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b15e87a205ae4f3fbe4e8218058e8c19 : SUCCESS in 1h 11m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9a8ee427f11d4ccc906855a31179e5bd : SUCCESS in 1h 52m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3ad857281104c04832039084919f6b6 : SUCCESS in 1h 58m 20s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5327c2282e1449849cbea921155dd15a : SUCCESS in 1h 59m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/227842bae3ee415d8c640bca51c8863a : SUCCESS in 1h 47m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dcb77e0775624c12ac3f12d0fc355dcc : SUCCESS in 2h 14m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b78cacb1ba94c419aae6dd7357b8334 : SUCCESS in 1h 53m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/981c5e45698847aaba69064ec2c30ce6 : SUCCESS in 2h 08m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0099211ec18342a9a88d80aa6c502f96 : SUCCESS in 2h 02m 33s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/927542d9037841dfa64934100c97ea41 : SUCCESS in 2h 06m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce6cc6deff714b52919e08961f3dd179 : SUCCESS in 1h 50m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a6c65466214b0f95f9fc62a75d4b6c : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6d6271cd4f6ded12c3f70de0498c3f5fd4a7244b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-24 17:10:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6dbf383a76773d0eb4f10fc665e559d5373e5f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 20:23:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4006eccf30aa40c98d14954c53a63a26\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e2e96ed47cb4e3d9770fd9a2204bb6a : SUCCESS in 11m 35s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0144a5a05e904070be92c746762a6c70 : RETRY_LIMIT in 8m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60161a1649cb401191d764a8c0d82369 : RETRY_LIMIT in 7m 56s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e1fe5a281ae9459c9b257d88c893a581 : RETRY_LIMIT in 13m 48s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/92a7e240819a49faa85b3b5ed51027f0 : RETRY_LIMIT in 18m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d57cc7d78dd94ef6a8081e89a1f28788 : RETRY_LIMIT in 15m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33dc5e57179c42b7baf87f459b6a8e65 : RETRY_LIMIT in 17m 03s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5e306b246a3146d6934d0cff418b28d8 : RETRY_LIMIT in 19m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f2aa4d331f7412cb0370f51ae3c0caa : RETRY_LIMIT in 11m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a999260763734379b4cb111f7b1dca32 : TIMED_OUT in 3h 04m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9448a9c7a74b44bf90b8d5c07131fb85 : SUCCESS in 2h 52m 26s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7e89f2eba87649178a10062f6708045a : SUCCESS in 2h 53m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aaf8069fe59140338b71302c2f48514d : SUCCESS in 2h 43m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2939f2b2ebc74ecabe851154f4259ed1 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"47bbe05cd0b417dfd3410734e3bd0fbdd7a9f0bb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-25 08:58:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"312194d69c26e0640253096a96a81c991ceb5bdc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-25 10:34:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa26164409ae52fdbb267eb692113c2bef8e67cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 12:05:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6da2cd00855346b48b5259d0cfefea42\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0422bf30a9e40dc92efaefb757a370f : SUCCESS in 12m 54s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d19e82890cf6469bac4256688ecb83b4 : SUCCESS in 1h 08m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ffa0e48c37349eb90f95f4f4a00d726 : SUCCESS in 1h 02m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/61c9a267c4a743bbb9a38bb02af2d6a4 : SUCCESS in 1h 26m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6924a2284834b6f8725432d3e7987b4 : SUCCESS in 1h 40m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b1ad7916077249839cc3216fcccc4d54 : SUCCESS in 1h 40m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0ac1f7a990449d28af65a29c90de4e7 : SUCCESS in 1h 41m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/202219feedc047219900ba76fec75c55 : SUCCESS in 2h 03m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b07e43063774d8a9ce32de01d8efb33 : SUCCESS in 1h 30m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1a5684114a4e4a2898c257a0f470be71 : TIMED_OUT in 3h 04m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/359db8d32de747cca4f9a3fbf49a6b19 : SUCCESS in 2h 35m 58s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eb3d150fbcd1403da276e7d5b9dd4c30 : SUCCESS in 2h 42m 04s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f4992f578864ff19b9782295a4a3549 : SUCCESS in 2h 48m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e27cbeb94445259286952203a9529c : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"1b73bb46d577f395bb57711c5f2d168662166047","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-25 14:51:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4608a7d49b3f960ab83e3ae8215155b7abc83b2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 18:05:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/425bd11bae6541018fc770bf5b8da139\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29c1c50d67c5416caf58e81551fbed37 : SUCCESS in 13m 21s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2128b05e922f4f3d92ecf5d12a24e692 : SUCCESS in 1h 18m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70719374a5a646a5b5c2e57ff75926a5 : SUCCESS in 1h 14m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8c7d5ccb80bc47eaa5e88cdeb37902cc : SUCCESS in 1h 44m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f95e123a2f02478d80f34e526513371f : FAILURE in 32m 34s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac99784df118464ca9d00854b2d551b7 : SUCCESS in 1h 57m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e3838336c067497c8d4f40dc661d96e9 : SUCCESS in 2h 00m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22405eeefc024ec1a68ce1ed67eb3532 : SUCCESS in 1h 47m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d6e696a5ebec4650bc8060fc52c1f778 : SUCCESS in 2h 21m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0639f4a27f9d4571b5c6209aa54fa167 : SUCCESS in 1h 58m 20s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac0f95b2e6ea4862966d85fb0a43b2db : SUCCESS in 2h 47m 54s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7ae43a24239491b88786a8b1d2a70a3 : SUCCESS in 2h 17m 57s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d3a3e4bd3a514cdc873d139d43a7e2e0 : SUCCESS in 2h 46m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74f762d9126443d6a5b08d67c77969c2 : SUCCESS in 3h 03m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a491a55db6bc441ab47766c3e550706f : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"503561167999e09e130a2c4fbbf16b86d11f8717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-26 16:08:04.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b1dc091c02703da1b7b9f2bffdf1c2410957ee52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 19:13:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/661bf819e73a432cbc2789508c28f507\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db47afc3bc22443886696e7aa80e4101 : SUCCESS in 14m 23s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4635b9a171ae4293adf38c925f45218a : SUCCESS in 1h 06m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f14cff0151114b799d9043c9827baf78 : SUCCESS in 1h 16m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/400a95de08fe42feb26c69397dae50c0 : SUCCESS in 1h 30m 19s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/dba3a1bdc61e445aa9163613deabf19d : SUCCESS in 1h 39m 48s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/852cf8e2861145b3bd53577c58408c67 : SUCCESS in 1h 36m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/674ee1e188ec49ae94f85fb0b73b5319 : SUCCESS in 1h 46m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b35d953c8a884f6faf0abbf65166928e : SUCCESS in 1h 39m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b081b569f69f4b19a82ac86b906726f6 : SUCCESS in 1h 44m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a04f6a611fdf4f2f9ec5b11be2ba9d21 : SUCCESS in 1h 25m 23s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/64b6eb86468d429e97861f1534bf818b : SUCCESS in 2h 47m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33061619400e44baa952e91beca8de1b : SUCCESS in 2h 31m 04s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/79ce295022a24d3da8b848a3ec70e21b : SUCCESS in 2h 57m 36s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1f4e3b004d446d5970a945101d1ad73 : SUCCESS in 2h 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466b1401c8ce4d6396f578516909a039 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"27a49153d429f6880016f3734f720aa772dcaf37","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-26 20:37:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b4ca2480aa83fa1c43d6ff842c18fd5210cb3530","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-26 20:41:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5970ea0d4f38119c4a5c64f634b227dc458cfeea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 23:29:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a2470b51d5e4ff8b737f13a173123bc\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c1c9abc8a1349a192daf3492dda93ff : SUCCESS in 12m 36s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6f9f6688902d4ac09457a69c5325c8b6 : SUCCESS in 1h 05m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2de3cd8371504ead8d74d18d3c688009 : SUCCESS in 56m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/31a2fc67f71e4e86b4edd3a93833166a : SUCCESS in 1h 12m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/96cdba2dddea4df1b714e2bf9a61076f : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/119f03c706be4263aa6c8b61492092d9 : SUCCESS in 1h 18m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a55cd4dfc4e45cb99afefa98a7847cb : SUCCESS in 1h 47m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f3e887fe31444ee90808e1117a39157 : SUCCESS in 1h 14m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c4d56bce0aaa4984924bf69eecce890f : SUCCESS in 1h 38m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b04556f44aaa487482e70ac13234b7ab : SUCCESS in 1h 16m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/336f9ee96dba4f189efdfeb3c9c1401a : SUCCESS in 2h 44m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/383484b7c9ee496691cc18e29b63aaa4 : SUCCESS in 1h 53m 39s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/af4e0c794aca4f40bc151f0d948fcbea : SUCCESS in 1h 59m 40s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/871779f98254453f8188a15935ef0307 : SUCCESS in 1h 58m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c491a924bd4814a9307d226af79828 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"b2b3d6603d2855253a037a35b85a304e6b42ecee","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-27 10:57:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5e1f6746edf260d9fec73f23672aeb8872907ee9","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-11-02 08:07:12.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ccd934a0621c6f273c317006963723d4a4b4923a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 08:09:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"956e09ab8a5fcb8b3d487b1c1d4c1563b9d3ceda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 10:03:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a469aa45c884664a0605f77060bfefd\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4179248ce1984ca5a32ded9e37eb7816 : SUCCESS in 12m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/19a8979886a44e4f9422217192b05910 : SUCCESS in 1h 16m 58s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c5efafb271ea47a5baa88291e946b77a : SUCCESS in 1h 28m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4a91025058eb4bc29401e33178b95e7a : SUCCESS in 1h 45m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e37f9880e122464c91bd166ef32e31ee : SUCCESS in 1h 47m 03s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/44e337b89d034d9d8af012f4620da88e : SUCCESS in 1h 40m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16113a449f5d4e02b8b6ffcc80b99a55 : SUCCESS in 1h 29m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed7b895998649e9bfad77c13d0609a4 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"559d984db7d0aea6b534334e7ad410e6eab5cb36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 10:03:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a64c3d3624a7539fd3eb6a3065fbdf7c85ba1203","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 10:04:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/406fa3026e924466a4ce3cc7e0f460f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1fa204d0d0645c0b9ef94da0731c1a3 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e7f7e7d122aa6c0fc8c99355f9888406b495c528","revisions":{"1575dd64cdc338542e6be031a704553832e31a96":{"kind":"REWORK","_number":1,"created":"2023-10-19 12:31:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/898818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/18/898818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/1"}}},"commit":{"parents":[{"commit":"db839d98c76809aca9b5aeeee2c24ce6e3776a94","subject":"Merge \"Use proper galera port in configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/db839d98c76809aca9b5aeeee2c24ce6e3776a94"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-19 12:30:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-19 12:30:56.000000000","tz":120},"subject":"Add quorum queues support for service","message":"Add quorum queues support for service\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nIn order to ensure upgrade path and ability to switch back to HA queues\nwe change vhost names with removing leading `/`, as enabling quorum\nrequires to remove exchange which is tricky thing to do with running\nservices.\n\nChange-Id: I8c418906b75edb633948f2c074170454a8f3e2d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/1575dd64cdc338542e6be031a704553832e31a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/1575dd64cdc338542e6be031a704553832e31a96"}]},"branch":"refs/heads/master"},"39b5a33b33524310eb4ea463930ba47aec22bfac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-24 17:10:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/898818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/18/898818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/2"}}},"commit":{"parents":[{"commit":"645fcde6024a889924e12c6a4ec4113d316248d5","subject":"Fix example playbook linters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/645fcde6024a889924e12c6a4ec4113d316248d5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-19 12:30:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-24 17:10:41.000000000","tz":0},"subject":"Add quorum queues support for service","message":"Add quorum queues support for service\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nIn order to ensure upgrade path and ability to switch back to HA queues\nwe change vhost names with removing leading `/`, as enabling quorum\nrequires to remove exchange which is tricky thing to do with running\nservices.\n\nChange-Id: I8c418906b75edb633948f2c074170454a8f3e2d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/39b5a33b33524310eb4ea463930ba47aec22bfac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/39b5a33b33524310eb4ea463930ba47aec22bfac"}]},"branch":"refs/heads/master","description":"Rebase"},"e7f7e7d122aa6c0fc8c99355f9888406b495c528":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-10-26 16:08:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/898818/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/18/898818/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/18/898818/3"}}},"commit":{"parents":[{"commit":"645fcde6024a889924e12c6a4ec4113d316248d5","subject":"Fix example playbook linters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/645fcde6024a889924e12c6a4ec4113d316248d5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-19 12:30:35.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-26 16:08:04.000000000","tz":0},"subject":"Add quorum queues support for service","message":"Add quorum queues support for service\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nIn order to ensure upgrade path and ability to switch back to HA queues\nwe change vhost names with removing leading `/`, as enabling quorum\nrequires to remove exchange which is tricky thing to do with running\nservices.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-zookeeper/+/899386\nChange-Id: I8c418906b75edb633948f2c074170454a8f3e2d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/e7f7e7d122aa6c0fc8c99355f9888406b495c528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/e7f7e7d122aa6c0fc8c99355f9888406b495c528"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
