)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~898906","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I11649981895a482cea97c373d2de9ca7df654cd2","project":"openstack/openstack-ansible-os_designate","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-10-31 16:31:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I11649981895a482cea97c373d2de9ca7df654cd2","subject":"Fix example playbook linters","status":"MERGED","created":"2023-10-20 08:47:51.000000000","updated":"2023-10-31 16:33:00.000000000","submitted":"2023-10-31 16:31:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898906","meta_rev_id":"e6306a928be879999f7e6208c47e3a2b39c1818a","_number":898906,"virtual_id_number":898906,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-31 16:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-26 07:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-26 07:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-10-26 07:52:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 11:01:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-26 07:49:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-10-26 07:52:08.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"7b96e4fb7482fa19f9dd7eb3ac627d9e1eb41956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-20 08:47:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86d83ab6d771a02fe96d328da4a46afc6e517e3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 11:01:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/917e28c858324c8aa46e0d6c0e217a0d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d9a27d096fa4dccb0ebba14de3c211b : SUCCESS in 14m 11s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8d8e8fdeac7840759c2d7b44ac035694 : SUCCESS in 1h 07m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdaac7e4fbbc45a2a5ebb209af8f5730 : SUCCESS in 1h 08m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2ed5594d55554ba3bee19053a5a42961 : SUCCESS in 1h 21m 37s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cecbf031e02c4207ac590bf323fb746c : SUCCESS in 1h 55m 43s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/209bdd2836e84a7b863ddee0306770d7 : SUCCESS in 2h 06m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/243e599bf5744734bd6d32913a727b9e : SUCCESS in 1h 34m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/337fd5b330ec43dd8192061166a4b469 : SUCCESS in 1h 55m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3a9f487e12042d9b134c9b8562936a3 : SUCCESS in 1h 43m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/246bfbed0c1649a79f9d2e5e8efc1a87 : FAILURE in 32m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93aa24e4960f471daff7dd291b4aec89 : FAILURE in 35m 48s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b83c1e37a6f04d84a6e538143a263130 : FAILURE in 31m 26s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58789afaaf764843ba929e530e049831 : FAILURE in 33m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b425e3750ec746eea1a5c57706eac935 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"54d2a348cb0b4a50cc5486323ec738223f9961c3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-24 17:10:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9c6be817efff3f3543ed3822b984a7e89736e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 20:21:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e372366a63034349baeb4f59c7d72ace\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a54d3d4cb32b4de791cd5244dfb43969 : SUCCESS in 12m 38s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0d3053330bb141bab58e37d0c4aa0bd4 : RETRY_LIMIT in 8m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd6b2ca980234fb8b603d70b5b1b8a8a : RETRY_LIMIT in 8m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/daec8542fe3a4436a5041cf4723a5764 : RETRY_LIMIT in 16m 40s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfe1fd653bfe442b948f6d19a9d1914c : RETRY_LIMIT in 13m 46s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/403bd5a92a5f428386ec2414af29adfc : RETRY_LIMIT in 12m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b17fb5160b3d455db93d49f96a96b0b6 : RETRY_LIMIT in 11m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3278ea135a4541a7af39065541645cf2 : RETRY_LIMIT in 13m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59d7d29fbe8b4714a0be804504089f20 : RETRY_LIMIT in 13m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf7f76ea3efd4925a3fdbf7cd6ab92e1 : SUCCESS in 2h 41m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2fe517d8f5143959adbf9249dce75c9 : SUCCESS in 2h 52m 54s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/af7e3c372d4140bf8cb889f240f4dd6a : SUCCESS in 2h 37m 04s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6002074871914aa1be7ff444786754c3 : SUCCESS in 2h 19m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89912494091e4ff3b7bad30ad324ca16 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ada3cfeb927bcfc64f92b6d1e5bc38e71ae0b00c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-25 08:58:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"542bf62f3938a584b10350297b7c3ce6c3d311d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 11:50:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d7444a6e47e4bf6b400d1176040b615\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31cb21b3cf7a4df4aa465e258088cb91 : SUCCESS in 12m 56s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dddfe684c3dd4badacce01a5e932587b : SUCCESS in 1h 14m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f969247b744342dd84aa3b19a8424e63 : SUCCESS in 1h 13m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ea416ab6c7d345bfa7af94e02f5fb989 : SUCCESS in 1h 47m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb196ffadb244cdca70da4d37a615a5d : SUCCESS in 1h 56m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e751ae625be744469bc1baa5e41d6b45 : SUCCESS in 1h 45m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3e8471a4d59443e8f9859d9d43695cd : SUCCESS in 1h 30m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/520834471784457d9606bb50a81a52c6 : SUCCESS in 2h 08m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f224170c9d445d5960bb1b1d3019ddc : SUCCESS in 1h 39m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9b1331ceaee9468b893ba2e1fc674c2d : SUCCESS in 2h 38m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b811beb181ef4b23b09306b5b439fb01 : SUCCESS in 2h 42m 40s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c38a5bba093b48cea4e83737f8788dbb : SUCCESS in 2h 16m 41s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27c91c3c8efa4fdb803f33691124bf10 : SUCCESS in 2h 15m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a204a2354cee46cfb004965aec83159c : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"91d566636570850594797794c28b92c705029549","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-26 07:49:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"23b04308e008e8c8d0e1e26b3bc68db8f4ba2913","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-26 07:52:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5344cb78374e1170bd198bba337c9dfa8786f524","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 07:54:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c65b9b2879db1a134420a1ea1bf23a7d1cf429d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 10:09:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcbbfb6afa7e450ab0db8ce1c82c8fad\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2f5849d90264b5b94d64e12a98565a8 : SUCCESS in 10m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/77ecd3a50bb64010bfba9f637c883d52 : SUCCESS in 1h 34m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/75e1360c4b9e4331a3428787fac021dc : FAILURE in 32m 19s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ab44794c62040d1b2bc16970b8b94d1 : SUCCESS in 1h 48m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f503e59f55344c2090c765a4e31940e7 : SUCCESS in 1h 48m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/13c7f31c45354ed080c2b88f2747421b : SUCCESS in 2h 08m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e56bc4d305f94c78bfa34ab1d9f7de7e : SUCCESS in 1h 47m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3243c58feb6947aea66b98fc8d65cd7a : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d9df726d4ce7fb44d8892968bac473b040a5df83","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-26 20:38:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75972dd097c17dee4db6d9b9fc82bd77d22fb605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 23:35:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1846b5d95f424bd4bf3f6a5b75a39013\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e89dd40a54924e4387b7ebfec189bd8d : SUCCESS in 12m 30s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7fcc80ddebfe4c63b7d1ccbe46fc84f9 : SUCCESS in 1h 03m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ad25e21ffc24611ba9019d587b666b7 : SUCCESS in 1h 08m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e4846726141f45329342d6e2fdc526cf : SUCCESS in 1h 33m 01s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e4d84513d6ca4a799974dbb21e886fb3 : FAILURE in 29m 23s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebae600bdede446282addc3e856826fc : SUCCESS in 1h 34m 07s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c815effe8b9b4110a0fbfabfd25d417d : SUCCESS in 1h 33m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3b91003366448c681d6988c6395a3ed : SUCCESS in 1h 43m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ed49fdf3d162444ebbc9c64e6a647418 : SUCCESS in 1h 28m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0790875843e84129a359255aa01c6a8b : SUCCESS in 1h 22m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/68634d274ea34439952b6b9449b72d77 : SUCCESS in 2h 50m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e650a1eb020a4232b8a15632ec778bac : SUCCESS in 1h 50m 09s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c47d946550e84f429135166d8a5f4483 : SUCCESS in 2h 48m 46s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e805ed07d29e45e491b60a453977c42a : SUCCESS in 1h 54m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b86eac3ce2e7481fac70eed586b62230 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d1450e9c604519a10f183362f5e2f10d0417237e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-27 15:15:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b64262cef60f9bc28c8e628e2fbbcf663e7221d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 17:30:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b99866433d4e4fbdba18de963861b29d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4256fcb05bfc4c79a8ed5deaf51ec5cf : SUCCESS in 11m 42s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/209d4f490b3347ceb6b5cadd43836479 : FAILURE in 48m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c8d3d395c7246f9b8430f21615f2327 : SUCCESS in 1h 07m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/870374744c124df192f387fc65a08abd : SUCCESS in 1h 43m 07s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bfb85902a90a4e1099c21faba2695eb4 : SUCCESS in 1h 37m 33s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b0f45a2f3b347c0bdfa9e5c5b32a7fe : SUCCESS in 1h 28m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/abfbb363404e4b369b677dc3ccd6f3d8 : FAILURE in 1h 07m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb63db77b47441ef91711b3fa76f15d5 : SUCCESS in 1h 32m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/02b347ecffc04be2908ad430d483874b : FAILURE in 56m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d139b46ee2b04149afc887cb2742ff9d : SUCCESS in 1h 32m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0755c813000643f59015e1d9af8b60b2 : FAILURE in 56m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59fad7d8ce1e4efab00f178f3a2417c6 : SUCCESS in 2h 11m 22s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/331e73281e414b45831c62185d2164b4 : FAILURE in 59m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/115871888a614051b1abbc277b313847 : SUCCESS in 2h 05m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e9e10b71cd494b9c837703b6d57d16 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c45c1364f2fdc05aef32fe964ecbbea6187e424f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-31 11:55:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe0d32a69df52f67752180329fc281fd150db925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 14:33:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/969e53581c7541a3bd4a618c0d8a068e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02d158b82c17436db3b8f2fd0d9de6fe : SUCCESS in 13m 37s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ceced7f43379418195cd9774e5eaeb42 : SUCCESS in 58m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0c80005d70e4782b30b32920cc0cd0d : SUCCESS in 1h 11m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7655d95c85e44a49844042bf257bf923 : SUCCESS in 1h 15m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d7ebba2a0ea04c4bb72044e72f2ba29e : SUCCESS in 1h 36m 46s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/06d5dc3335774229ab832d3224c0f454 : SUCCESS in 1h 45m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8e1f94635dc147fe862a603814135d13 : SUCCESS in 1h 34m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e604e6c448d042c0a54928aa266de94f : SUCCESS in 1h 31m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39ad8ebdc5bf472da97c91a1c7118003 : SUCCESS in 1h 48m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/356c4ad72884454797d9f98e13a4eb1c : SUCCESS in 1h 41m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3e53b7a4e07247dbb96c0484a136c862 : SUCCESS in 2h 30m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d2a8835b0a54a728f5f54fa1610d211 : SUCCESS in 2h 14m 32s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/054bad92d3124fdb944331728334eec6 : SUCCESS in 2h 24m 07s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ca41e41957d49f6af37fe55e3dbfba3 : SUCCESS in 2h 23m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256d3adf14ef46a0b0cc0eb724a55a09 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"201a4a758beeef0898d8f1fca93ef11c4cd0704b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 14:35:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19d2d143b2821a37d44e5408c1da03f27c283098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 16:31:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6084705bc7c342949d88ce5dae8bb75e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7acdcb51518426d9b0b5c8309d79ca9 : SUCCESS in 12m 49s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8311c24fd15844139fad8f97c884dcd6 : SUCCESS in 1h 32m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/50b6437eda8f48c99307b5e6892a191b : SUCCESS in 1h 38m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/251db1f7f80f4f64a167eafd5f605f84 : SUCCESS in 1h 32m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1a651c9cd6143ec8e582c6828aafaa2 : SUCCESS in 1h 24m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/875b0255ce7c4a8182de3cc49062c680 : SUCCESS in 1h 46m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccdad98e56e7465289e558b1fc30021c : SUCCESS in 1h 46m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bff048bcd4824f038c7faf6149b1f4ae : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c1a4c7f56422bc10e0eaf8b0ea9236d2355fc7aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 16:31:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e6306a928be879999f7e6208c47e3a2b39c1818a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 16:33:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63858eeeee8545d482cbbf19130a15f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32287196ca954828bd728834867b2b04 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"645fcde6024a889924e12c6a4ec4113d316248d5","revisions":{"645fcde6024a889924e12c6a4ec4113d316248d5":{"kind":"REWORK","_number":1,"created":"2023-10-20 08:47:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/06/898906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/06/898906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/06/898906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/06/898906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/06/898906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/06/898906/1"}}},"commit":{"parents":[{"commit":"db839d98c76809aca9b5aeeee2c24ce6e3776a94","subject":"Merge \"Use proper galera port in configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/db839d98c76809aca9b5aeeee2c24ce6e3776a94"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-20 08:47:43.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-20 08:47:43.000000000","tz":120},"subject":"Fix example playbook linters","message":"Fix example playbook linters\n\nChange-Id: I11649981895a482cea97c373d2de9ca7df654cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/645fcde6024a889924e12c6a4ec4113d316248d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/645fcde6024a889924e12c6a4ec4113d316248d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
