)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~918112","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~Ia29caf1633d5467e870226763088065fde54f12d","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"osa/messaging_improvements","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-09 11:43:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia29caf1633d5467e870226763088065fde54f12d","subject":"Add variable to globally control notifications enablement","status":"MERGED","created":"2024-05-03 17:28:59.000000000","updated":"2024-05-09 11:44:34.000000000","submitted":"2024-05-09 11:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918112-osa/messaging_improvements","meta_rev_id":"4f7472a7732604d2d97e68e1ce0b2d8ef58fba19","_number":918112,"virtual_id_number":918112,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-09 11:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-07 13:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-08 06:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-08 06:52:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-03 20:05:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 13:13:21.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-05-08 06:52:45.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"755cd9b062b0b354e56c2c3f1a661fe5106a51cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-03 17:28:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"873a568f4c8c451b3a696911be096fa756b0347a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 20:05:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dffac2a14c3540db9269dc00cc094d50\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbb28deeb8b6493aa4b20c346e3e9242 : SUCCESS in 13m 40s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3af1e2b97ef54774bbec1f36f717b3e9 : SUCCESS in 1h 13m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe556233c29e446f9448d8e0cf7d84e1 : SUCCESS in 1h 10m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/aeab7647916448cab4ab6c761f86dab7 : SUCCESS in 1h 26m 03s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c8e71a9543af4d2bbb107830e6b80c8f : SUCCESS in 1h 38m 14s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5db2564cb1f14130a184c9e5ae50e350 : SUCCESS in 1h 30m 44s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0c137a4389264a528f376b06d321e6d3 : SUCCESS in 1h 40m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/445614b8817c4e7da42b65f07131900b : SUCCESS in 1h 34m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a9bfaeeb96ec458d8efe274b6ac96c78 : SUCCESS in 1h 36m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a749009f320743b0824a82069456878b : SUCCESS in 1h 38m 23s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/056918ccd95a4f17a104fa8ecfc10f42 : SUCCESS in 1h 31m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/76dbfb63521a423a8356fcc44457948d : SUCCESS in 2h 01m 01s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ff0ee041d2f54e7eadaaedb0a05bb5a4 : SUCCESS in 2h 29m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6019d09dac6743fd9ae2ff0acfff8394 : SUCCESS in 2h 08m 40s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1529cba9fa4245eebaea627f0ae77192 : SUCCESS in 2h 28m 26s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd381e54dcf14967b24c2acc143e6024 : SUCCESS in 2h 07m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b26de38a45b4366ab26ec04e683b4d1 : SUCCESS in 2h 06m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d7fb61427cd45559a9706fd6b44dca9 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0c42a21d86176ca8449ed5c84baef8b6e0f5f6f4","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-07 13:13:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f16164f148117fdc39989f6636f4061b4c54336a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 06:52:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe9c93e021c1dad19de1aff60a0aee2e3f935b20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 06:52:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/841d7474f7cb44c7be5ef00dd488c7cc\n\n\nWarning:\n  Change 914144 in project openstack/openstack-ansible-plugins does not share a change queue with 918112 in project openstack/openstack-ansible-os_designate","accounts_in_message":[],"_revision_number":1},{"id":"5704b515b2118efe999f97dbf4d2cf83aac538f5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-09 06:35:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b32fd56bc73f2f0f473f10e315a630d60a7143a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 09:24:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68492c0e68c64ebeae522b4a62f8522a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edb8d8317bac4e3eb9ee7b8109871283 : SUCCESS in 14m 02s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1078f24e26ce4b6e89fe26c047649abd : FAILURE in 25m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74f77d4a9eaf483f9387875ce94b9fb2 : SUCCESS in 1h 08m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b1ff2d3a742b4ad99ab82b010b12e37b : SUCCESS in 1h 10m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f98be5ae32c04ad8816a10c48a1cde68 : SUCCESS in 1h 25m 31s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/21a5bb87b96f45d7a974d8a2825f152e : SUCCESS in 1h 31m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67a244f6604b4c6ba5351402e6952c8e : SUCCESS in 1h 34m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eeeb4581c0084042a18f73c044ac88b9 : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d1601851e97b4d298243e026fb563878 : SUCCESS in 1h 30m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f3a00074cc74b439bb56e45c5788248 : SUCCESS in 1h 24m 01s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc2767de3d744d52823b4d4a10a0ba6e : SUCCESS in 1h 12m 11s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59d18839501b4b39957392442edf7f55 : SUCCESS in 2h 25m 31s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e802d5f536a54f3488fcfd99fb179bcd : SUCCESS in 2h 22m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d55bcedc08b4df49d8610af7e8ffddd : SUCCESS in 2h 11m 59s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3492dee3424d4bca9f70465a9cba8a63 : SUCCESS in 2h 15m 10s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a58c2a3c7fa94988812401f93a5a03e7 : SUCCESS in 2h 19m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/066c873147ea4e7488366035ee658e8a : SUCCESS in 1h 52m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65996de79d7c42a282fbe28aee107eeb : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3c17e503a44b3ef9d6d40d439b4bead3f36e9fb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 09:26:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"edd9b878ec9356498c8f1e3718c4c9292716c718","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 11:43:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52a8a39ed39d4d188851430836387d1e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/436e38bd22bb40489319f731ded98a24 : SUCCESS in 12m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c19cf488439d46d5a38d8d1c30c2ee35 : SUCCESS in 1h 21m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4892e444132647e0b1293f071f0363b3 : SUCCESS in 1h 05m 37s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d144558bf1ab455792038965bae0c840 : SUCCESS in 1h 19m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/639db70e4a47450f8d4a672413d1b6f3 : SUCCESS in 1h 09m 26s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a02252c966324e75a0244b6171b4be12 : SUCCESS in 1h 21m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6879156a296e45b79f7d8f97d549b1fd : SUCCESS in 1h 12m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0470cc78a4a4d35a2bcdbc11319ea13 : SUCCESS in 1h 27m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4894c7cfbd3b4123855801e83d2d05d4 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0a2519d8f5b951e70bb7a14b02398452d8a24f15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 11:43:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f7472a7732604d2d97e68e1ce0b2d8ef58fba19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 11:44:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8546f369acdb486cb00733fb8504abca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f308341d81db42c88f188b0e3afca098 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6367c6fa2662f05fca5ba4275c4d1d7b844d277","revisions":{"f6367c6fa2662f05fca5ba4275c4d1d7b844d277":{"kind":"REWORK","_number":1,"created":"2024-05-03 17:28:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/918112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/12/918112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/12/918112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/12/918112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/12/918112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/12/918112/1"}}},"commit":{"parents":[{"commit":"5fd3b32e1e038310be322483d30a0ee00d7e798d","subject":"Add service policies defenition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/5fd3b32e1e038310be322483d30a0ee00d7e798d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-03 17:28:14.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-03 17:28:49.000000000","tz":120},"subject":"Add variable to globally control notifications enablement","message":"Add variable to globally control notifications enablement\n\nIn order to be able to globally enable notification reporting for all services,\nwithout an need to have ceilometer deployed or bunch of overrides for each\nservice, we add `oslomsg_notify_enabled` variable that aims to control\nbehaviour of enabled notifications.\n\nPresence of ceilometer is still respected by default and being referenced.\n\nPotential usecase are various billing panels that do rely on notifications\nbut do not require presence of Ceilometer.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/914144\nChange-Id: Ia29caf1633d5467e870226763088065fde54f12d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/f6367c6fa2662f05fca5ba4275c4d1d7b844d277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/f6367c6fa2662f05fca5ba4275c4d1d7b844d277"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
