)]}'
{"id":"openstack%2Fopenstack-ansible-os_freezer~1003081","triplet_id":"openstack%2Fopenstack-ansible-os_freezer~master~Idb6134dd073d490a11a89c7cc82a47fe2732ac05","project":"openstack/openstack-ansible-os_freezer","branch":"master","attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-09-22 20:17:49.000000000","reason":"Change was marked ready for review"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-09-22 20:17:49.000000000","reason":"Change was marked ready for review"}},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-22 20:09:52.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Idb6134dd073d490a11a89c7cc82a47fe2732ac05","subject":"Set persistent scheduler jobs directory","status":"NEW","created":"2026-08-31 16:24:00.000000000","updated":"2026-09-23 11:50:49.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee1b85cb0803d4d5d008671d59f679adf7da0898","_number":1003081,"virtual_id_number":1003081,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-23 11:50:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 20:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 09:55:00.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-09-18 09:33:33.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-09-22 14:22:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"6663c225e85506e8ecc5d6fe2af6af43aa6c3cb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-31 16:24:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75ad28b2eaa2906e9d9f926eff7c438a035e5dd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-31 16:34:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"19ffb3ab7a8aba01e8c01638e876d68f88754862","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-31 16:38:35.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"adfaea521295a1f96d824a5062e126c625b4ad4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 20:18:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/def3b38f29724d72985772c57350ed5a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f447ef9d579c4d0188f40faa66154d19 : SUCCESS in 11m 27s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/7cf25a242daa43878458f6737581d58b : FAILURE in 2h 32m 24s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/391b1ffe2acd4afcb23ad08efdd6ca71 : FAILURE in 1h 23m 03s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/68fc2ddd20fc40539125df44cefe2748 : FAILURE in 2h 31m 21s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00c28d17f26f4cb795e1449a14714149 : FAILURE in 1h 31m 14s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/3e96220ae3b240c496a1d53fc6db1557 : FAILURE in 2h 38m 11s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/688a26ceac10481896f734c77569eda4 : FAILURE in 2h 09m 10s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd795a206a82405dbccc4beec3e46d33 : FAILURE in 2h 40m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e472e5ca01b544ae8b63b310670d4755 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbcbd92f6f534a12822c994425d3da05 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":3},{"id":"9261cc9330728c307dbecdef346ece3ed48fab13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-01 12:13:55.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"80b1c481d7eabc20d0d18c687f0c28f41f5e1ea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-01 12:18:18.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"4d194f64ae676bb98bafa6d4675063f231cc7632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 15:21:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beae99433cc74ac487d7c9fe5e6a7dd2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c48d8c0faf54c899d52bf2854d53701 : SUCCESS in 9m 51s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2c6cb5568c714aa6af454ca62a564634 : SUCCESS in 1h 32m 39s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/23bdee93885548fd8ea9484c7224c7dd : SUCCESS in 2h 32m 24s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/64c0a3d7e24c473a88fa3ac07c47c526 : SUCCESS in 1h 36m 17s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1ed95a9190f428b836bde07e728d57b : SUCCESS in 1h 33m 02s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d0bce1fb5e6a4a998f4c65ddd5c8979b : SUCCESS in 2h 42m 33s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d7d280b72ad74a25ba2bd42acae93432 : SUCCESS in 1h 40m 56s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db4eb005ac834c598ee23ea2d87cccd8 : SUCCESS in 2h 52m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/641f094fd7484eb4b3991b2aad3a526c : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b84448dd4a514e34b3d5cb16261fb122 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":5},{"id":"e62271cdd30403bdf665a7f36addea7fa473f1a9","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-09-16 09:55:00.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e21dfcd10cf58eef29a3599052e7eb080e477ace","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-09-18 09:33:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b872aa4305ef958c2ab87a1d3b5079d048e46d00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 09:38:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7f1aafb870e75987db35815c6612197123a20e2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 12:52:25.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a64b3b36b574f7d935211d42e8c117f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/954f06a97fc14a1bac749c6878636f67 : SUCCESS in 14m 42s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3780f0a7c1524534808b8733d15c8d5b : SUCCESS in 3h 04m 55s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a46093221334458fae7ecb85b5460bbc : SUCCESS in 3h 02m 14s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/18fad03f41c24ff0be5768e4956b0cbe : TIMED_OUT in 3h 11m 56s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68e9ec8e35564326be0af78d18689d59 : SUCCESS in 3h 08m 29s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/856aad6f55fb48e3aaf5ffb82e24181e : SUCCESS in 2h 45m 47s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6b70f3db1d2d48ea8f21cd229feaac3a : TIMED_OUT in 3h 11m 15s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b98d2c9221ff46a59adf0839158fbf07 : SUCCESS in 3h 06m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b86937480dc4575988b1c77d067b9ce : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8394d60d20b4751986813d2bed9fc0b : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":5},{"id":"3fd22d68a91431caefa3659821f6383288703f56","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-21 21:16:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d283728d6770929780db658c044ee281d5adc880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 00:31:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7743661ceff74b3cb959185624b8f1a3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4ebcfddf1744dcab37e5cc349f50032 : SUCCESS in 8m 43s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/aa49b1a96aba41c38f092e69fb5dc0de : FAILURE in 2h 04m 13s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8cddb1206a38476197ff3ad3dd86e09d : FAILURE in 2h 44m 54s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9a2a581161c846a8b259229ca879cc43 : FAILURE in 1h 49m 16s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c1b72a8ac8a47a992aaf60c8e86090a : FAILURE in 1h 25m 46s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9940edbe05e3412bb63e12c099c4ce13 : FAILURE in 2h 57m 08s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d3322a231bcc408198a4c98613e298cf : FAILURE in 2h 37m 01s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8187f75684ab422aac835c6e9b6bb7c0 : FAILURE in 2h 50m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4cffe6602c4a008a31cd49cee5c7cf : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce93bcf38ffd464eacb803028656bb99 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":5},{"id":"3cdc99fb323aad6301409b66fc740d0840f5795d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-22 14:22:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f58a3efb83b066e5c37c998b240f424c273713bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 17:38:16.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ade258a97a34de18bc4a7be14a5138d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab74e3d23762407584e2eb5abca62ab6 : SUCCESS in 13m 34s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f6985b5bb7dc4d969afbda0302a09ff7 : FAILURE in 2h 44m 20s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a078f0aaa4d8459f95293455b0759eee : FAILURE in 2h 57m 19s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e508180157c34f8f8513618f3657f914 : FAILURE in 1h 50m 18s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63dd87fe174f4074b1a75ccda9c972a7 : FAILURE in 3h 07m 23s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b10b15cbb8db41cdac83efee2d5633e7 : FAILURE in 2h 48m 33s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8b7c29964f784d1a9d7ca238bb5597a8 : FAILURE in 2h 46m 34s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e10abf08c694025b750b7e86cfd7147 : FAILURE in 2h 52m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2659f96a3e44ec98a11c0ff22b99d9 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6eb72df62ff9462bac8b4552881f5e60 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":5},{"id":"ed8eac83c321873e760564b0e68ae21633e33fc6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 20:09:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c93173c22951fde250c2907cd99ed42d2d4ca446","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 20:16:55.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nThe following files contain Git conflicts:\n* tasks/freezer_pre_install.yml\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"f500f3ea680882389662b3fcd615aca1dbbc4b2f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 20:17:43.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"1dda4d2946f31eeb76663ee8b8a6b489ccf16107","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 20:17:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"af2576d49a9c49a2d7043b59369f70d2f04c3a7d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 20:19:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"04635c10aab42351d7d2134bbe941acb1d9c8183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 23:09:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6715ae3bd1f4fdd967ad63d2a4e185e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c6f24bcc1354aaf8ce6fc1ec856a745 : SUCCESS in 14m 50s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f09ce0e213514b029d6d0bb948c83c9b : FAILURE in 46m 32s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2c92f3362f22452c8cd87e8f76f499cc : FAILURE in 1h 38m 33s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8d1b722bc13042c0bb75b70144d75c3f : FAILURE in 43m 33s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/baefd0067dce492cabada20ede1a2549 : FAILURE in 1h 48m 18s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c730ca9759944608bc8321fed2a0c6c0 : FAILURE in 2h 48m 44s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d79f50fae69142dabbd6e5aa940a0dec : FAILURE in 48m 30s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b2fa5fdeb394dbaa39262349bb5bf34 : FAILURE in 1h 26m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baed8d15a4e44dae9c0e36e9bec33375 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2eff7da78e348d2a80a92391b3e35f8 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":5},{"id":"d80dac5bc13b8b164484194bb13a124c6025a928","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 08:25:33.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ee1b85cb0803d4d5d008671d59f679adf7da0898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 11:50:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa0da333ecd40a4ac84ae5884b898c4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/749369c47c4349ea978a916acc43f065 : SUCCESS in 19m 14s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/1ee05278805347bdbbdb05d271b3f494 : SUCCESS in 1h 39m 48s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/735eca9f7ccc495da87d589853c9a085 : SUCCESS in 2h 23m 17s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1db0c33cd7d7410db066b973ae20f82a : SUCCESS in 1h 44m 11s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef3383890cff4f32b4191848c6665367 : SUCCESS in 3h 12m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/758599845ac8458b892938bfcd57b4f3 : SUCCESS in 2h 58m 51s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0bc687d6517946aeaa36cc060db183a6 : SUCCESS in 1h 43m 53s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35f848f0cd6240bea042763625c2aed0 : SUCCESS in 2h 34m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4179e881e0c44b4ab3b58216a25580b4 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29dc91817c0944a0845bd264caa2cd26 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"40810821c38c6aaaf7be46b284cdc4efc3f48e81","revisions":{"d6f8e8209db9355160ae2d722fad9983c5bfcd97":{"kind":"REWORK","_number":1,"created":"2026-08-31 16:24:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/1003081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/81/1003081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/1"}}},"commit":{"parents":[{"commit":"4b512d5304399dcb97611735a01d06f2c40c04b4","subject":"Merge \"Add information about manage configuration by OSA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/4b512d5304399dcb97611735a01d06f2c40c04b4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-31 16:23:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-31 16:23:53.000000000","tz":120},"subject":"Set persistent scheduler jobs directory","message":"Set persistent scheduler jobs directory\n\nAt the moment /etc/freezer is a dynamic path symlinked from the venv.\n\nIn case of venv_rebuild is set to true, old venv is deleted along with\njobs dir.\nAs jobs may be defined staticaly, without API, it\u0027s important\nto avoid the directory deleted along with venv.\n\nThus, we move jobs dir to /var/lib/freezer to make it persistent\nand independed from venvs.\n\nChange-Id: Idb6134dd073d490a11a89c7cc82a47fe2732ac05\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d6f8e8209db9355160ae2d722fad9983c5bfcd97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d6f8e8209db9355160ae2d722fad9983c5bfcd97"}]},"branch":"refs/heads/master"},"18ec2a06ad8cfd311c6384fe8c43917ddd4c76e4":{"kind":"REWORK","_number":2,"created":"2026-08-31 16:34:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/1003081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/81/1003081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/2"}}},"commit":{"parents":[{"commit":"4b512d5304399dcb97611735a01d06f2c40c04b4","subject":"Merge \"Add information about manage configuration by OSA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/4b512d5304399dcb97611735a01d06f2c40c04b4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-31 16:23:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-31 16:34:57.000000000","tz":0},"subject":"Set persistent scheduler jobs directory","message":"Set persistent scheduler jobs directory\n\nAt the moment /etc/freezer is a dynamic path symlinked from the venv.\n\nIn case of venv_rebuild is set to true, old venv is deleted along with\njobs dir.\nAs jobs may be defined staticaly, without API, it\u0027s important\nto avoid the directory deleted along with venv.\n\nThus, we move jobs dir to /var/lib/freezer to make it persistent\nand independed from venvs.\n\nChange-Id: Idb6134dd073d490a11a89c7cc82a47fe2732ac05\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/18ec2a06ad8cfd311c6384fe8c43917ddd4c76e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/18ec2a06ad8cfd311c6384fe8c43917ddd4c76e4"}]},"branch":"refs/heads/master"},"215794a9785b992637f0846d62e66092249674e4":{"kind":"REWORK","_number":3,"created":"2026-08-31 16:38:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/1003081/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/81/1003081/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/3"}}},"commit":{"parents":[{"commit":"4b512d5304399dcb97611735a01d06f2c40c04b4","subject":"Merge \"Add information about manage configuration by OSA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/4b512d5304399dcb97611735a01d06f2c40c04b4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-31 16:23:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-31 16:38:35.000000000","tz":0},"subject":"Set persistent scheduler jobs directory","message":"Set persistent scheduler jobs directory\n\nAt the moment /etc/freezer is a dynamic path symlinked from the venv.\n\nIn case of venv_rebuild is set to true, old venv is deleted along with\njobs dir.\nAs jobs may be defined staticaly, without API, it\u0027s important\nto avoid the directory deleted along with venv.\n\nThus, we move jobs dir to /var/lib/freezer to make it persistent\nand independed from venvs.\n\nChange-Id: Idb6134dd073d490a11a89c7cc82a47fe2732ac05\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/215794a9785b992637f0846d62e66092249674e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/215794a9785b992637f0846d62e66092249674e4"}]},"branch":"refs/heads/master"},"e49452c9575ada426b74aea5f64ba9387303f67b":{"kind":"REWORK","_number":4,"created":"2026-09-01 12:13:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/1003081/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/81/1003081/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/4"}}},"commit":{"parents":[{"commit":"4b512d5304399dcb97611735a01d06f2c40c04b4","subject":"Merge \"Add information about manage configuration by OSA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/4b512d5304399dcb97611735a01d06f2c40c04b4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-31 16:23:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-01 12:13:54.000000000","tz":0},"subject":"Set persistent scheduler jobs directory","message":"Set persistent scheduler jobs directory\n\nAt the moment /etc/freezer is a dynamic path symlinked from the venv.\n\nIn case of venv_rebuild is set to true, old venv is deleted along with\njobs dir.\nAs jobs may be defined staticaly, without API, it\u0027s important\nto avoid the directory deleted along with venv.\n\nThus, we move jobs dir to /var/lib/freezer to make it persistent\nand independed from venvs.\n\nChange-Id: Idb6134dd073d490a11a89c7cc82a47fe2732ac05\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e49452c9575ada426b74aea5f64ba9387303f67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e49452c9575ada426b74aea5f64ba9387303f67b"}]},"branch":"refs/heads/master"},"835265f04f8c6a7620b47349f85c33e0deddd71d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-01 12:18:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/1003081/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/81/1003081/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/5"}}},"commit":{"parents":[{"commit":"523e1c11bfcd78aaf1c7b47289e12a416648146e","subject":"Implement freezer-scheduler clustering support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/523e1c11bfcd78aaf1c7b47289e12a416648146e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-31 16:23:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-01 12:18:18.000000000","tz":0},"subject":"Set persistent scheduler jobs directory","message":"Set persistent scheduler jobs directory\n\nAt the moment /etc/freezer is a dynamic path symlinked from the venv.\n\nIn case of venv_rebuild is set to true, old venv is deleted along with\njobs dir.\nAs jobs may be defined staticaly, without API, it\u0027s important\nto avoid the directory deleted along with venv.\n\nThus, we move jobs dir to /var/lib/freezer to make it persistent\nand independed from venvs.\n\nChange-Id: Idb6134dd073d490a11a89c7cc82a47fe2732ac05\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/835265f04f8c6a7620b47349f85c33e0deddd71d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/835265f04f8c6a7620b47349f85c33e0deddd71d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"4b512d5304399dcb97611735a01d06f2c40c04b4","ours":"e49452c9575ada426b74aea5f64ba9387303f67b","theirs":"523e1c11bfcd78aaf1c7b47289e12a416648146e","merge_strategy":"recursive","contains_conflicts":false}},"688920deb980b72c5e72d32ad3c5c83e42aa7107":{"kind":"REWORK","_number":6,"created":"2026-09-22 20:16:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/1003081/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/81/1003081/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/6"}}},"commit":{"parents":[{"commit":"f6c3fd5e4e274867ba16937f7a74351945ed7247","subject":"Separate API and Scheduler venvs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/f6c3fd5e4e274867ba16937f7a74351945ed7247"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-31 16:23:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 20:16:55.000000000","tz":0},"subject":"Set persistent scheduler jobs directory","message":"Set persistent scheduler jobs directory\n\nAt the moment /etc/freezer is a dynamic path symlinked from the venv.\n\nIn case of venv_rebuild is set to true, old venv is deleted along with\njobs dir.\nAs jobs may be defined staticaly, without API, it\u0027s important\nto avoid the directory deleted along with venv.\n\nThus, we move jobs dir to /var/lib/freezer to make it persistent\nand independed from venvs.\n\nChange-Id: Idb6134dd073d490a11a89c7cc82a47fe2732ac05\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/688920deb980b72c5e72d32ad3c5c83e42aa7107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/688920deb980b72c5e72d32ad3c5c83e42aa7107"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"523e1c11bfcd78aaf1c7b47289e12a416648146e","ours":"835265f04f8c6a7620b47349f85c33e0deddd71d","theirs":"f6c3fd5e4e274867ba16937f7a74351945ed7247","merge_strategy":"recursive","contains_conflicts":true}},"40810821c38c6aaaf7be46b284cdc4efc3f48e81":{"kind":"REWORK","_number":7,"created":"2026-09-22 20:17:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/1003081/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/81/1003081/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/81/1003081/7"}}},"commit":{"parents":[{"commit":"f6c3fd5e4e274867ba16937f7a74351945ed7247","subject":"Separate API and Scheduler venvs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/f6c3fd5e4e274867ba16937f7a74351945ed7247"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-31 16:23:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 20:17:41.000000000","tz":0},"subject":"Set persistent scheduler jobs directory","message":"Set persistent scheduler jobs directory\n\nAt the moment /etc/freezer is a dynamic path symlinked from the venv.\n\nIn case of venv_rebuild is set to true, old venv is deleted along with\njobs dir.\nAs jobs may be defined staticaly, without API, it\u0027s important\nto avoid the directory deleted along with venv.\n\nThus, we move jobs dir to /var/lib/freezer to make it persistent\nand independed from venvs.\n\nChange-Id: Idb6134dd073d490a11a89c7cc82a47fe2732ac05\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/40810821c38c6aaaf7be46b284cdc4efc3f48e81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/40810821c38c6aaaf7be46b284cdc4efc3f48e81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
