)]}'
{"id":"openstack%2Fopenstack-ansible-os_freezer~975794","triplet_id":"openstack%2Fopenstack-ansible-os_freezer~master~Icfc63d099183460a8ef706d2a017fafc759cab17","project":"openstack/openstack-ansible-os_freezer","branch":"master","topic":"osa/freezer","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-04-16 02:04:02.000000000","reason":"Change was submitted"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-04-16 02:04:02.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-04-16 02:04:02.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-15 23:09:31.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-16 02:04:02.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-16 02:04:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icfc63d099183460a8ef706d2a017fafc759cab17","subject":"Place content for os_freezer role","status":"MERGED","created":"2026-02-05 13:11:25.000000000","updated":"2026-04-16 02:13:32.000000000","submitted":"2026-04-16 02:04:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"975794-osa/freezer","meta_rev_id":"99aa6a97478dba729b3d1ba3e43d05c01536ecbf","_number":975794,"virtual_id_number":975794,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-16 02:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2026-04-15 23:09:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2026-04-15 23:09:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-05 14:42:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-05 14:42:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-05 14:42:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-02-05 14:42:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-02-05 18:08:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-31 15:38:03.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"CC"},{"updated":"2026-03-31 16:06:09.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"01cdd37dd32c0de8924615b99ed094e05f6e86ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-05 13:11:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"883f016c178eabe08141cd7708338a9de2d9ab6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-05 15:19:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e146e16d1a89a9a572b6c4a7f9093aa381f4bbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-05 15:21:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"84ecd5677f60bffc1178aeaa94aa4523b58a7737","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-05 15:25:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dddf623cedce3fe5191edfdc4cea75e06746940e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 18:08:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd435b0cc0af4238b0ba6197f632f211\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a1ddffcfd0f41e48c2a0fdbf34d2deb : SUCCESS in 2m 48s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/110508ed311d48089da39391d7553743 : FAILURE in 10m 58s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2a511b000cc049b7b23c39600124aca7 : FAILURE in 1h 08m 55s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9fbdb106cffb47d09c28ad22fc17d6ee : FAILURE in 1h 15m 56s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5421f96204144dd795f46d496598e1fe : FAILURE in 1h 19m 46s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8b24c309579846f398b1dc66f9861781 : FAILURE in 42m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c46681fbd637415781a4b4d5543b9fd1 : FAILURE in 1h 13m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98ed4f7c49f34042920d079ea40522f6 : FAILURE in 45m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d4d996110dd4caebc28b02205c9355c : FAILURE in 43m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4d60e9882e424779a9fc8f14c88b1108 : FAILURE in 58m 42s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2d747eba99744db1800a33cb64257ceb : FAILURE in 2h 24m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b528ef92fdc4b04bffaab480df6d99d : FAILURE in 1h 00m 16s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb1fcfbb190d4dd8bb3baf3b2f193f88 : FAILURE in 2h 08m 14s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3612b5f679834df082972ef26173d2f0 : FAILURE in 2h 17m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d06f22fbcfcb498ea92d3f06707f85ff : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cf5479b62a541b88b2f58050fb60c01 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":4},{"id":"56d699e333f3be88638ba6afbd0f4063fc32f4fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-06 15:03:15.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5727a92059a410638fd3b775b59bb8799a9a867a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 18:11:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5564953e214647be8a33a6df4ee1ed60\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8fcb8e339aad4eedbf0ecbebe9631158 : SUCCESS in 2m 50s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a5bf1ab530b4164b92c27d3a3ff335f : FAILURE in 6m 29s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/48e4e0398df8484e8e621accab630cde : FAILURE in 1h 23m 35s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d81f2253d5604807b7cc65aad7c12e2b : FAILURE in 1h 15m 36s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b78c179902584214a23412ee41dfd216 : FAILURE in 1h 21m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d7ab3e6bba841e79f0df5b662aa5d9d : FAILURE in 1h 19m 29s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/13b448b931a54d2c844988e2753451e1 : FAILURE in 1h 34m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/367ce32dc468438b98da3ffcca72939b : FAILURE in 1h 20m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e80118abf085403cb810b50c96a4ef17 : FAILURE in 1h 03m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ba8dd9c56af34fdd93a012e20852a099 : FAILURE in 1h 07m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f59027f373f54794addfa4897528fb45 : FAILURE in 56m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90f48a925b434e448bc378e70a23d8da : FAILURE in 1h 58m 25s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/343bdb20935d48efb4bfb8eca358cbaf : FAILURE in 3h 01m 34s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c896dbbcaa6f47a897417d5f994927b8 : FAILURE in 2h 24m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73883026e14741f889b17b01ec0887c7 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/189df1af54074878929d996431c86a52 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":5},{"id":"435bcfbe405f512e9f2f524ab9f6bddbf900709c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-25 09:45:31.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1e661ff773a659221556ffeb1766b8d2999c7268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 12:20:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbbb106147cb4e7f8b9192b753056d0c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/97e158a8dfac43c4aeb9a2a3cecb7dd8 : SUCCESS in 4m 07s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bf02fb7426e4c1e98b95600bbe258d2 : FAILURE in 8m 59s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7341f7e5b46b4cea907f7d1543c46c8a : FAILURE in 1h 09m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f60ec618762a4e57a484e7403e7f57c5 : FAILURE in 38m 12s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/1dcedb9a7b1449858a2ebc2004decc52 : FAILURE in 1h 10m 15s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/54f1690db99e4575ab0f147be47e58bc : FAILURE in 1h 11m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8e830fb2be7b4258a8ae6e621dfa421a : FAILURE in 1h 22m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96b7272d4eae48ecb19a3b42e452b061 : FAILURE in 1h 01m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b225a84142046169fb35990651eee93 : FAILURE in 1h 14m 15s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a31548f22d8b43959b0301424e48ec8b : FAILURE in 59m 08s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bab1a52b933b40788f25234d492b5622 : FAILURE in 1h 54m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a651df6cc83406ca20d21bf5deecfac : FAILURE in 1h 54m 30s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df431c4ddf1946959dc3fcbce18b8bee : FAILURE in 58m 55s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d613440f0294c58955119ca6939486a : FAILURE in 2h 25m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b74ea58da6541968c3a985df3f797ea : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44a8ea5002084f078d0cef704442303e : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":6},{"id":"66bdb3734301383a7da6a17470cd9e479436b4d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-12 10:29:45.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5ce1c6dd995be2331bcfa70bfe1527d68ce7c0e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 13:01:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3743b0c5206e469eb7d122797fa9ae7a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/54a1671b305344eaaed6982d3b4ae913 : SUCCESS in 3m 48s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/386402af362a4984b31e19279325ea29 : FAILURE in 15m 42s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ed4ebb1abde944839c6961008c2a5f2b : FAILURE in 1h 28m 05s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2d40ea6512b841408c22fa67cca5c0f5 : FAILURE in 46m 51s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/057f18665bd54bcb9dc128f5fee847d4 : FAILURE in 46m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/45281e88fad14b87aba4d7aead64edd3 : FAILURE in 46m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e3f03054f4a6427982b6662f10ee6ac0 : FAILURE in 47m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4603833903c549ccac487f5055a87d96 : FAILURE in 1h 25m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5e275ce82ca44e0af54c9b935c83960 : FAILURE in 1h 26m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/734d07c749494989b9eb105ba2c63f6b : FAILURE in 2h 16m 03s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cfee116bd73b4cdeaeef87cf3cea1298 : FAILURE in 1h 04m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd2665300b214a839fea6618fb4863be : FAILURE in 2h 14m 55s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16b66cf5ee57485ba1d7a3b9f3413016 : FAILURE in 1h 50m 05s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c39dc5a3fc1848ce9fa8d182a0d565f1 : FAILURE in 2h 21m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/988c254af39047d098f365d284122e7b : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2de65c49d75a42fda8d6eb980ae7a20e : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":7},{"id":"695ee4438213b2d98fffd35449e6756517f8f96e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-12 16:08:40.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"392d85da679ff6bb520a9a8917dede85cc418e72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 18:37:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36f77d7017d74672a44a3341ae28bdf3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/85d27023e9fd4f70bc47a87b4d11c6d9 : SUCCESS in 3m 16s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/552b1a2b49214f9f931161cb526f052b : FAILURE in 11m 27s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c18e972281ce4eb7935f7041b8d6fc63 : FAILURE in 43m 56s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/93dc5236f12d402eb559042368b124ba : FAILURE in 1h 24m 29s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/29c6b4947c7f47daa405ae1d4e57089d : FAILURE in 1h 13m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bfd798de7e134bddb95e7abb2dcff323 : FAILURE in 1h 17m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fb088dcbadba40c99b4e245e2ac10ff9 : FAILURE in 1h 13m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7134f1745be49c2a9b3a90ad8686d40 : FAILURE in 1h 23m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37dd1ad279de49a3a3595589d8c27dc7 : FAILURE in 1h 30m 09s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eb3c92850fae4dbfaf836a1930c0d314 : FAILURE in 1h 43m 41s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6a73222939d14a68ba393f32025e1ba6 : FAILURE in 1h 43m 22s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b67eb79a33a54e869b9e93b31935fadf : FAILURE in 57m 08s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77e74150efc648299a5d9ad515f67af0 : FAILURE in 1h 35m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e6cb9566efe4a659997d209a0bb8ab0 : FAILURE in 2h 18m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb0c237e34147939db445a4417c7916 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d8b667f763840928f494f078996399a : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":8},{"id":"3f0ef184357422e5d3cfa3ad01587b47f171f435","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-13 09:25:25.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0f97147b7d5fb3cdcf247c2d804f27b8ddddb9db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-13 11:46:04.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"3d7d5a90848f618c0b6e1e83ce6f0bf72c35fadd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-13 13:48:14.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d38afab3adbb8427e224e4e0b4834a49f7130e16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-13 13:52:12.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"dcf30fc12e0bb944518b69b085ed82b5f7e87aea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-13 13:57:42.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"962a348948aaf40cef6c3ed530ac7ae42310cf21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-13 14:20:13.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a7b472cace7e934e34751be89cda0ba92d24535a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 17:02:21.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2655d5e5f9244cb581afeea17f525760\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/01f9c35c21854551bb94a0e3237dc267 : SUCCESS in 2m 35s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85e6396975d047dbad9c1f83e4163042 : FAILURE in 10m 30s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3aa6198f60e942eead2b8afeb7becb1c : FAILURE in 1h 35m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a95998558d87482e886db9337ee995e8 : FAILURE in 41m 48s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/04a29c17ec084895b7b4a01787e9ce86 : FAILURE in 1h 19m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c50615c811db4f558be47de71c62121c : FAILURE in 1h 10m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f2f6e9bcc7c64ff497d9a3e79383d7bb : FAILURE in 48m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1295b1af6d324e358b6bd569bf8e01e4 : FAILURE in 1h 21m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/273b0a28e8f94e1783b6d9c3bf4fcede : FAILURE in 1h 27m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/04d649e3cba84c9ba2d3f4439d16618f : FAILURE in 2h 33m 10s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/53fc03217f914225b3daeaff50c12b9f : FAILURE in 1h 03m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dea8a795fe143d8a8adf34d4a87e3f6 : FAILURE in 2h 14m 37s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a4620340d3147c5b003d16076ee3f1f : FAILURE in 1h 05m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cda45ae8631e4dd4b6dfcd58c65dc838 : FAILURE in 2h 25m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46ffa7ded2a439c82ea8ef892157145 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c23bec3f2204cf6be53f35cb392b0c3 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":13},{"id":"3822521e118e62d78812498ddb04b308196384b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-20 14:28:57.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"2c991136b096194c66d964acbe9ae8b69831de5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-20 14:29:28.000000000","message":"Uploaded patch set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"af89ab8a151dc077c078e4bbecf69de7a39347b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-20 15:06:01.000000000","message":"Patch Set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"c9b777640b4fd0f402a0ea7036a70d8e777af320","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-20 15:06:07.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":16},{"id":"e05ac0bf7099d01deac65459b5a78e6e8a4a6ae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 15:14:02.000000000","message":"Patch Set 16: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible/+/975512 is needed.","accounts_in_message":[],"_revision_number":16},{"id":"b0695ca819fff2e8ffc390c6870e362df3bbdf12","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-20 15:16:32.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b3a8706abb3c739f4541414bfcd4d9784aa3216c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 18:47:41.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b747e5f40c5a4a5a89328ae381f4c145\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b8ab0e5ef134252b56a2c9c8a1f386f : SUCCESS in 3m 17s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55df57f05fed495391e29da3c7d2b373 : FAILURE in 12m 47s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/da060262a155422b89d278d398fcb091 : FAILURE in 1h 01m 00s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7470e0b77ffe4ae99800b4e4bd3878ac : FAILURE in 1h 34m 38s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d86d62e053a7490e9062576024742ed5 : FAILURE in 1h 38m 10s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8cd4a3536cb54dc082886ce5eafe49fe : FAILURE in 1h 30m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2dcee7265269409c8c8ec780f38e8dd1 : FAILURE in 1h 30m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e754ff1994c4f428c591271b1fa3445 : FAILURE in 1h 35m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3389733b988e442fb0826ed2cba5f427 : FAILURE in 1h 46m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/be033753fd024cabaed926f68a290257 : FAILURE in 1h 21m 59s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bf82e70cb7d34b3db612cccf0acb768e : FAILURE in 1h 48m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81a348abd5fd4192b825fb95d10d8724 : FAILURE in 2h 17m 08s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d21c5fd572f4431af4040abea5e338a : FAILURE in 1h 10m 48s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02d8747bdc44464dbf1873190de995c6 : FAILURE in 1h 24m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569f57aa1f5b4efbab6a15a6d3a71f6d : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d6e6743a57640ceb26e66ab96f7279c : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":16},{"id":"d67a05a70f9358b1a62c1ca5e09e2e162fc526d2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-21 09:39:34.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"df1817b5f92cac39e5fc91b3d066983a51d64dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 12:22:30.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a6ad9ca5c5f4cb881af65bb16237602\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/280ae8e6af614912b7acd14600c7b184 : SUCCESS in 2m 32s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/217dfcd0017b4e76adb5dc974bd19d53 : FAILURE in 11m 26s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e677467a51404e398d92350759c64302 : FAILURE in 1h 46m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/34b137b254024d0799dda198cbba348c : FAILURE in 1h 43m 17s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ded431c1b9a24ab5b0383a46bd576927 : FAILURE in 1h 37m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/260ff365cf934d60ac8eb7be0fcb9c14 : FAILURE in 57m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5c1ad31842a747efac00e22613640cfb : FAILURE in 1h 53m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/148c9e5828e244e6852f408a89561db4 : FAILURE in 1h 43m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2f6f7c0be3946988b5870a66e61436a : FAILURE in 1h 42m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/359067292d5a4edc9b504e06dda50a6c : FAILURE in 1h 09m 09s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/761f3ef929aa4cdaadfe16f7ac9e4028 : FAILURE in 1h 51m 10s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6998bebb36c640cb94223ceee9b82fd3 : FAILURE in 2h 34m 44s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1b00e859b0744faa44ebe5a845991b1 : FAILURE in 2h 20m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/048eca01b85243f18fa352ed0294122d : FAILURE in 2h 28m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f48a318f7ee4c028e36a286b1794f08 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be186fdf2d514aa29a8e4e88792add04 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":16},{"id":"e434884224c118879c0a40ddcce675e35859a2b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-26 09:31:17.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"3a296361377f0726d5fd1b5591ae29c24aba75c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 11:56:39.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/273ad0fc553c4868a404d8d11d59584e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d0b477b398dc4d52b672c96e6ff9aecd : SUCCESS in 2m 26s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f31a16a734847c29289bb4882491462 : FAILURE in 10m 38s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/1635c813a0a44e5da47870adb62dfaf0 : FAILURE in 1h 00m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5c18308b2bd240da88e67b9dbdbb26cb : FAILURE in 1h 01m 15s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b8b20879564145989fd3082beb7c761e : FAILURE in 1h 36m 37s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/457e5fee5f194f8987397722cc32e005 : FAILURE in 1h 02m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e8bf6948099c4e07b4b3bcbc4384979c : FAILURE in 1h 04m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f09b004b667a43bbaa5d47075b404805 : FAILURE in 1h 47m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77f981816a744cd6b82498eac4f9bc7a : FAILURE in 1h 42m 41s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/23619b5c7cd747ce805a228c1be7f7dd : FAILURE in 33m 42s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a33d231dffae4927bc992fc238cb527d : FAILURE in 1h 20m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/024ef8d2d8584399975691ca7359ef4b : FAILURE in 1h 15m 11s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75082bfeb43540ccbb50128203b8ab54 : FAILURE in 2h 00m 06s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5e71395bd9d47c086e36d3b70ce055d : FAILURE in 2h 17m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b2090c320c14cf1a86fb5c7c6d6aa02 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a573f71d4774ce3a8bbecc5b35d3991 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":17},{"id":"d44c86eacb02340e1c97af64c234276a0e108928","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-26 13:30:46.000000000","message":"Patch Set 18: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"5f03f3f89698a095a5f6a52dc89b53889bb35f3c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-26 14:55:17.000000000","message":"Patch Set 19: Commit message was updated.","accounts_in_message":[],"_revision_number":19},{"id":"a3af9fd9752186f61bdef6ec4c55f1e7c6b4dfd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 17:20:53.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cea5bb49a144164b09200b1577de8c1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b3bfa9b7fae404db383a94d11303ac4 : SUCCESS in 2m 46s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1eb329b8109f424880420596aed6c7b7 : FAILURE in 10m 04s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ef90ee2fbc464b2d842c15d8fddb5214 : FAILURE in 1h 38m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9df275481f834c6aa31125ac1cee48eb : FAILURE in 1h 29m 56s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/58407ef390524a858ed5880ec79ffd66 : FAILURE in 1h 26m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/24bb0fe4f00f4efebb657d93aa3baa2c : FAILURE in 1h 16m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7234404a0c0f4b779b1d1ab3156c74bb : FAILURE in 1h 44m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/739da1d12151402f9dbcc1aaf41c2f07 : FAILURE in 1h 34m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11eff440bdca4f3cad49f0730aa66ae9 : FAILURE in 1h 40m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b4dd1bf88f7746429688fef2f26d623c : FAILURE in 1h 53m 53s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e2badb5a5bf54ed5a2e11d3bc8856c5a : FAILURE in 2h 07m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a7a423d324c47aa8159ddca3a6e9a30 : FAILURE in 2h 16m 52s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7f8b5f153f14e7d86739b00661082c3 : FAILURE in 2h 12m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9bb3535ec40433384932188dd30dfd1 : FAILURE in 2h 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b7eea8471d4fe083dfa1cc155da317 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48211803d7c441258a1554d19b436016 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":19},{"id":"b0a922105a660ee73d1f38bf1b4f4bbc92848d4a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-27 10:58:44.000000000","message":"Uploaded patch set 20: New patch set was added with same tree, parent tree, and commit message as Patch Set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"90219c12bcdabfe89e2875d3053cbeee9f6f0699","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-27 11:10:38.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"a2990b9c48647777d5c54647d29a3fa9e1ec4308","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-27 11:16:35.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"55a1690e13502400254ca5cc0a2aff0a527febe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 13:59:09.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f405bcac78db45dfa674eb6a98dd399a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f4ebbe8593c84dcaae136b644d1b4abd : SUCCESS in 2m 20s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e754ca7a8a54730928c06a19643670e : FAILURE in 9m 53s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ae5e4c5c7e994fbcb95ba4855f5b6918 : FAILURE in 1h 34m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/66fb058578e64baaa7d83985fdd305a5 : FAILURE in 1h 50m 47s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/edd7e5920c204232b5321dec6a0b98a7 : FAILURE in 1h 06m 40s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f483b48cadd649b397740ba12fdc4f16 : FAILURE in 1h 56m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/470d9f8a6e8245509042c893a2301b2d : FAILURE in 55m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c65fa7b2b43d484c865c299c296fd72f : FAILURE in 47m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd8704e91e5848a1b2bbf7e1812cac51 : FAILURE in 58m 41s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/38f364e6293143e1976e53024297aef1 : FAILURE in 1h 53m 22s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3b1490dfc9474b0aa583d411ab2161ec : FAILURE in 2h 31m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/724712905ceb42b896fa1cf6862b11d6 : FAILURE in 2h 33m 56s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/565febdc3c9543af85bbfadaa33c4d4c : FAILURE in 1h 28m 58s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffe78add53a34a00a2b43a9445d27130 : FAILURE in 2h 28m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6cb61a09724bae8356fdd1eff25367 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c46237f4601c4889b5818524dd5044c2 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":22},{"id":"0c56d66cbc0a010ad94f9078584cc66d266c1782","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-27 14:38:07.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"d1368a67f5377bc504b2edfbb9b5bc7d02e8bf37","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-27 14:38:58.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"c4f7acef8f5558f191adf92643a18d545075337f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 16:45:42.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e4f91234ee547c4a25c00dfa2fe5fea\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d088da78585146bb901c15e8f504f19b : SUCCESS in 2m 30s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a5d7abb5cc04908b396992c4dfccfbe : FAILURE in 10m 18s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e61a5f699ebf46adbb93405f2c8caa4e : FAILURE in 49m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/dd8508c5e03f4e28820b9b678a2b6662 : FAILURE in 1h 19m 00s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/56fc83b2e0e443018f778129a05e3106 : FAILURE in 1h 25m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9307c4d8642640aa83d1f606f2cdd5fd : FAILURE in 1h 18m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/217d0b1f150345d7a0413106f8e9f105 : FAILURE in 55m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d76215c0ea045c58955fbcc5fe08f34 : FAILURE in 51m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e66a35abeeb4bdb92e4eb7ac4d174c1 : FAILURE in 54m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cd1b8e2e56fc4c969aa7211d7685bd31 : FAILURE in 28m 07s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0cc24a9d2e11456c897f63c6ca040042 : FAILURE in 2h 01m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3ec990a272a4922ab0098ecda91e9b1 : FAILURE in 1h 55m 15s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bfd3bb4f4894749af05b3dea766f5bd : FAILURE in 2h 01m 20s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8542a6cee922465387ed9fa957e4ff77 : FAILURE in 1h 16m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4681c3a4e3504768b093ab229c89f66e : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04e9877258de4f9299c5cdbef6dfeaeb : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":24},{"id":"0a68956291196ff7920b4a9c722880ac6f0ba4fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-28 08:23:17.000000000","message":"Patch Set 25: Published edit on patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"668db69548cf1870d5f86460a24e967c940c367b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-28 08:34:42.000000000","message":"Patch Set 26: Commit message was updated.","accounts_in_message":[],"_revision_number":26},{"id":"fdcacb492a764b03579d4b9743a81f098cb2694c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-28 11:26:00.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3fa334c20804f959ff5a135116376d5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7d815301b4204423a363b91d65745868 : SUCCESS in 3m 12s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a27d6a7cc194af7adb5a8ebaa645fd5 : FAILURE in 9m 02s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/027439fa44c14dc190c8407bece5e5d3 : FAILURE in 1h 02m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0d436a423f86485283647a462f44f865 : SUCCESS in 1h 02m 54s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/12a9dc057d754c26a5fed21d16c0b370 : SUCCESS in 1h 07m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bcde56360ec24fc286e7dde926b2ae08 : FAILURE in 1h 05m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ced2fc31ea1a4601a77b945e63853e1a : FAILURE in 1h 08m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3c0896ce240402092ae36b36a242bc5 : SUCCESS in 1h 55m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8d0533622814270a48b1d8878ca6239 : FAILURE in 1h 26m 55s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1039dcc5ff114565b7949abb0613c391 : FAILURE in 1h 18m 06s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d6b21b8bf95b45138fb5b06f35f91455 : FAILURE in 1h 17m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c49c9718b6c4e3baf986f11bce1ea73 : SUCCESS in 1h 57m 04s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5994f227df76428f92841aad69131ac3 : SUCCESS in 2h 32m 17s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79f96aa5cca1456d9292df539af96653 : FAILURE in 2h 22m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6973117275b14e2bae21deb02085da81 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1114361641bb4290bb5bd3684519675e : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":26},{"id":"86f520a792389a32073092b699308175431b296d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-29 06:59:03.000000000","message":"Patch Set 27: Published edit on patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"cbe89b549012fbf2195933c90ae9ab7b34d255a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 09:46:11.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b089d28fb3124f84aa1e241b06264ecb\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4bc27aaca6e4408b8b35e62f7a5f0140 : SUCCESS in 2m 58s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/831943a91c634245b4645b9c84223eb0 : FAILURE in 7m 02s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9a90e44d5cb8497d9cbe0fc08cc52601 : FAILURE in 1h 22m 55s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/47c7a398d35e4e14927eb5458bc6161b : SUCCESS in 2h 39m 55s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a9c0c910ead64769a076d62335390771 : SUCCESS in 2h 40m 01s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8579e8c5b2224c3f98520dcda01ba335 : FAILURE in 2h 09m 45s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ab806fdf54c41c798d9111528630054 : FAILURE in 1h 16m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b311594b7b9c488593e061cb612af01a : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fc2cd0cbdc143748077ab1edefd9115 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":27},{"id":"037350179388eafce331ee35270552e099d48bb0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-29 18:39:55.000000000","message":"Patch Set 28: Published edit on patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"17bf8ad0cda00bf7ead168122d2b8832c960adbf","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-29 19:39:45.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":28},{"id":"8d7fb24097c34bf1defd6210a59551d77862ce91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 20:48:40.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac3c79db60a94a1c843faa2120fbeabe\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec96cca617f749d2970adf275355dfec : SUCCESS in 1m 45s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aac9a2471d6d45c4bba4820d844e6837 : SUCCESS in 8m 53s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e4c0c6143a7541338386ca42f520dd4d : FAILURE in 2h 00m 41s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c947ad2e501f46ffa33fcdb0c7e8b686 : SUCCESS in 1h 26m 56s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/75bd22e7d36b4eff80f714b92b9c3a8e : SUCCESS in 1h 38m 36s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b315a40e087443c29c1c4995e594f3bf : FAILURE in 1h 34m 51s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b9d3fa0e856495791d1b1abbabd1107 : SUCCESS in 1h 30m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d534da2e584746ce951bcee13cedc50e : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7084c7f19d0b48e0a1a0e11228ca188a : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":28},{"id":"cb3cf539f659fd33c85daa9d00dacdf12a68d1e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 05:35:55.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"0480f05559823a647469512877875172a1d084b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 07:24:59.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"54699cf3c2be66ed02ccba7c625fcdd38daa7f0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 07:25:57.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"a7e0249a5284a76eaf5ba685bb0c32da7aad3ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 07:26:25.000000000","message":"Uploaded patch set 32: Commit message was updated.","accounts_in_message":[],"_revision_number":32},{"id":"a4b260450665ab5700b804e5d63a83aea71c871e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 07:27:29.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"a0e24d9b5926c53eeaf7e55f400cbd29cdf4080a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 09:59:37.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68912f0caf3a484f916b0964eba8ff1c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/58ab537b60464180b2f2b580db7c6eff : SUCCESS in 1m 36s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b141f04bd4be4c6695a58a47ddf323fd : SUCCESS in 5m 58s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3ec68e1e5b4241dc8d22bf115f89f09d : SUCCESS in 1h 35m 21s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c5eba8a3686a4399bb097fdbd2b9315a : SUCCESS in 2h 17m 14s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c9a1fe66151247bd934c0461cfe99757 : SUCCESS in 2h 28m 47s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/87021a9f7a364cd7ab744c5d169292de : SUCCESS in 1h 29m 16s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd1e8324d6384280b4f0c37d68be6823 : SUCCESS in 2h 07m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4497ee36fa04401bb84449c772760dd1 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/791b723e81f94c16b7d03867fc311975 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":33},{"id":"9d5cc7794416b72ce360e5fb6844c43a89f161c5","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-31 15:38:03.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"d2847b8172bc20e0a5587344846e486f9c7d2953","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 15:52:28.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"5de9b54db312491675b6e45be5517775f2af8b0c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 15:53:40.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"0ce41488104a14842c34e6409db6a345536c3b9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 15:56:05.000000000","message":"Patch Set 34: Published edit on patch set 33.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"58a494a51d50b91b43028d08f3eabb6f743dbd63","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-31 16:06:09.000000000","message":"Patch Set 34: Code-Review+2","accounts_in_message":[],"_revision_number":34},{"id":"b9f6e7da5093206d07753c6c943c080f28e9c54a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-31 17:08:08.000000000","message":"Patch Set 34: Code-Review+1","accounts_in_message":[],"_revision_number":34},{"id":"bac3df9501d3a486dd2843ba1232ee48b2e1fa71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 19:11:03.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb685c53ad8f4dae82ec40a81f7aa757\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/74c48723383640f18002ede7d2007811 : SUCCESS in 1m 42s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f59bcb6359c418bb8610525f1142209 : SUCCESS in 9m 18s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/16b4d149a83d4da0b2a3eda4222f25c4 : SUCCESS in 1h 28m 38s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9ff0f6951f40448c81a7ea8222b7757f : SUCCESS in 2h 33m 01s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/470f12c538e2436885cfe6613b2b81b8 : SUCCESS in 1h 33m 01s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0dfb7e8091bf4e46bac80800a3d23ebd : SUCCESS in 1h 33m 34s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/257ef66d9f064cbe9be87f718cab1b1e : TIMED_OUT in 3h 07m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b041bd0cac3414b86b199adefa56c22 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e57d45360f3e44ceb4f473c3dbe2855b : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":34},{"id":"c59ba0381d774582288a80b84141f5eb3af5094c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-05 08:57:27.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"d60551077ad8fc0b9f6e47ec704feae12bec6151","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 11:52:11.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/759d84ded69544b49f242c01439de2a6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d697fc5e4475464ba2e571e995f9c1ad : SUCCESS in 1m 27s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ad110893a5349d5b9f394804fa87cdb : SUCCESS in 7m 45s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/db3d818d2f3542699033679697eea63d : SUCCESS in 2h 19m 19s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c078023b07684d5ba24802532883a83b : SUCCESS in 2h 38m 55s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d5ad4c615c5b4f1593bb4bd815734f31 : SUCCESS in 2h 42m 40s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b95336147eae4a9ebb5eadbc97ad88fd : SUCCESS in 2h 19m 56s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f651eaca921b40c7ab75383b968c0890 : SUCCESS in 2h 47m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec010582979a42309b91a5b0ab65fede : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7374926bdb8a4e8c81718c9126006b75 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":34},{"id":"8b094cd4c429042782c2b146bc8535cdd53bb473","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-13 10:44:11.000000000","message":"Patch Set 34: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":34},{"id":"bce061645c7e5c4d26121cbda2abf5d3642dc2b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 10:47:36.000000000","message":"Patch Set 34: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":34},{"id":"dfd24db3dba4f3861f748eaddf4fa72842e8bc87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 14:15:38.000000000","message":"Patch Set 34: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/988d79fda8454266ba68371bed5e4717\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/354e562d8390447281f6aca11279e446 : SUCCESS in 3m 53s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3067bb1be8464dd294657ab46dc3631a : SUCCESS in 14m 02s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4b9b8bdb2e584375849c35fbfe320bc8 : FAILURE in 2h 18m 46s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e5b36a263310418492390bfc8ab6a21b : FAILURE in 2h 37m 22s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/cc0dde9592484d8aaf71be459bef406d : FAILURE in 3h 13m 45s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9c80b0eac2ea450d9959cf36d3199c11 : FAILURE in 1h 54m 25s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edb1569b29f7444187ed83a2c8010db8 : FAILURE in 2h 43m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/453069b59bfd4af4b0c0a2fdb199e6a8 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b7a23487a084776b323152a3d57820e : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":34},{"id":"8aafa0346dff9e1c8f84cd851622d286be221ff4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-13 21:36:14.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"8fa0128a7d131a26cb93f09ad367afc12a197271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 00:32:11.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcfe709af04b4cb4addb640a963301b3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7e97d8d7d2c84187a448ee98e57506a5 : SUCCESS in 2m 10s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a615702d2c6b47d59787598199cb1785 : SUCCESS in 8m 44s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/04356340b64d4f5897d2daccaca46e21 : FAILURE in 2h 40m 51s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/455d3e78a4f64b2192ccd79ac80a8fb2 : FAILURE in 2h 29m 04s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/bd1556488e4b4f8b8eccb558583c1d50 : FAILURE in 2h 42m 35s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ceed10bceec84b689d91b6836fb61269 : FAILURE in 2h 51m 06s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/128e12f6db1448d49374192bb4dabd06 : FAILURE in 2h 47m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610a8eb29abc464682cb702b1e96eab6 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/465aef5d6bf04e6194889ebd02001273 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":34},{"id":"017ba6466efa7b6871fcfe83bd0540903ded7057","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-14 15:56:41.000000000","message":"Patch Set 35: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":35},{"id":"9fa7f45b3b6cbcf59e1f99d346e4f65b59fa88db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 19:15:04.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1c68034c4cd4a3985e18eb229e1e025\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7704c9a59f3b4f13b57fd3b3cd3701b7 : SUCCESS in 2m 44s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d0560d7219e4d8f9d062b7d9f778c54 : SUCCESS in 12m 34s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/0fab3e26e44d4b6c99529e42978b5ee1 : SUCCESS in 2h 16m 02s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0fc5d97c24364a4cb8c45b2a9e198acc : SUCCESS in 2h 56m 48s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/5e922c77b1c44a55bddbbb52879fded8 : SUCCESS in 2h 44m 09s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/45453ade57254b3c833865d60d811a4d : SUCCESS in 1h 33m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4807b01db4f04e0884b5bf03827d6577 : SUCCESS in 3h 08m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85df2e287b0b49a4a549f4571bf76675 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd8f35186943423199b3a12a25d0cff5 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":35},{"id":"c2fdaef39724d0f9c58f898c547a2e7fb3c70b17","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-15 23:09:31.000000000","message":"Patch Set 35: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"41acdda026efe3fa1c45e4a4fd9d382aa4ea4300","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 23:12:34.000000000","message":"Patch Set 35: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":35},{"id":"b39df2e569e44e1b0841807e1f50ab715bb7ed5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 02:04:02.000000000","message":"Patch Set 35: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb68d7b2564a49f8bd7e46c40279607b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4ce022c6e96545838724d58b1cb57cc3 : SUCCESS in 2m 05s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a460f1d8eabb47daae1206343487612d : SUCCESS in 7m 54s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/1bf9a922720640fb9b013c91046894bf : SUCCESS in 2h 05m 31s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8bc0233b7d32466da6a367962b4815ba : SUCCESS in 2h 33m 28s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3d1cdd3bb5bc4601b292f2b9b74dae65 : SUCCESS in 2h 46m 57s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e828a8b2e0f341cdbb7aa19baa34a184 : SUCCESS in 2h 20m 03s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eaeab6eb8fa7427bbf564d2e93867225 : SUCCESS in 2h 40m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b051b7ca8b5e46dfb6b9df339c0de5ae : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ddcb67b328b4b8290e2202e92d3841f : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":35},{"id":"c25a2a403b33ceea75c8cc919430ab62754160ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 02:04:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":35},{"id":"99aa6a97478dba729b3d1ba3e43d05c01536ecbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 02:13:32.000000000","message":"Patch Set 35:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1140c6de4be24200a49c3f2ef0e96a65\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2597ac14ebd41bfb09d99416cbc687c : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":35}],"current_revision_number":35,"current_revision":"7b89b1f162935d1d918f4b1d8b8f720130a4c519","revisions":{"8dd7bbb34e73f5cb25b956800703ab309b42eadb":{"kind":"REWORK","_number":1,"created":"2026-02-05 13:11:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/1"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-05 13:11:05.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/8dd7bbb34e73f5cb25b956800703ab309b42eadb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/8dd7bbb34e73f5cb25b956800703ab309b42eadb"}]},"branch":"refs/heads/master"},"78f5a5412fb123877ff0ab0c1b2022d276022c5c":{"kind":"REWORK","_number":2,"created":"2026-02-05 15:19:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/2"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-05 15:19:09.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/78f5a5412fb123877ff0ab0c1b2022d276022c5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/78f5a5412fb123877ff0ab0c1b2022d276022c5c"}]},"branch":"refs/heads/master"},"15171bbe264eaf814482d6609664427f37693077":{"kind":"REWORK","_number":3,"created":"2026-02-05 15:21:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/3"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-05 15:21:05.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/15171bbe264eaf814482d6609664427f37693077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/15171bbe264eaf814482d6609664427f37693077"}]},"branch":"refs/heads/master"},"728629cdf3fdac8db51f6b473b1ea831e4c54b11":{"kind":"REWORK","_number":4,"created":"2026-02-05 15:25:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/4"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-05 15:25:20.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/728629cdf3fdac8db51f6b473b1ea831e4c54b11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/728629cdf3fdac8db51f6b473b1ea831e4c54b11"}]},"branch":"refs/heads/master"},"d2c06a858a2f45d48b7436c9d4e96d96d8eebfe1":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-02-06 15:03:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/5"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-06 15:03:15.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d2c06a858a2f45d48b7436c9d4e96d96d8eebfe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d2c06a858a2f45d48b7436c9d4e96d96d8eebfe1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7751d125798ac043e48d1be3498a06100dbdfd86":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-02-25 09:45:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/6"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 09:45:30.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/977841\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/7751d125798ac043e48d1be3498a06100dbdfd86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/7751d125798ac043e48d1be3498a06100dbdfd86"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e109e518c5f579b616a875d8b2bf894e39015d86":{"kind":"REWORK","_number":7,"created":"2026-03-12 10:29:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/7"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-12 10:29:40.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e109e518c5f579b616a875d8b2bf894e39015d86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e109e518c5f579b616a875d8b2bf894e39015d86"}]},"branch":"refs/heads/master"},"74f979b6a67a78645bf120d8ce55f45f4971867a":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-03-12 16:08:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/8"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-12 16:08:39.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/74f979b6a67a78645bf120d8ce55f45f4971867a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/74f979b6a67a78645bf120d8ce55f45f4971867a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e06eb48eefb85e04b8b4b3e73b0dc06c4f3b0284":{"kind":"REWORK","_number":9,"created":"2026-03-13 11:46:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/9"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-13 11:45:58.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e06eb48eefb85e04b8b4b3e73b0dc06c4f3b0284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e06eb48eefb85e04b8b4b3e73b0dc06c4f3b0284"}]},"branch":"refs/heads/master"},"22019bb002865a736928c54ddd694015a41d9a35":{"kind":"REWORK","_number":10,"created":"2026-03-13 13:48:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/10"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-13 13:48:08.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/22019bb002865a736928c54ddd694015a41d9a35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/22019bb002865a736928c54ddd694015a41d9a35"}]},"branch":"refs/heads/master"},"052f152aa35197755879c94eabe99a1386c0a5a8":{"kind":"REWORK","_number":11,"created":"2026-03-13 13:52:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/11"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-13 13:52:07.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/052f152aa35197755879c94eabe99a1386c0a5a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/052f152aa35197755879c94eabe99a1386c0a5a8"}]},"branch":"refs/heads/master"},"fddeacf5caa82f80c2b7bddd423b8c5cbdf5a6c9":{"kind":"REWORK","_number":12,"created":"2026-03-13 13:57:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/12"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-13 13:57:36.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/fddeacf5caa82f80c2b7bddd423b8c5cbdf5a6c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/fddeacf5caa82f80c2b7bddd423b8c5cbdf5a6c9"}]},"branch":"refs/heads/master"},"02533b536ca0547288d596efd232ce825f7dd4db":{"kind":"REWORK","_number":13,"created":"2026-03-13 14:20:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/13"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-13 14:20:00.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/02533b536ca0547288d596efd232ce825f7dd4db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/02533b536ca0547288d596efd232ce825f7dd4db"}]},"branch":"refs/heads/master"},"2e4b43cc7384106e75cb153cd01b43cec89b0f9a":{"kind":"REWORK","_number":14,"created":"2026-03-20 14:28:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/14"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-20 14:28:49.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/2e4b43cc7384106e75cb153cd01b43cec89b0f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/2e4b43cc7384106e75cb153cd01b43cec89b0f9a"}]},"branch":"refs/heads/master"},"506e18bc79b5dcab4b2719412a8ea170df788e0c":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2026-03-20 14:29:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/15"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-20 14:29:01.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/506e18bc79b5dcab4b2719412a8ea170df788e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/506e18bc79b5dcab4b2719412a8ea170df788e0c"}]},"branch":"refs/heads/master"},"63ee867b0b515b10e01951cf01816b027550a203":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2026-03-20 15:06:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/16"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-20 15:06:01.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/63ee867b0b515b10e01951cf01816b027550a203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/63ee867b0b515b10e01951cf01816b027550a203"}]},"branch":"refs/heads/master","description":"Edit commit message"},"22b20b98c0461ff08eeb908031425dfeaa539e52":{"kind":"REWORK","_number":17,"created":"2026-03-26 09:31:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/17"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-25 20:23:45.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/22b20b98c0461ff08eeb908031425dfeaa539e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/22b20b98c0461ff08eeb908031425dfeaa539e52"}]},"branch":"refs/heads/master"},"9cecd94d8b1753e36d7c9c4599d067d41fc7ba8e":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2026-03-26 13:30:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/18"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-26 13:30:46.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/9cecd94d8b1753e36d7c9c4599d067d41fc7ba8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/9cecd94d8b1753e36d7c9c4599d067d41fc7ba8e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ffdc166ba04e30bff24ab0bb6f66b56438b92600":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2026-03-26 14:55:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/19"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-26 14:55:17.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ffdc166ba04e30bff24ab0bb6f66b56438b92600"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ffdc166ba04e30bff24ab0bb6f66b56438b92600"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0432fe086c0c41fa905aea582e0a9add5b7310da":{"kind":"NO_CHANGE","_number":20,"created":"2026-03-27 10:58:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/20"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 10:58:18.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/0432fe086c0c41fa905aea582e0a9add5b7310da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/0432fe086c0c41fa905aea582e0a9add5b7310da"}]},"branch":"refs/heads/master"},"73417228e032c846d25c6faeea7e3526542447f4":{"kind":"REWORK","_number":21,"created":"2026-03-27 11:10:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/21"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 11:09:26.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/73417228e032c846d25c6faeea7e3526542447f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/73417228e032c846d25c6faeea7e3526542447f4"}]},"branch":"refs/heads/master"},"b16e7ac3e265113653046b776b34ad885cdd070e":{"kind":"REWORK","_number":22,"created":"2026-03-27 11:16:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/22"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 11:16:29.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/b16e7ac3e265113653046b776b34ad885cdd070e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/b16e7ac3e265113653046b776b34ad885cdd070e"}]},"branch":"refs/heads/master"},"89738b1c45282422f2c9d940d72a17fcb8a872e2":{"kind":"REWORK","_number":23,"created":"2026-03-27 14:38:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/23"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 14:38:00.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/89738b1c45282422f2c9d940d72a17fcb8a872e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/89738b1c45282422f2c9d940d72a17fcb8a872e2"}]},"branch":"refs/heads/master"},"05b01ead683037333047a1f1d85f70e42838bb82":{"kind":"REWORK","_number":24,"created":"2026-03-27 14:38:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/24"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 14:38:53.000000000","tz":60},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/05b01ead683037333047a1f1d85f70e42838bb82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/05b01ead683037333047a1f1d85f70e42838bb82"}]},"branch":"refs/heads/master"},"d66bc4c0589a617c1dcc59496a03c926e92c327c":{"kind":"REWORK","_number":25,"created":"2026-03-28 08:23:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/25"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-28 08:23:17.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d66bc4c0589a617c1dcc59496a03c926e92c327c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d66bc4c0589a617c1dcc59496a03c926e92c327c"}]},"branch":"refs/heads/master"},"dd8dda984df99f092c34153832029d8e8f17a23d":{"kind":"NO_CODE_CHANGE","_number":26,"created":"2026-03-28 08:34:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/26"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-28 08:34:41.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_glance/+/982526\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/dd8dda984df99f092c34153832029d8e8f17a23d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/dd8dda984df99f092c34153832029d8e8f17a23d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3ae02303cfe69795324fb03cb8205b945fc40f0d":{"kind":"REWORK","_number":27,"created":"2026-03-29 06:59:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/27"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-29 06:59:03.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_glance/+/982526\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/3ae02303cfe69795324fb03cb8205b945fc40f0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/3ae02303cfe69795324fb03cb8205b945fc40f0d"}]},"branch":"refs/heads/master"},"28b73531ea5133e6b4143b1fb5b43f015731676f":{"kind":"REWORK","_number":28,"created":"2026-03-29 18:39:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/28"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-29 18:39:54.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_glance/+/982526\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/28b73531ea5133e6b4143b1fb5b43f015731676f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/28b73531ea5133e6b4143b1fb5b43f015731676f"}]},"branch":"refs/heads/master"},"1d23376b1b6c0ce33863e3168f8157d270a8785c":{"kind":"REWORK","_number":29,"created":"2026-03-31 05:35:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/29"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-30 16:53:31.000000000","tz":120},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/1d23376b1b6c0ce33863e3168f8157d270a8785c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/1d23376b1b6c0ce33863e3168f8157d270a8785c"}]},"branch":"refs/heads/master"},"ad852a8abc38d341e1b5782eddd80aba1bd448e0":{"kind":"REWORK","_number":30,"created":"2026-03-31 07:24:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/30"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 07:24:51.000000000","tz":120},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ad852a8abc38d341e1b5782eddd80aba1bd448e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ad852a8abc38d341e1b5782eddd80aba1bd448e0"}]},"branch":"refs/heads/master"},"ebf850e2c8ea12bfb50d8b24219bd67918d3a3d8":{"kind":"REWORK","_number":31,"created":"2026-03-31 07:25:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/31"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 07:25:52.000000000","tz":120},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ebf850e2c8ea12bfb50d8b24219bd67918d3a3d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ebf850e2c8ea12bfb50d8b24219bd67918d3a3d8"}]},"branch":"refs/heads/master"},"511098d4bb31820a509827e95180f8b48e2df83b":{"kind":"NO_CODE_CHANGE","_number":32,"created":"2026-03-31 07:26:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/32"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 07:26:11.000000000","tz":120},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_glance/+/982526\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/511098d4bb31820a509827e95180f8b48e2df83b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/511098d4bb31820a509827e95180f8b48e2df83b"}]},"branch":"refs/heads/master"},"6b5632f5b4f381195600d21b0a5268fc22e29390":{"kind":"REWORK","_number":33,"created":"2026-03-31 07:27:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/33"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 07:27:24.000000000","tz":120},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_glance/+/982526\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/6b5632f5b4f381195600d21b0a5268fc22e29390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/6b5632f5b4f381195600d21b0a5268fc22e29390"}]},"branch":"refs/heads/master"},"d54b2b3b0125e7c7c27b3bf3e23a3cec58976a44":{"kind":"REWORK","_number":34,"created":"2026-03-31 15:56:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/34"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 15:55:39.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/975512\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/981533\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/981496\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/975793\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_glance/+/982526\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d54b2b3b0125e7c7c27b3bf3e23a3cec58976a44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d54b2b3b0125e7c7c27b3bf3e23a3cec58976a44"}]},"branch":"refs/heads/master"},"7b89b1f162935d1d918f4b1d8b8f720130a4c519":{"kind":"NO_CODE_CHANGE","_number":35,"created":"2026-04-14 15:56:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/975794/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/94/975794/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/94/975794/35"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 13:45:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-14 15:56:41.000000000","tz":0},"subject":"Place content for os_freezer role","message":"Place content for os_freezer role\n\nThis takes state of the externally managed role [1] for\nos_freezer as the base and updates it with required changes to work\nwith modern OSA version, as well as ansible-core 2.19.\n\n[1] https://github.com/noonedeadpunk/openstack-ansible-os_freezer\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/984593\nChange-Id: Icfc63d099183460a8ef706d2a017fafc759cab17\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/7b89b1f162935d1d918f4b1d8b8f720130a4c519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/7b89b1f162935d1d918f4b1d8b8f720130a4c519"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
