)]}'
{"id":"openstack%2Fopenstack-ansible-os_freezer~975805","triplet_id":"openstack%2Fopenstack-ansible-os_freezer~master~Idbfbfac4bf69a2f8a02e7e57d07da85c7f9b681d","project":"openstack/openstack-ansible-os_freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-17 06:50:05.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-17 06:50:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idbfbfac4bf69a2f8a02e7e57d07da85c7f9b681d","subject":"Add contributors guide","status":"MERGED","created":"2026-02-05 14:45:10.000000000","updated":"2026-04-17 06:50:54.000000000","submitted":"2026-04-17 06:50:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975805","meta_rev_id":"db027144ca93213e2f4a894fd92dfa7bff7835b7","_number":975805,"virtual_id_number":975805,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-17 06:50:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-04-17 06:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-15 15:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-04-17 06:34:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-29 19:50:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 08:00:23.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-13 10:56:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"7bbd6dc10a0d6ccb9bce830a89dfc4f252f78fbe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-05 14:45:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60c558d71c4045d497513f930cfa97fbfa798b34","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-05 14:45:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c17eabb8b463b37eb355ec937fb8c9b737987329","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-05 14:46:47.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ebb800774d72a57bc0681801dab80f1d219cf178","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-05 14:47:10.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"e658b1b8133a172cb2460591d711a79133432442","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-05 14:47:13.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"7bba1f399894cd7048c504e175041ac89455b9b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-29 19:39:50.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_37598\u003e","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":5},{"id":"e04a94f0e9a207c153c13196d3f15f6633fe1064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 19:50:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8be77d6a105943c88b86290400338d1a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bbdcc11de4f42c483b5c7cd22f68134 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32464d3e565544799d4922a1a3715771 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":5},{"id":"0bc67b1485a019e63a68b1cae8574ff853beaae9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 10:08:05.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_37598\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":6},{"id":"61296bb3dbee9e9906983dfc66096a0d297f4392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:29:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d821db50fb1946e2a54c70b658f662b0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba3fccf72e1343dd83f611695591257e : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31a81c29c294ec28ca4a16a1743ea14 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":6},{"id":"b3db174b797148c9a5a6e998a83088ec99c35fbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-10 14:10:58.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_37598\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":7},{"id":"df6e4949f8ee920fdac6a04ea00b3459170c85a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 14:42:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e00d15d67dbd42d3a0f8688dac086028\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be71951bef7c4404a9fd8c58f8873aef : SUCCESS in 14m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab25ef003e04adbad885e5a8af1d7f4 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":7},{"id":"d1b209b60c8d1f2250bb71969cab37d5b6458e6c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-13 08:00:23.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"fb08d7595d172289dbc699aeaa1430dc0bbb64d9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-13 10:56:26.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"ecc8c4023b81cf5b85168e8bbffed220740fe8ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 11:00:52.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"bf30c3aed2419ceed384d9249e23ca36e09bf7f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-15 15:29:17.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_37598\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":8},{"id":"a380c4cb32fd2a02839e893c56dc592fd0c35aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 16:01:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b20dea27e354a48abd57c8b12e6c9a5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d0dcf8ad77a42f2bfbfac4e01400420 : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b29e718c0654f3bb43bb979b91631a3 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":8},{"id":"26640de69f2b49da9eee54d064e785d7c51b7b47","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-16 09:01:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"92019427a6dbc63b8b24a4a8b354f6b03c1fac0c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-17 06:34:43.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"578ebea9b893909c5b1daee06e32c3556cca156a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 06:37:15.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8d423a0a1ef66d630680ed4d482b6dfa2c66e6cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 06:50:05.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2230405a14ac461b9ac74355ba8450ad\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c3a5d4a14834cfeac8e21ede46c9dff : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a176743da2d543879c86a52f8304bdfb : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":8},{"id":"493d705e7e8146c4dfc3936dd9fcde4210cf1904","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 06:50:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"db027144ca93213e2f4a894fd92dfa7bff7835b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 06:50:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a7c7886cdff43128ca0f94e42530c0c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/916cdbc308f9481a8f504cdd23c3ce8a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"3e00a18fee82f3b0ba7e7817475f3e67ef2a78a8","revisions":{"0ffce04c6b2ab93cb1e76a0ae864b179278f4d1b":{"kind":"REWORK","_number":1,"created":"2026-02-05 14:45:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/05/975805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/05/975805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/1"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:45:10.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:45:10.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Idbfbfac4bf69a2f8a02e7e57d07da85c7f9b681d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/0ffce04c6b2ab93cb1e76a0ae864b179278f4d1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/0ffce04c6b2ab93cb1e76a0ae864b179278f4d1b"}]},"branch":"refs/heads/master"},"fd2b5fc495e032ae108f313a2bd5335855a1f9db":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-05 14:45:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/05/975805/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/05/975805/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/2"}}},"commit":{"parents":[{"commit":"8dd7bbb34e73f5cb25b956800703ab309b42eadb","subject":"Place content for os_freezer role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/8dd7bbb34e73f5cb25b956800703ab309b42eadb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:45:10.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:45:50.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Idbfbfac4bf69a2f8a02e7e57d07da85c7f9b681d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/fd2b5fc495e032ae108f313a2bd5335855a1f9db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/fd2b5fc495e032ae108f313a2bd5335855a1f9db"}]},"branch":"refs/heads/master","description":"Rebase"},"4fddb3d1c7828791dfb3a51db6144b415ad25935":{"kind":"REWORK","_number":3,"created":"2026-02-05 14:46:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/05/975805/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/05/975805/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/3"}}},"commit":{"parents":[{"commit":"8dd7bbb34e73f5cb25b956800703ab309b42eadb","subject":"Place content for os_freezer role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/8dd7bbb34e73f5cb25b956800703ab309b42eadb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:45:10.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:46:47.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Idbfbfac4bf69a2f8a02e7e57d07da85c7f9b681d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/4fddb3d1c7828791dfb3a51db6144b415ad25935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/4fddb3d1c7828791dfb3a51db6144b415ad25935"}]},"branch":"refs/heads/master"},"e3dd6be1668255ef3b028d4e176743e95ce95f19":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-02-05 14:47:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/05/975805/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/05/975805/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/4"}}},"commit":{"parents":[{"commit":"8dd7bbb34e73f5cb25b956800703ab309b42eadb","subject":"Place content for os_freezer role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/8dd7bbb34e73f5cb25b956800703ab309b42eadb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:45:10.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:47:10.000000000","tz":0},"subject":"Add contributors guide","message":"Add contributors guide\n\nChange-Id: Idbfbfac4bf69a2f8a02e7e57d07da85c7f9b681d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e3dd6be1668255ef3b028d4e176743e95ce95f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e3dd6be1668255ef3b028d4e176743e95ce95f19"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c190bf6dd0eb90afc70724b29f5e29a7cee82fa1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-29 19:39:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/975805/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/05/975805/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/5"}}},"commit":{"parents":[{"commit":"28b73531ea5133e6b4143b1fb5b43f015731676f","subject":"Place content for os_freezer role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/28b73531ea5133e6b4143b1fb5b43f015731676f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:45:10.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-29 19:39:50.000000000","tz":0},"subject":"Add contributors guide","message":"Add contributors guide\n\nChange-Id: Idbfbfac4bf69a2f8a02e7e57d07da85c7f9b681d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/c190bf6dd0eb90afc70724b29f5e29a7cee82fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/c190bf6dd0eb90afc70724b29f5e29a7cee82fa1"}]},"branch":"refs/heads/master","description":"Rebase"},"41a8440bee81682e657602dcc5404a6f325690d8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-03-31 10:08:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/975805/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/05/975805/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/6"}}},"commit":{"parents":[{"commit":"6b5632f5b4f381195600d21b0a5268fc22e29390","subject":"Place content for os_freezer role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/6b5632f5b4f381195600d21b0a5268fc22e29390"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:45:10.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-31 10:08:05.000000000","tz":0},"subject":"Add contributors guide","message":"Add contributors guide\n\nChange-Id: Idbfbfac4bf69a2f8a02e7e57d07da85c7f9b681d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/41a8440bee81682e657602dcc5404a6f325690d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/41a8440bee81682e657602dcc5404a6f325690d8"}]},"branch":"refs/heads/master","description":"Rebase"},"0fac182fed4873051aecc61f23209528f753cb6f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-04-10 14:10:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/975805/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/05/975805/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/7"}}},"commit":{"parents":[{"commit":"d54b2b3b0125e7c7c27b3bf3e23a3cec58976a44","subject":"Place content for os_freezer role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d54b2b3b0125e7c7c27b3bf3e23a3cec58976a44"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:45:10.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-10 14:10:58.000000000","tz":0},"subject":"Add contributors guide","message":"Add contributors guide\n\nChange-Id: Idbfbfac4bf69a2f8a02e7e57d07da85c7f9b681d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/0fac182fed4873051aecc61f23209528f753cb6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/0fac182fed4873051aecc61f23209528f753cb6f"}]},"branch":"refs/heads/master","description":"Rebase"},"3e00a18fee82f3b0ba7e7817475f3e67ef2a78a8":{"kind":"NO_CHANGE","_number":8,"created":"2026-04-15 15:29:17.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/975805/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/05/975805/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/05/975805/8"}}},"commit":{"parents":[{"commit":"7b89b1f162935d1d918f4b1d8b8f720130a4c519","subject":"Place content for os_freezer role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/7b89b1f162935d1d918f4b1d8b8f720130a4c519"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-05 14:45:10.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-15 15:29:17.000000000","tz":0},"subject":"Add contributors guide","message":"Add contributors guide\n\nChange-Id: Idbfbfac4bf69a2f8a02e7e57d07da85c7f9b681d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/3e00a18fee82f3b0ba7e7817475f3e67ef2a78a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/3e00a18fee82f3b0ba7e7817475f3e67ef2a78a8"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"0fac182fed4873051aecc61f23209528f753cb6f","theirs":"7b89b1f162935d1d918f4b1d8b8f720130a4c519","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
