)]}'
{"id":"openstack%2Fopenstack-ansible-os_freezer~983870","triplet_id":"openstack%2Fopenstack-ansible-os_freezer~master~I5e0fb5c7377f9f22c52df28c4a11793311efba83","project":"openstack/openstack-ansible-os_freezer","branch":"master","topic":"osa/freezer","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-27 16:05:42.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I5e0fb5c7377f9f22c52df28c4a11793311efba83","subject":"Add support for centralized scheduler deployment","status":"MERGED","created":"2026-04-09 14:00:06.000000000","updated":"2026-08-27 22:24:44.000000000","submitted":"2026-08-27 22:23:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983870-osa/freezer","meta_rev_id":"5d3a340c37e70940b8cc8ec3e293bfb1811eabba","_number":983870,"virtual_id_number":983870,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-27 22:23:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-08-27 11:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-21 14:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2026-08-27 11:15:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-10 15:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-19 22:23:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-08-21 14:12:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-08-27 11:15:17.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"44c396dbdd8431de6631ee8c0b4c51b295ec6d20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-09 14:00:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a6e66e4a7de833e335d3520694c0c766f743f3b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-09 14:00:14.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"1e4780a695755049332de7758603d51980134759","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-10 13:57:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25cba8400fa7bb42828f6affbac33aed33bfd569","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-10 15:06:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"98f1636d4ee4ccc18d6d714a9b22fb0935858254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:08:15.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"fa9bea074521a4befaf8ea527cf1bcf5d0ab67b5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-10 15:08:20.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5e975f0503af65e99e33da8bed27016a1ee18249","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-10 16:08:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"35bc6632b457bdba5ea48b19456485422edd8ec8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-10 17:06:42.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"e312011bd15e343949974ae921d57d1d38134440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 19:38:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32a66126728840b387a6eeb12ce0dc7e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d26e7196984d49999e59b829954773f5 : SUCCESS in 9m 08s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/5570e431976c4782bce06262aa99efac : FAILURE in 1h 55m 35s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3be437689f214e74ae4a97cabbce7853 : FAILURE in 2h 15m 13s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/bf4da94fae7e41959165ff63edb93f7f : FAILURE in 2h 05m 38s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ad7c32595526418e812c92cecfb67849 : FAILURE in 1h 23m 47s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9eb2206600594e5e97712e1c0b5473ca : FAILURE in 2h 22m 33s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c63002d2519b402b84bde1425f7d98e0 : FAILURE in 1h 20m 26s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8f1dd96d00140568643b1d3820c8039 : FAILURE in 1h 58m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8cc85cb7812483eb39874eff7313ebe : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8e2d27ecd664853af0d4a50dc0350a2 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":5},{"id":"fbdb48707afe941d1a1790ed28cb70d8567992ad","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-18 08:00:33.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_28619\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"_revision_number":6},{"id":"2a31fcc935cc4e4485c573077272cd0a59e6cc8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 10:37:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81014034ee924de1bf5a10ab5fde842b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27e8f2bac23d4af78d31dcc996f8ddc9 : POST_FAILURE in 12m 17s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3df418093f0a40d0a4607cc235b3627c : POST_FAILURE in 1h 19m 32s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/32f13222f4334e1ea8536c39345ff3c6 : POST_FAILURE in 1h 13m 57s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/196b029a14ee4e349343787ea36715ce : POST_FAILURE in 2h 21m 37s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5b9b338e47584b27b26fbc24e1c7566f : POST_FAILURE in 1h 24m 34s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/becae5a259dc4cd690c9987d1c0d0f44 : POST_FAILURE in 1h 13m 17s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2d328aa2a5184cbd8aa4d08c62455d1a : POST_FAILURE in 1h 25m 20s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6164ad8e55b4eefb88bd19ae04e5914 : FAILURE in 1h 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/176fa8a799ef4c2eada7aac31abc98da : POST_FAILURE in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a72114776ed471ca9f7ad04ccb1cca5 : POST_FAILURE in 3m 36s","accounts_in_message":[],"_revision_number":6},{"id":"c7edbb5f802bce080e7d33274b093bf047d0dde9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-19 22:23:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b3bd94844ebb9500b0bcf5e67751ef9c11fd030d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 00:50:18.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf4f9dd4402a44949eddddd067c6858b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f41e003addb242f68c469a6ef3829497 : SUCCESS in 8m 56s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/af348d2dead8421892f874513cb07810 : FAILURE in 1h 50m 37s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b4740e2247c449c19753cf880d69abb8 : FAILURE in 2h 18m 54s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0a6d503341ca4e7e8246b8d704ed16ac : FAILURE in 1h 54m 30s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e628c361f3d643e1af1d169d6545719e : FAILURE in 1h 50m 58s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e334a8164b24c599db8164f54563cae : FAILURE in 1h 42m 08s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b6b0e21619364776b5fe971708df8f33 : FAILURE in 2h 01m 18s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba8b9d5f019043dcbe8dfca142c79a02 : FAILURE in 1h 43m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a73d9569f574b3d905e05e9d9cad97d : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0cf7f8e9da6448d8122e9d0ea41aebe : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":6},{"id":"5a4bd32c4657f62374dd7f4ade0801c3a8ce8a16","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-27 01:09:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"db53384d3bf0e3aa697510c347d5f9cdc20c190b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 03:55:09.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a36f032f000146b7ad7b30fb97cfa7ef\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30d069e4eadf42ecbb074c9145e12dc8 : SUCCESS in 8m 20s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/917b5138edad42f28b26c8b50f6a20e3 : FAILURE in 1h 15m 31s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ba23c3cb5a1f4e9c8983548f1a58f4ad : FAILURE in 2h 26m 17s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9e4d8833746246588fde71f4321d57d4 : FAILURE in 2h 22m 11s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eaa9eb4936f54b3a96d7d193942628a6 : FAILURE in 1h 26m 23s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5115ae95eba94733ad10c7fe38585ea9 : FAILURE in 1h 09m 28s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d5ad94f5f7394874b947e90526131a42 : FAILURE in 2h 36m 40s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/608e250694dc4943b19302ad679acd58 : FAILURE in 1h 58m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e021cd4452db425c81a30ab21d820b13 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6be808900a3a4ec9b7909c1713ecd975 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":6},{"id":"6808283734b7533005a4fb84a3e4f61daae23735","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-13 00:00:45.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_28619\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"_revision_number":7},{"id":"08316e4785b6a29c5217eca7e562cb9a7614879f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 02:33:24.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e569e00c12a4e6b86872048ce073e57\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/110293449aa7405c9b4a7b19b83a6dae : SUCCESS in 10m 10s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e586a062b66e4d6182f8df0c490349ec : FAILURE in 2h 25m 05s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7c0e13b402024aec8470599e4c4c3302 : FAILURE in 1h 38m 39s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/66997c4ad3c543f5a2c557b4a91edee3 : FAILURE in 2h 17m 00s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/36db0d6985574be9855b20eefef10769 : FAILURE in 1h 20m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/337beb5199424b7197250296ae0e68eb : FAILURE in 1h 40m 23s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/225aa1f9ce8c41dcb7e71dbe947722b2 : FAILURE in 1h 13m 51s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6aa92155d0ad4cbf9229bd5ea2e62430 : FAILURE in 1h 56m 09s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b4560bc111743f4bcb1d56ad5004572 : FAILURE in 2h 22m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6179a1273904a9c9ee6cf3381c87f6f : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12ec4a517bab4c12a3f13acf4d40ecbb : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":7},{"id":"0aee875db3b52c820eadd0aa6f20549d70599343","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-29 08:52:13.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"a9d5ac13446b9b35caafbf58a55c921745a7fa2e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-29 09:26:07.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"64c486231d7d3b395300a3fd704d721e624550bc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-29 09:29:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d422caf35f1ddbb48eed12d1fe18e2a8d41acf06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 11:32:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4301168a804403aa7ddcd2377864b8a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d2575490ca741048a0a9b0b1db25cbb : SUCCESS in 9m 19s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/0d4f07b432f643f6bf89b3414e080dc1 : FAILURE in 1h 11m 07s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8f2705d4e3604559a83347c67b76b718 : FAILURE in 2h 19m 43s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7ed4fe0831bf412cb3c0f13d52f26031 : FAILURE in 1h 14m 35s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/294eeec3b4f148c187b8f5456afd0651 : FAILURE in 1h 08m 03s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/45bb12aca1a345fea8aa33f5a0024d63 : FAILURE in 1h 05m 58s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7d881876f6474de198211a32571c01a6 : FAILURE in 2h 35m 39s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07da2360ffde42d9b3815f27dcb656f7 : FAILURE in 2h 01m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afb347bea8954a2f88f43067c65b8610 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fee87f60a1e417f923c72e3190b5ffc : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":8},{"id":"278774dcf4c0bd8a93a5f1f51c77d73cf5a7bcb8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 10:45:44.000000000","message":"Patch Set 8: Code-Review-1","accounts_in_message":[],"_revision_number":8},{"id":"8c374e68d6ad88a6e2167acc7c9a3362df6a175b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 10:50:20.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"b862809b2d2fac5c33a5e6553f0eec3c9a77de19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 11:59:32.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"8e70b41baf65a527aeb35592f7d2fba246b1a51b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 14:47:27.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"719e2b0ed16ef3258c0a8dbcf018eadc23aacd66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 15:00:28.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"d1ca131745436df680e3ca287197f31373710253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 22:16:43.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f92aa7d1036d4f4da27d458323254b5c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7cb0aa83e4c436e836dad43dd58c5ef : SUCCESS in 12m 39s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/fe3377cee55d40beaf0ce258709d1c36 : TIMED_OUT in 7h 08m 31s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/326976f783064b7ca636ab7ab965e91e : FAILURE in 1h 34m 28s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/933411a0cdc54867be58540a5d18da80 : FAILURE in 2h 09m 16s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7355623e30449cc8b48a310b6f21e05 : FAILURE in 2h 49m 29s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4df98c4ef2994c42b106e0487930797b : FAILURE in 2h 47m 33s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/16b4e3aba4c944acb69464acd577ef61 : FAILURE in 1h 35m 17s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19e0c47661d04c6f9d948ba123a56bed : FAILURE in 2h 45m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f878549255d4784815037b36c9f9253 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/537f692717a8476181056e8a648b5c70 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":12},{"id":"d896b512b7d62f20fdab45cc68fa5f7136849c7f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-02 07:21:05.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"92af5a76cccb97a044c95fbddbb3ab486b7ae1f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 09:13:55.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995696 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"ceaf9e1cb9d2008a00673b0b16dbf5c8d5a85426","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-02 09:14:11.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"28def0baace02aef1821da170e2260aa7ac2a073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 10:30:59.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible/+/995111 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"32dbea09b5eaac4ab5242287b4b12fc49d2eb5db","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-02 12:54:44.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"30e7307e543b44937a5a8da9ae01449c6e35b3e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 16:10:57.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40cf03708d6c41bebb2f56ed78552e08\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c6bc7dd6d3940ec8e5223ffe8aeaca1 : SUCCESS in 9m 28s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/aea463ccc9364883a09dce119ba9c2ec : FAILURE in 26m 42s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b502cbde098046b38c8d439a9a216a01 : FAILURE in 2h 09m 14s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0ad8e5e94d9348d28a110eb8a9269eea : FAILURE in 2h 57m 37s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c452a2e36bbe4fe4aaaf4026b94239bf : FAILURE in 2h 45m 14s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/556f4ac29b4240bdbc86947ccd9216bb : FAILURE in 2h 41m 50s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8e40a0ca831f40c5888942f385cf798e : FAILURE in 1h 46m 01s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdd645dfe6604795ba04ccf7b7a1c100 : FAILURE in 3h 04m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/161b520d21c4461a9eda2d762a58d2f5 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69fe7cc5c8864766b98bce819b2f2df8 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":12},{"id":"275c93b760071baf4053289f5f1e763dc5279cd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-02 18:52:50.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"3279a7b15ecb40fa1e303cf1232d38dfb8642102","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 22:20:13.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c12c30a8aa84993a1dcd55b69219c01\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/705e424ae53c402791479b4532ae0d01 : SUCCESS in 10m 45s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/91ba326dea6e4b2eb68d00ca7b5ee24e : FAILURE in 29m 08s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/08629b1baf9a41b2b79511373c2f0ffd : FAILURE in 2h 37m 41s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2740cc07b73f47c488cca6062cb5d168 : FAILURE in 3h 01m 37s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f3aa6be33f94f409ee8386535d2e184 : FAILURE in 2h 52m 00s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c6e9deb52e0c4a4b9bd5ea1bbf6fa157 : FAILURE in 1h 52m 59s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0f7aea1a0d084f418033c473be9f700b : TIMED_OUT in 3h 12m 35s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c39de746884a44d2afad0e60916a2bf2 : FAILURE in 3h 03m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96edeaae2a3496c8c7fe059f9bee0b1 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24baaab438a0440fbadbc96aeed70dce : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":13},{"id":"d689122d47db529e232af0db93f63e13de32f41c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-03 06:20:02.000000000","message":"Patch Set 14: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"8ece43766664250b7cb215bb3a3014e7792e0d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 09:10:45.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81649754f7374c8a88c2d99909749fb9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/822ae17945ab4fa19d9a205ba4a9a8b5 : SUCCESS in 9m 58s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/00b6fdb98a654492af5944185249f9a9 : FAILURE in 26m 38s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/78c111213ea84c079fbb10365ec7269b : FAILURE in 2h 42m 16s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4439e4f43feb45b5b522137b9b110705 : FAILURE in 2h 39m 26s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/685079a72aa74408993b719d167d3a1b : FAILURE in 2h 38m 41s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/dbbca52d8070445395c6799bdca11c25 : FAILURE in 2h 01m 01s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/010302ddc82a43c18ce71880113cb545 : FAILURE in 1h 38m 04s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a85fd13f0b704ebba005e60b6eee4079 : FAILURE in 1h 25m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ce885c7bfc485baaad2340cfccf0e3 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1a073400eb940bb97bb63096f62ac6b : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":14},{"id":"73696b0d418b48c187d90b08a4a4be8530658ab9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-03 15:13:43.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8bdd34d548c788b11f8f45917fc2b4928f822ab8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-03 15:23:26.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d90249739259d1023934b4d9f38d0fc076219f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 19:01:05.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed9af01b159e475388055f239bd0f53b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0b027ca2a08423197e2783475bc892d : SUCCESS in 13m 47s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/0ffe957e7a224ff1949eff74f6b1365f : FAILURE in 20m 57s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2b47ac3e40b34c83b9d4028f5dbd708f : SUCCESS in 2h 51m 59s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a6cc15cbd5814afda058cdebdc046016 : SUCCESS in 2h 34m 49s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0aed4aaadb7a4aeba7ef6ffe5760ca66 : SUCCESS in 3h 01m 14s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2399a9bb0f664f5c9af75f1791291b47 : FAILURE in 2h 27m 37s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f2d233ea0de64876a25dd91224d39da6 : TIMED_OUT in 3h 27m 54s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4849f6b73474701879c24d33b50c899 : SUCCESS in 3h 07m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f23419a0b14994bf7fdc6da9d797b4 : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63f930a67daf412eba39d4c56d432f89 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":14},{"id":"878f4c654207242c50d7bed903ef0348d4dade6a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-04 17:02:35.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2c9b64fc11dd752f4ec2c5c58c70c9445e85bab5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 23:37:28.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/afe695917d8747faa8ff4c3f7c841b5b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/424559c3512a491fa65394e8eaa26fa1 : SUCCESS in 10m 18s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/be0ee6ce84f54255b450c9f06a54e3a1 : FAILURE in 23m 33s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/4fb16e7d8ad74888b096a93b405bd39b : SUCCESS in 2h 44m 53s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0bdc7dc5fe7c40f4a180207a6827a702 : SUCCESS in 1h 40m 04s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08d1a1e47f674ae0b5f3bff0a34f2b14 : SUCCESS in 2h 45m 23s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/cb86982b2dfd43a6a28d133b89d6e10d : FAILURE in 2h 53m 57s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/38d0e5b6802744d5b661d5f409e575c3 : SUCCESS in 1h 42m 34s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d760ec9b7b247789003536ff9f5bbd3 : SUCCESS in 2h 51m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c71c2c502e486a9d41e2f13212436f : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e98455471f394b5a8100e0d179ad7832 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":14},{"id":"8d76af38bc43f4f5bb093ae4d46d57ba41aa05c2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-07 09:53:04.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e550e9b3c6eedadb31e2f3a436ec47a7f3a7bf11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-07 12:26:24.000000000","message":"Patch Set 15: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"9d6e734dc9e5ab49ab7a6a54e911e6607718c041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 16:33:39.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a467a15f269949d595c8a96cb8d75fd6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69d6c41363b14291a7d8bc0fb1782a34 : SUCCESS in 10m 47s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/cce0f96b6bde47ddb5e702b7ef88a5ac : TIMED_OUT in 3h 15m 04s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/768d919bb75a4a21b89f3a3306474bdd : SUCCESS in 2h 52m 15s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c7f96c9f38234f3bb1a990c9f834d839 : SUCCESS in 3h 03m 51s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1aaa41bf3dd0465db435a60a7606799c : SUCCESS in 2h 58m 49s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/29f4f8dca65e4280abe206d17210a502 : FAILURE in 2h 55m 04s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/adda42b6fd8a4bc59de1a70709b1c833 : SUCCESS in 3h 05m 49s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffcb675c9d9f4f45ab7dc5e322c54989 : TIMED_OUT in 3h 08m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd0671322bb64809ab4ccf7b98c94cab : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/179eba60c0bd4136a1e2973a10b5db35 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":15},{"id":"16de0905df722a73c7354355ef24c4309fded7e2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-13 07:12:36.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"836a89c851934029ed5e4e70e27b2658660efcc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 10:41:40.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a363d5c064c43f188068c38a53cea26\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9667c9028864438bbb0aacd89ce5af60 : SUCCESS in 13m 51s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/60a5edd1da804f618645b00dc08e0019 : TIMED_OUT in 3h 15m 55s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/49b20546f6e541728e8457ad9e98a22c : FAILURE in 2h 45m 04s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d179fd8a0d2d4504a83502df985669dc : FAILURE in 2h 58m 52s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e2d2412688d408fb20f82fd7cf85538 : FAILURE in 2h 39m 10s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/45dfef02af91464b83022b4d46ba1bf0 : FAILURE in 1h 32m 48s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3f5c266f82df4274b69f7742fde55ab2 : FAILURE in 1h 44m 01s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fecf2a8be634d85b92ea3847c6ccbbf : FAILURE in 2h 26m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09212a8a3c7489cbf7af5d84ac6d2a3 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ed92b54ab7142a997af590e53f25dd6 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":15},{"id":"2ef2dadb1278eaad7d1a30c1b7b39dfa7ba91597","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-14 15:20:57.000000000","message":"Patch Set 16: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"b26814bc765e1860471f669d6a2a7f7367555df1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 18:55:00.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57ae77278232415dae002c713d08c945\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e801fee011e49d783adf9996a275e86 : SUCCESS in 12m 56s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/fc1d961d53ae47a995c6bd8a9eee5d07 : TIMED_OUT in 3h 18m 50s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6c6d27640fb14357a9c6934f38c94402 : FAILURE in 2h 47m 26s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3c76111a7c854fc6a7d4d54ac881b547 : SUCCESS in 3h 04m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/138a90acf2e74914b3bd0827842e8173 : SUCCESS in 1h 31m 44s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/70e9e43aff6645ab8a37583185f3a126 : TIMED_OUT in 3h 09m 57s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f975c0257ce844e1a7a06320080fb1b9 : SUCCESS in 2h 22m 18s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c6a13d7a3b64dc4926b6a6b6778978d : SUCCESS in 3h 00m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b22b9a9029845a9ab2dc3890590922b : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e82cbce699b248dc8d4ae390d27154d9 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":16},{"id":"5c6b803ffaafac41bd1c65bff7e3b59f1539777e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-22 09:59:42.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"ec5e57ac3d1850520c00c54bc4cfea71a5ccfc7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 15:44:16.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/330c1ba62aa34d4b84581fb46287d7b6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8910eed43029480999966921351efb4e : SUCCESS in 11m 09s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/bfbb578c319449eba989d62f9388477b : TIMED_OUT in 3h 12m 56s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/4d701fcfbbfe4a26a99e8e0d852332d6 : SUCCESS in 2h 47m 57s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/422a2b2e73e14561ab1c4a088ad50f5e : SUCCESS in 1h 48m 00s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abb2c748b270489bbc0b4ffbe76fcdfe : SUCCESS in 3h 04m 04s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d5a90a8e8b8d46b29d9a670e1e28bf2a : FAILURE in 1h 57m 52s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/666e92a4c7ca4141a497a3f1ffab8c51 : TIMED_OUT Host unreachable in 5h 38m 15s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db907b88b66b43d586fbf833c447a67f : TIMED_OUT in 3h 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00291969f2e24283af4b284d54490607 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7144a4eb3bc4aaf8f01af95394446aa : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":16},{"id":"ff1ea35840796a6e0e781bd28289bb991cd71d22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-22 16:08:42.000000000","message":"Patch Set 17: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"aff422d01bbea77743e44624d7779d17cab91047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 19:42:16.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5550c1f94fb5488f83fbe3d0885b3d38\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7f0749fa34043a1a0bdbdfb1a6af0f7 : SUCCESS in 12m 55s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a78bd947e21f40669135218ca5269638 : TIMED_OUT in 3h 23m 58s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0928a9180ce5433c8c1e8d0acb022948 : SUCCESS in 3h 00m 10s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f4f29d64e7104d13834281a2ee71462b : TIMED_OUT in 3h 09m 03s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c89c725cdde4f428f22ea56545b2915 : SUCCESS in 3h 03m 19s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/904356f5c7204fd1a10694ab2fa6f2a9 : TIMED_OUT in 3h 12m 17s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d60420df94424471a8c1e5a4a2e7bb50 : SUCCESS in 2h 27m 28s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40777ddb9484478fa40ff4de992ff4c3 : TIMED_OUT in 3h 07m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fbf6890e54f4a72b6660f1dd8ce1727 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab3c4275d765491db9ac32ef973b4ef8 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":17},{"id":"a228c789d1f413538ea124c8dbf3025af81f727a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-24 07:06:56.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a535fbc2b8fa37f086b7d55120ff85a505545d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 10:33:45.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9a24f4ec9634914813760fafd44cbc6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08c055539c0a486a95b239baeb9576cc : SUCCESS in 8m 47s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/014667b70a7c495b992efc74db7531b0 : TIMED_OUT in 3h 13m 01s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c7120a2ba0d441048476fce47e40893f : SUCCESS in 1h 42m 59s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/324a74d28452407fbb55e1e53611cd84 : TIMED_OUT in 3h 06m 39s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/228347c56c7840289d2166002c0f301c : TIMED_OUT in 3h 08m 22s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/1b9a76c3b02444a5a6d23fd0181e74fe : SUCCESS in 2h 41m 56s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/989b09cc4b5c4a7789c7ddd25d2b178f : TIMED_OUT in 3h 06m 46s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/495dd01ce88c494483c7427f97873463 : SUCCESS in 1h 39m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ba1f0f609b4ae3a9e9a7380523a4b4 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59e22fe618104448b92e5d94041242ab : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":17},{"id":"82dce673f23dda9f35f54ea9d234b6b370b354ae","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 08:33:58.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"87e89e37c583ab3d0e86250bae1ce6d9efa7d833","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 08:34:44.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"f0fb3f4773132ab50c927e11b6d77d3e249bbbc5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-19 08:19:57.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"ffbbadbb522ced64db3f161ec4bd65bc341fb188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-19 08:45:01.000000000","message":"Patch Set 18: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"dfe73e240ec12621e6aae778eff39e1896c7c99e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 11:47:36.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdb803ad736e4d2b9f49528fd622eeb7\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d2d6bf176314aca9a146dc09eb87773 : SUCCESS in 10m 22s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/cf5ed3790f4649b5a452b706530db7f3 : SUCCESS in 2h 48m 34s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9f4b640103e745ffbb504d141fbcc361 : SUCCESS in 2h 40m 11s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/566f26f1f06543148b765a76c0fb343f : SUCCESS in 1h 29m 17s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/729d2c9fb5db45d88da86426b5cc0a7f : SUCCESS in 1h 23m 29s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f71624481e8f481ebecd15b3437b4625 : SUCCESS in 1h 33m 28s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7f8a157dfb0645f69564dc99c32c5d85 : SUCCESS in 2h 24m 54s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6c1d532a7eb421b9249a33b9082e3ac : SUCCESS in 2h 45m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9959b8847477417bbfbf9b2754e8e3d8 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d53290afbbe64b53afb17bd00caa5734 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":18},{"id":"1c5a476a9d03ce085eaeff5da2314cf1b0dfade4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-21 14:12:17.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"c588c4372ab8677af631f15e0a77fa870bab6cfd","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-08-27 11:15:17.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"29c83cd13b395e6779dd82a6536086d51a2ad1bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 11:18:51.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"c905415afdb3a66bcdf5f5becea59d4edb418481","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 14:33:21.000000000","message":"Patch Set 18: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc08639791534f1fa4d9ac80d903c067\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a54cd2efbfc74a4b87b2c21d69cc0148 : SUCCESS in 9m 29s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/cc24a625c9554375989dd4886cdac043 : SUCCESS in 2h 06m 14s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/5513d92808a3441baf0c7b5f021c58dd : SUCCESS in 1h 37m 38s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9511ff541d4a47bb9ce37ef95181d8e6 : SUCCESS in 2h 40m 48s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/148f4a4a306d474d914da60edb5b4cd2 : SUCCESS in 2h 26m 32s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/672507358e6449009fa49e35a7f81fca : SUCCESS in 1h 21m 19s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5d97f0b8c0644affb416d982f4d0e56f : TIMED_OUT in 3h 08m 40s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f2d429767c84afa8a09929f5c97f711 : SUCCESS in 2h 41m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2acba01614944d78b550379d971f26c6 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c216c6699e54d7ba9ee920da391eb62 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":18},{"id":"550828a6900bb0fbf5d2e75420f32ee868d88826","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-27 16:05:42.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"5ba04eb953fb72d92411cf762a9f585e6db9fedb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 19:16:03.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bb9cb62b3de4d4782577f1494f884be\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26ea5102a79149368dbdc6aa32b01bdf : SUCCESS in 10m 21s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/35f5ca989c064b18b0b1bd6a6756a31b : SUCCESS in 2h 38m 19s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8216175563014a989329e59c469bfdc0 : SUCCESS in 1h 34m 13s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/302d8a6b082f422b9eafe06ca402198e : SUCCESS in 2h 27m 28s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e65e282772240d49d556974d266ea5a : SUCCESS in 3h 04m 23s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2af90eab9b76494c8ed628261b62cfab : SUCCESS in 1h 24m 47s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/53182f16ae7649d4b325eaac38e5e86b : SUCCESS in 2h 47m 57s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d42f7735b32a4b449817beab9736bb38 : SUCCESS in 2h 41m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79b6c9a71e2b4ca78c1d2e0d46c9368e : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcfde46a42f442e59ef9a1b8061a61f5 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":18},{"id":"6ee9a1df2aa4c41a2ac9eb4a50f89d7312624540","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 19:18:22.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"58a8a926a84e2d3a2e9434fdd9e81c85508a8c3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 22:23:56.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44b37c69fc6c4275b903fc48222b0a57\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a9f7630db4f48f6b290a2c951f4a1bd : SUCCESS in 9m 41s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e58c5ed964bf4383a377b6901905404e : SUCCESS in 1h 25m 56s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/97b86230ba274a7d822caf75542f9cae : SUCCESS in 2h 44m 50s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/60b450e1074c4e8d92c7284d4ab13368 : SUCCESS in 2h 09m 40s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90764c2f08b84b51aed21eab288ba9a1 : SUCCESS in 2h 23m 12s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a9d720274f9f49449122dee7c72454ea : SUCCESS in 2h 01m 04s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/144c03afd67d4ae9b0b37db1427010f1 : SUCCESS in 1h 34m 11s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b855bfbced642ccafb86658745e92c0 : SUCCESS in 2h 58m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46166d271db444268b4ea1e41ffd94d1 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17f206b24ac74b9b993b45b6d5000413 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":18},{"id":"0536920bbc61c9b1addbb106464f2d632bbb9262","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 22:23:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"5d3a340c37e70940b8cc8ec3e293bfb1811eabba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 22:24:44.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c77fc2aeacb940c6b07682e0d079084a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488213a91a7349bc9124c7566c407d02 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"cb1eb1f8a1f6cc01ecbe1bdc495d4e616d8f22c4","revisions":{"78511dc37b26be96a791b0e496e86a0054715ca6":{"kind":"REWORK","_number":1,"created":"2026-04-09 14:00:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/1"}}},"commit":{"parents":[{"commit":"d9cb143e28d56aae6bc780b9e7cb2537201765e4","subject":"Add internal TLS coverage for API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d9cb143e28d56aae6bc780b9e7cb2537201765e4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/982786\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982872\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/78511dc37b26be96a791b0e496e86a0054715ca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/78511dc37b26be96a791b0e496e86a0054715ca6"}]},"branch":"refs/heads/master"},"ffcc73d3c7f013b621516e5ceebc380f46dc7cea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-10 15:06:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/2"}}},"commit":{"parents":[{"commit":"4ba088b252bce64b335dc994646c0a912379cade","subject":"Add internal TLS coverage for API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/4ba088b252bce64b335dc994646c0a912379cade"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-10 15:06:30.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/982786\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982872\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ffcc73d3c7f013b621516e5ceebc380f46dc7cea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ffcc73d3c7f013b621516e5ceebc380f46dc7cea"}]},"branch":"refs/heads/master","description":"Rebase"},"81f375bd6fade46dc3b7f1811af2b6ccb30f752c":{"kind":"NO_CHANGE","_number":3,"created":"2026-04-10 15:08:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/3"}}},"commit":{"parents":[{"commit":"20635dad1f46795a59fdb74df59df2d1725cfb24","subject":"Add internal TLS coverage for API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/20635dad1f46795a59fdb74df59df2d1725cfb24"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-10 15:08:20.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/982786\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982872\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/81f375bd6fade46dc3b7f1811af2b6ccb30f752c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/81f375bd6fade46dc3b7f1811af2b6ccb30f752c"}]},"branch":"refs/heads/master","description":"Rebase"},"ad09aede9635dacf35fce46e4f66a2e2ac6e6cc0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-04-10 16:08:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/4"}}},"commit":{"parents":[{"commit":"480544df580aec68544d945630e52128033902a8","subject":"Add internal TLS coverage for API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/480544df580aec68544d945630e52128033902a8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-10 16:08:42.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/982786\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982872\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ad09aede9635dacf35fce46e4f66a2e2ac6e6cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ad09aede9635dacf35fce46e4f66a2e2ac6e6cc0"}]},"branch":"refs/heads/master","description":"Rebase"},"cab3947ef95e23316aea0cc3ab03b20f06859aa4":{"kind":"REWORK","_number":5,"created":"2026-04-10 17:06:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/5"}}},"commit":{"parents":[{"commit":"480544df580aec68544d945630e52128033902a8","subject":"Add internal TLS coverage for API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/480544df580aec68544d945630e52128033902a8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-10 17:06:41.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/982786\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982872\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/cab3947ef95e23316aea0cc3ab03b20f06859aa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/cab3947ef95e23316aea0cc3ab03b20f06859aa4"}]},"branch":"refs/heads/master"},"702f8d358b08cc8b8335d67b0429260138ac076b":{"kind":"NO_CHANGE","_number":6,"created":"2026-04-18 08:00:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/70/983870/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/6"}}},"commit":{"parents":[{"commit":"e0bd2373eeb01e0427a55166d61dc0e76c3a3bca","subject":"Add internal TLS coverage for API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e0bd2373eeb01e0427a55166d61dc0e76c3a3bca"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-18 08:00:33.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/982786\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982872\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/702f8d358b08cc8b8335d67b0429260138ac076b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/702f8d358b08cc8b8335d67b0429260138ac076b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"cab3947ef95e23316aea0cc3ab03b20f06859aa4","theirs":"e0bd2373eeb01e0427a55166d61dc0e76c3a3bca","contains_conflicts":false}},"d750c4db252baf07b5a4b5fa742a5790a846076f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-06-13 00:00:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/983870/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/7"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-13 00:00:45.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/982786\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982872\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d750c4db252baf07b5a4b5fa742a5790a846076f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d750c4db252baf07b5a4b5fa742a5790a846076f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e0bd2373eeb01e0427a55166d61dc0e76c3a3bca","ours":"702f8d358b08cc8b8335d67b0429260138ac076b","theirs":"e154108444b41516c1f5e23cdbff463012fd6acb","merge_strategy":"recursive","contains_conflicts":false}},"e75816729b9f345a6a6749e669f09f7994af341b":{"kind":"REWORK","_number":8,"created":"2026-06-29 08:52:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/8"}}},"commit":{"parents":[{"commit":"1c5db7373574beb37c34338d75eb7e66f4b6549c","subject":"Remove v1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/1c5db7373574beb37c34338d75eb7e66f4b6549c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-29 08:51:15.000000000","tz":120},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e75816729b9f345a6a6749e669f09f7994af341b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e75816729b9f345a6a6749e669f09f7994af341b"}]},"branch":"refs/heads/master"},"9699719417abaaf27e323bd8a193bf74017be3e6":{"kind":"REWORK","_number":9,"created":"2026-07-01 10:50:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/9"}}},"commit":{"parents":[{"commit":"1c5db7373574beb37c34338d75eb7e66f4b6549c","subject":"Remove v1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/1c5db7373574beb37c34338d75eb7e66f4b6549c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-01 10:50:12.000000000","tz":120},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/9699719417abaaf27e323bd8a193bf74017be3e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/9699719417abaaf27e323bd8a193bf74017be3e6"}]},"branch":"refs/heads/master"},"bbab61b1571f4f8e93c9641d2365a20683af6ab1":{"kind":"NO_CHANGE","_number":10,"created":"2026-07-01 11:59:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/10"}}},"commit":{"parents":[{"commit":"6501d759104e24e9e5852876f07cec5892313d19","subject":"Remove v1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/6501d759104e24e9e5852876f07cec5892313d19"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-01 11:59:32.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/bbab61b1571f4f8e93c9641d2365a20683af6ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/bbab61b1571f4f8e93c9641d2365a20683af6ab1"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"1c5db7373574beb37c34338d75eb7e66f4b6549c","ours":"9699719417abaaf27e323bd8a193bf74017be3e6","theirs":"6501d759104e24e9e5852876f07cec5892313d19","merge_strategy":"recursive","contains_conflicts":false}},"2da8165411412d8deb0f56efb60ea711d878a15c":{"kind":"NO_CHANGE","_number":11,"created":"2026-07-01 14:47:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/11"}}},"commit":{"parents":[{"commit":"e7ba90a65a288e19da6e10fceec332e68116f92a","subject":"Remove v1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e7ba90a65a288e19da6e10fceec332e68116f92a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-01 14:47:27.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/2da8165411412d8deb0f56efb60ea711d878a15c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/2da8165411412d8deb0f56efb60ea711d878a15c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"6501d759104e24e9e5852876f07cec5892313d19","ours":"bbab61b1571f4f8e93c9641d2365a20683af6ab1","theirs":"e7ba90a65a288e19da6e10fceec332e68116f92a","merge_strategy":"recursive","contains_conflicts":false}},"612c2b9d2938cdcdeaaa6ab51ad0e7dc05449b40":{"kind":"REWORK","_number":12,"created":"2026-07-01 15:00:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/12"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-01 15:00:08.000000000","tz":120},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nDepends-On: https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995696\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/612c2b9d2938cdcdeaaa6ab51ad0e7dc05449b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/612c2b9d2938cdcdeaaa6ab51ad0e7dc05449b40"}]},"branch":"refs/heads/master"},"f3df8e5b890e56d88a9489430e9ca8f617fd954e":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2026-07-02 18:52:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/13"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-02 18:52:50.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nDepends-On: https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995668\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/f3df8e5b890e56d88a9489430e9ca8f617fd954e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/f3df8e5b890e56d88a9489430e9ca8f617fd954e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"04a3371bb60a130f5ee34289a2a6db934e688033":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2026-07-03 06:20:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/14"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-03 06:20:02.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nDepends-On: https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995696\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/04a3371bb60a130f5ee34289a2a6db934e688033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/04a3371bb60a130f5ee34289a2a6db934e688033"}]},"branch":"refs/heads/master","description":"Edit commit message"},"70ce3ce2e29538050c6c6573c87fd5e88b6bee14":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2026-07-07 12:26:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/15"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-07 12:26:24.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nDepends-On: https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995696\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/996268\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/70ce3ce2e29538050c6c6573c87fd5e88b6bee14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/70ce3ce2e29538050c6c6573c87fd5e88b6bee14"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f2cd1e0e73f783d7ba7e3abfe53ce34ea2812370":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2026-07-14 15:20:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/16"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-14 15:20:57.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/997224\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/f2cd1e0e73f783d7ba7e3abfe53ce34ea2812370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/f2cd1e0e73f783d7ba7e3abfe53ce34ea2812370"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3e8a62598a350ad73cc751634e3f55923ff396d8":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2026-07-22 16:08:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/17"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-22 16:08:42.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/997224\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/998330\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/3e8a62598a350ad73cc751634e3f55923ff396d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/3e8a62598a350ad73cc751634e3f55923ff396d8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cb1eb1f8a1f6cc01ecbe1bdc495d4e616d8f22c4":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2026-08-19 08:45:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/983870/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/70/983870/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/70/983870/18"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:59:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-19 08:45:01.000000000","tz":0},"subject":"Add support for centralized scheduler deployment","message":"Add support for centralized scheduler deployment\n\nThis adds basic support for deploying a freezer scheduler in central\nmode, which is going to be introduced in 2026.2 release\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1001467\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/998330\nChange-Id: I5e0fb5c7377f9f22c52df28c4a11793311efba83\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/cb1eb1f8a1f6cc01ecbe1bdc495d4e616d8f22c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/cb1eb1f8a1f6cc01ecbe1bdc495d4e616d8f22c4"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
