)]}'
{"id":"openstack%2Fopenstack-ansible-os_freezer~984473","triplet_id":"openstack%2Fopenstack-ansible-os_freezer~master~I656ca85ed19178227f195435ebcf2c344d5b3cda","project":"openstack/openstack-ansible-os_freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-20 13:46:40.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-20 13:46:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I656ca85ed19178227f195435ebcf2c344d5b3cda","subject":"Release notes - fix for formatting","status":"MERGED","created":"2026-04-13 21:38:47.000000000","updated":"2026-04-20 13:51:12.000000000","submitted":"2026-04-20 13:46:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984473","meta_rev_id":"2bddd3662478ce257f51657a98a591b863de92fa","_number":984473,"virtual_id_number":984473,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-20 13:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-04-20 10:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-18 08:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-04-20 10:43:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 21:58:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-16 09:01:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-18 08:03:44.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-04-20 10:43:53.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"21457f88907e2a1e4ff2f5e7eb60d8f5f7553ed5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-13 21:38:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ec44a579fb2fb0920d0550993df220886c15c1f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-13 21:39:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d6ae84ca11c774b546cb26429805d1710e844153","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-13 21:40:10.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5787e38e8e8897d4e5f9df37b45d6d5584614fbb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-13 21:40:12.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"8d4880ce269569a64e69b37566149aca10f128b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 21:58:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f38a4b19433e475da1380b417e5fa8d5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4910c37fd2134b2595e613845e02de0c : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d16720285d043258331f49513bdadc3 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a06e6576926142219e44e48425f614ea : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":3},{"id":"4b4d5aa83a4e124ea5065e98b923bf620e84a54b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-15 15:29:25.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_37598\u003e\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":4},{"id":"4755e745e078965e7f2e134f9f47e573b7766132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 15:59:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca20ddde1cd64f22928138f3543741f3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/714f9eee5b154c4ea799fc1e7d26df6f : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0963e8399e454881986a9d5bb1efe64c : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b22a36c4586d45eb8176df1438b5aa95 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":4},{"id":"efc6c214b52b698491450fbcbf1609569e8f9b2a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-16 09:01:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6771bf024245c80540953b7477908bbfab8263b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-18 08:03:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ed5cf3f0ada98ebc62e551d81067537d7e678517","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-04-20 10:43:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"48649be3291a3f6d8b968d74ad48834b72d27b99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:49:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b6938c27bd1bc6b694c3ccf35d61e4c1ceca8776","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 13:46:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/610e4a04c74a46d2a5171996870e2198\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/961c0f1f09f547bd86d7974dec247a28 : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd159a5413504e2793e42dfbcbd6ce98 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c85d69ec225c407dad294a1b8cd8e5d9 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":4},{"id":"53cdbc7f2eaf468fc51166526f1285d6e4747ca1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 13:46:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2bddd3662478ce257f51657a98a591b863de92fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 13:51:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce32f3e0fe3e4b159170fe293cbd6fd8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84610188b6484e8d87a7c45ee6fd1739 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d2d9c55e867df8363b874084a78b74022b327979","revisions":{"1801db8b581fcc9c29aafc73101ae99307f1d2ca":{"kind":"REWORK","_number":1,"created":"2026-04-13 21:38:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/984473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/73/984473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/1"}}},"commit":{"parents":[{"commit":"ca177442ed32ee8019c075999b2982332bcfbfe2","subject":"Cleanup the retired repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/ca177442ed32ee8019c075999b2982332bcfbfe2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 21:38:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 21:38:47.000000000","tz":0},"subject":"Release notes - fix for formatting","message":"Release notes - fix for formatting\n\nChange-Id: I656ca85ed19178227f195435ebcf2c344d5b3cda\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/1801db8b581fcc9c29aafc73101ae99307f1d2ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/1801db8b581fcc9c29aafc73101ae99307f1d2ca"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"c31e84db3ad87c328f8de790e5e32e1d5d61de49":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-13 21:39:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/984473/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/73/984473/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/2"}}},"commit":{"parents":[{"commit":"d54b2b3b0125e7c7c27b3bf3e23a3cec58976a44","subject":"Place content for os_freezer role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d54b2b3b0125e7c7c27b3bf3e23a3cec58976a44"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 21:38:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 21:39:36.000000000","tz":0},"subject":"Release notes - fix for formatting","message":"Release notes - fix for formatting\n\nChange-Id: I656ca85ed19178227f195435ebcf2c344d5b3cda\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/c31e84db3ad87c328f8de790e5e32e1d5d61de49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/c31e84db3ad87c328f8de790e5e32e1d5d61de49"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"1801db8b581fcc9c29aafc73101ae99307f1d2ca","theirs":"d54b2b3b0125e7c7c27b3bf3e23a3cec58976a44","contains_conflicts":false}},"cc5b01138f499afdc594a4784781297096593ca9":{"kind":"REWORK","_number":3,"created":"2026-04-13 21:40:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/984473/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/73/984473/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/3"}}},"commit":{"parents":[{"commit":"d54b2b3b0125e7c7c27b3bf3e23a3cec58976a44","subject":"Place content for os_freezer role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d54b2b3b0125e7c7c27b3bf3e23a3cec58976a44"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 21:38:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 21:40:09.000000000","tz":0},"subject":"Release notes - fix for formatting","message":"Release notes - fix for formatting\n\nChange-Id: I656ca85ed19178227f195435ebcf2c344d5b3cda\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/cc5b01138f499afdc594a4784781297096593ca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/cc5b01138f499afdc594a4784781297096593ca9"}]},"branch":"refs/heads/master"},"d2d9c55e867df8363b874084a78b74022b327979":{"kind":"NO_CHANGE","_number":4,"created":"2026-04-15 15:29:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/73/984473/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/73/984473/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/73/984473/4"}}},"commit":{"parents":[{"commit":"7b89b1f162935d1d918f4b1d8b8f720130a4c519","subject":"Place content for os_freezer role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/7b89b1f162935d1d918f4b1d8b8f720130a4c519"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 21:38:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-15 15:29:25.000000000","tz":0},"subject":"Release notes - fix for formatting","message":"Release notes - fix for formatting\n\nChange-Id: I656ca85ed19178227f195435ebcf2c344d5b3cda\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d2d9c55e867df8363b874084a78b74022b327979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d2d9c55e867df8363b874084a78b74022b327979"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"cc5b01138f499afdc594a4784781297096593ca9","theirs":"7b89b1f162935d1d918f4b1d8b8f720130a4c519","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
