)]}'
{"id":"openstack%2Fopenstack-ansible-os_freezer~991716","triplet_id":"openstack%2Fopenstack-ansible-os_freezer~master~Id950ff110925b5fda2cac88539678a7c5047dc88","project":"openstack/openstack-ansible-os_freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-09-16 13:07:31.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-09-16 09:53:24.000000000","reason":"\u003cGERRIT_ACCOUNT_32666\u003e replied on the change","reason_account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-16 13:07:31.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-16 13:07:31.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-09-16 13:07:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id950ff110925b5fda2cac88539678a7c5047dc88","subject":"Remove v1 API","status":"MERGED","created":"2026-06-04 18:37:27.000000000","updated":"2026-09-16 13:08:28.000000000","submitted":"2026-09-16 13:07:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991716","meta_rev_id":"8a96f750742974d69f5c506aa6be512abc2b577c","_number":991716,"virtual_id_number":991716,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 13:07:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-09-16 09:53:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2026-08-31 16:29:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2026-09-16 09:53:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 21:49:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-05 11:59:44.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-06-13 11:46:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-06-15 10:00:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-06-16 15:51:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-06-16 15:51:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"bf7cedb0b354ebae5f7c5744a5241ac142d3e68f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-04 18:37:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfebd72b07ecd60285c018d97a4190eb42c1c763","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-04 18:38:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"34193380fc5473f409254432cb7f68caa92f175a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-04 18:38:29.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"ebd8f6ac23c038a041d2b34139e5504abcc58604","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-04 18:40:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4be5d99de480e9e9117b41dd4f7701eecd7ce689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 21:49:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2133321c08bd48419b6092e916a9a2da\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81696870f8744f1fa3cb59ec55585abf : SUCCESS in 10m 29s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/82dd271a350d4aadb9065c2281f0edc3 : SUCCESS in 2h 54m 32s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/104d3ccf8a1940198c413d0760cc16d6 : SUCCESS in 2h 45m 26s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/4af938b6edab4777bb3130c2721834ba : SUCCESS in 2h 55m 14s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/abe816ab1ea540fea79705d1eec5573a : SUCCESS in 1h 28m 52s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9670929f23e4edb9194a5b9d7f0dc2e : SUCCESS in 3h 00m 57s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ac517dc3be1149eb92290a0c13166941 : FAILURE in 2h 28m 17s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a7570ba08a8c42caacc693844c22ccf7 : SUCCESS in 1h 31m 07s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddfb8bdf93a14598aced32ccbe2da581 : SUCCESS in 2h 51m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a126707f46f74f1c88df9e5f652d5c66 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff04f697d7ce42ad9e2f81e5ae1a3c67 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2032ab5a35e3ef793278da4e8fe626b6fff3484b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-05 11:59:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7279113fe630bfde588aeab8e1a1f5d9087fa8ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 15:24:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2515b280f755499db8a2d77b1a1347df\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58751c776b8744149db56e2e10d5e358 : SUCCESS in 13m 26s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/36f81be5ff8147ae843569fbbaafcc31 : SUCCESS in 2h 23m 24s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7234906276354b438c04f48db8005386 : SUCCESS in 1h 37m 44s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/296e0cc8bb8b493881bfb5b3a52b0917 : SUCCESS in 3h 01m 52s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/503bebfedccf4c30a2ba83850bd919c6 : SUCCESS in 1h 33m 16s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cf14436153143ad8c2d2116f8b50561 : SUCCESS in 2h 44m 59s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/708df2d9d5aa40eeb30c0bc43e5fcd63 : FAILURE in 2h 21m 47s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6fb684c9011b4ebe895cefaa8085ed04 : SUCCESS in 2h 42m 11s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fed34d467954638a0ea494ff6b714a3 : SUCCESS in 3h 06m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b3c6bf95544b5090949ae1002e76a4 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16ac726fab2d4b3397e00310c96ae860 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"5147015d7aed69e8a127eedbf03542b6b1cdf5d3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-05 15:25:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9bafd59b0a5c164c54ffefd53b779cccebd0a6c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 18:56:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dd69e230f6a4157a8bd53100047bb87\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/479b6e2bc5594f5f80cd58809e0d9c0c : SUCCESS in 15m 44s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/8309c2a66d51400594c90ba85f228523 : SUCCESS in 1h 43m 01s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0b9639b3f1284967a1191a0a987c5d4f : SUCCESS in 2h 07m 48s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/859cc29e5a9a41fca161b584a4c8d7f3 : SUCCESS in 2h 45m 13s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/26b8f8b1ee834fe193964b56ce1196d9 : SUCCESS in 2h 33m 24s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb2be3b85b654aceb8805b8291028f32 : SUCCESS in 2h 48m 45s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/354c2adf0573452eaaf813ede56c2983 : FAILURE in 3h 06m 24s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/99e28588670d4dffb6aace60cf4b5026 : SUCCESS in 1h 39m 14s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/472fb1df27a744b5a9a6ed81b9f5eb38 : SUCCESS in 3h 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df080a8f3c774ca9b4adfa1cd3319655 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a33daad2d5284dbb8e724da91e564984 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"cc450daca148b98b3523bb259112d09fa7b8fc4a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-12 23:59:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c083b784cd63553c4a423fdd38b234a04ecb14ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 02:58:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6fdab1add5c4534b5e83ae63cbf3324\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca671f9e0b224a5aa59d24a8464cb7e8 : SUCCESS in 7m 59s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ba36cd01a27c4885bc80db8a4526fd9e : SUCCESS in 1h 59m 11s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1dcee44ef25748fbbde67dba4fc77389 : SUCCESS in 2h 37m 38s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f951ff5f19774ae8b7e38895b88c47b7 : SUCCESS in 2h 12m 01s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a16b5186d60f4f20ac030dfedcef85a8 : SUCCESS in 2h 53m 56s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92ac11e75d884db8a825b307c53655d9 : SUCCESS in 2h 33m 33s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a7e7e4fc2e81486abc91b63d9b6c1e4a : SUCCESS in 1h 21m 59s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4139a4c80a054099874efb6a5d0b16bc : SUCCESS in 1h 36m 16s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eab5c432c542424e9f8ad5254b139206 : SUCCESS in 2h 52m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f73a10dd63e045c5bc3b9445413ca586 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c141101dfa74528bf62cf5a155af26a : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b416895f7588d1b377f58aab0ec1354d932d51c1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-06-15 10:00:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2d612c8e21d9840fd46b325375d0537d7dca7cb6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-25 06:53:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0da88e11ea58a6368dfd3f803b9d794aa5c52e49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 06:56:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"99266b37925768b409ef7bd338da3f5c09cc3ea7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 10:11:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a85a6f3b32e94be3a42ab1b7bde5447c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7aedcd47b2f74ae9a1a6b29e9b0fb909 : SUCCESS in 7m 59s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/475f6ff25ff548c88e60489daa187ecc : SUCCESS in 1h 24m 49s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/114548f84386455391cd70a4e3f1b896 : SUCCESS in 2h 46m 24s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/977d687bc86b4ae58bfca913f4c97f77 : SUCCESS in 2h 21m 01s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18bf555362e14e50a59d07de178f8f06 : TIMED_OUT in 3h 07m 57s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/28d4835b7fb44964b50dd88bbcc044cf : FAILURE in 2h 40m 31s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d63d62f45d5480c9875422428476602 : TIMED_OUT in 3h 09m 37s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0c2be7c4f894316bc1a16848cc80b0e : SUCCESS in 2h 57m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c503c9701aa4d0c98c4d3e6f4e18939 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bab093be29f34ea88c8e47947142b0b6 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a1104ddef5925db3336fc5274f8a777c1293ba6d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-25 13:17:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc2d0b21ab423a440bfcd02df993063b4d5951bf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-25 14:31:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e77be450f69c96ee8d775ae95650381537a06392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 16:45:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56bcc799e9d74c7bb8a71be70e4e7518\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b620daf79d44d549434d2e3a275778c : SUCCESS in 12m 20s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3c81de7be9864ac493ed635d3c8435ed : SUCCESS in 2h 17m 25s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c2e2c37f0b0b474abf1d8e1c315012cb : SUCCESS in 2h 53m 17s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3aedf074b3824a0a9fdfa6160e44d659 : SUCCESS in 3h 05m 59s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70bf6026cc5843e092064bf5b69dc4b1 : SUCCESS in 2h 46m 58s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/58fde603292d40a5b1cc5e9a0b662308 : FAILURE in 2h 44m 33s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5ea168f1827d42b88583c58d5352fa62 : SUCCESS in 1h 38m 43s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33f066db937c48569b4bbe45fa83c338 : SUCCESS in 3h 15m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d522859f164cccb7bd29a2eaad1062 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c68fd8905b6d498cbd00879c49cf4e26 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"3f72bd0ea6812947a5c2ca8a7b5a4a589c62944d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-29 08:39:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91fdb4f77cf2aeb6f6190ca43d7da76a89c00520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 14:18:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4938f88d432a4eb88e5a9a12ca48bf33\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/825a3428728f40629c55a34c134eed80 : SUCCESS in 9m 12s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/7695edbd0509492093a3260c0d457da1 : FAILURE in 2h 14m 46s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/49f982bfa70941b99539ad723970e30a : FAILURE in 2h 49m 16s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/460416b773974190acc8ca97d48a218c : FAILURE in 1h 35m 40s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e427961bed74d13b151eed517aad0f9 : FAILURE in 2h 43m 23s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/078183a9c9f745a6894af89d017ce161 : FAILURE in 1h 26m 18s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7c345f266db04766a45aba6d7bf127fd : TIMED_OUT in 5h 27m 15s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4905181538943e1b7c7a108519cfb48 : FAILURE in 2h 48m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d891461956d4431a976348b445452d3d : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74b5310f567f4f81a1eceb42716fcfb0 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"39059fdd6e550d1ff2f6c0e17d4b59fdc895568f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 11:57:35.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"6f73174ab8bc094b9d8846d29db6e35bad450baf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 14:47:22.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"1930b04a9c36c2c2d1bc7abf3c848b2b9dda0108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 15:01:15.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"845267df6df626f328dfa980f3ab578de0572690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 18:23:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfa5e219e3dc447488db2e74f981fda1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b1071a0f2414d8096050e9b047e4102 : SUCCESS in 13m 22s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f718cc86ab074ee2ad641e99cb6c07db : FAILURE in 1h 43m 45s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2878ed29cd2a489fbdd116f031b90e3d : FAILURE in 1h 52m 09s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/120c5c4a451749518d352dbaa2e0de52 : FAILURE in 1h 43m 09s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb77fe902a8e428d93dd8072208c263f : FAILURE in 1h 44m 39s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/00e2855d79c14da7bdcae8ec21fd44c1 : FAILURE in 1h 33m 29s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/17df430938694863b50fde395224606d : FAILURE in 1h 47m 34s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc5d9428be9c4e889fa002b030a51327 : TIMED_OUT in 3h 08m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c944becc9b4341acc6b75c047f4006 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79683ff59d924d019b365e282738d166 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":5},{"id":"ea8bc26977e82f52af82fa9f16a4e7cd309d8f9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-02 08:37:39.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"01de84e492e24196a443f9c03db445238e206e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 09:13:56.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/openstack-ansible-os_freezer/+/983870, https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995696 are needed.","accounts_in_message":[],"_revision_number":6},{"id":"c75c4397204485936de863fa0e70fde1b3a44495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-14 15:22:11.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"cdf16ccba7bba4a6fbe0efe2031d1a06f3c56d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 18:50:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a78cfd7fd0c4433791edf4ec5bfc447d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9b4a2e4d2c64031b75823c813dceda4 : SUCCESS in 12m 01s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b9cd96ec41014078b8fb396b83e6a3ec : TIMED_OUT in 3h 23m 31s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1a872c0525f540c28a561bbe02c01e7a : SUCCESS in 2h 59m 30s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/142962b7f102486abbe5f73e954b1643 : SUCCESS in 2h 11m 40s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf2e16d0c3224391b6d28fdc91abcf7d : TIMED_OUT in 3h 10m 37s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/cb6bd04642df4a2485473e9a15404c54 : SUCCESS in 1h 27m 38s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9efef442140b4022b120d831780b5800 : SUCCESS in 3h 00m 34s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc9a8e07ef5c481c95cf4fcd51e2d3be : SUCCESS in 2h 56m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1a2a6b0792640dd927c62978e3bd3a2 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fa9aceaf91c43feb87e69fbaaa01ca2 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":7},{"id":"b9174ec2be1211b688036292ad654e88e3623631","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-24 15:24:40.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"b87b9e7defbc43d717afd344b827007562dae54a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 18:53:56.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0850d9fa1331402ba9bd89001e0bdb38\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99f7aeda2fc148fb99cfef1f99e88b56 : SUCCESS in 12m 39s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e886dcccd4504c6f8bb907f45438610c : TIMED_OUT in 3h 19m 42s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e43d4183fcb145eca7c60c2cbcfa1473 : SUCCESS in 2h 54m 01s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/23f9783abce14ce39b13ee26e4746da2 : FAILURE in 57m 51s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfd9322e5f664aeea9debb3d3feb1bd4 : SUCCESS in 2h 44m 59s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/0ff72427b25a4e20a53d894a60304630 : FAILURE in 49m 19s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3bdd4d8993f548d6849037e398589922 : TIMED_OUT in 3h 07m 48s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d24249488ea4f9e8ee015062f369813 : SUCCESS in 3h 09m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa5b2dded86496687ff183378071520 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1362f9fee7d41ce962bcc526c793ba8 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":8},{"id":"1f8ba17ff6d56e0275988a05ff8cbd07d9de0b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-20 11:49:21.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"5c5b302b94d055e2f7958a744fcd5476c775edee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 14:00:24.000000000","message":"Patch Set 9: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible/+/995111 is needed.","accounts_in_message":[],"_revision_number":9},{"id":"2e50c030e83b5491f757ce82fa586ff924a76a6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 16:19:24.000000000","message":"Patch Set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"626a4d601fdde314b25a6ec0ccb44132f9c58d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:23:23.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14bdcd287a7e4418af75ce45dcec2d66\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58dcbd6a8b28488898a52514c0536426 : SUCCESS in 10m 43s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/70c0533e66374376ab3e457ad4da5031 : SUCCESS in 2h 16m 49s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8552aee022a64dd28295ae4a7ad5ccc6 : SUCCESS in 2h 34m 10s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cb0ccd92954046b189bb6857ac134282 : SUCCESS in 2h 35m 22s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47dd522724f74e2fb1c18c4c3a8641c2 : SUCCESS in 2h 52m 46s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d034b0851a6242ae8e7a6fb929d5818d : SUCCESS in 2h 35m 34s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7f30843da7794ec2b36f8624257ee3c6 : SUCCESS in 1h 35m 29s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3080863b849d4594a2671b2a5c28f984 : SUCCESS in 2h 36m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/966aa26eea7545e989ad2f60b557a2b5 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5eb84be2eef1474394100221047d3b8a : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":10},{"id":"0374d9a2d6e78e9d24026a7fd52effab017478c9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-08-27 13:01:13.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"83d34547404c612fab6c96fea5aafe3493d606e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-31 16:29:03.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"0d8749b8da2db47d10021bc376bd4939aa9b6745","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-31 16:29:26.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"1058569b7e311a7652a1f2a076593e83efc6f9e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 20:39:32.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ce5cf0141314f32b1c21a5c35b70a62\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/058f3206fb904ce08d90b270f9db0fbb : SUCCESS in 10m 18s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/73d2ffbb798b4ca2afe1edf650ad9c4e : SUCCESS in 3h 01m 22s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/94aaace2965b44d4b32d8d2e1d1bba3f : SUCCESS in 1h 57m 15s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ece91d9a74084839a4516c9e827c935e : SUCCESS in 3h 13m 57s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e5c182a44944202b91672c15b1291ef : SUCCESS in 2h 41m 55s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f2590f457aed4a79a497e1bf619d4c77 : FAILURE in 2h 48m 39s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/43865088662046fbbe47c361e5aeb5df : TIMED_OUT in 3h 14m 26s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1b22dfc362040acb98fadc6373e556d : SUCCESS in 3h 03m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2886d80c71964a78be9e06011f657a62 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c6a38efa8c344debe1fab08e8447922 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":11},{"id":"a8005c662d08591c1021737fbad87d41721938e5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 23:14:24.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"91fb2d2eaaf43dfd4f20f11b02b58554c3dff081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 02:30:08.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce4033b4723d4992a537387dd65b69b2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f86c44ec0abd46d48ff752d351b8af18 : SUCCESS in 11m 45s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b85d5813f30149648d885f0baf16658c : SUCCESS in 1h 36m 19s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b2be545374894c78a06009481d2031d7 : SUCCESS in 2h 43m 32s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a235eb94b50e4dab87b3825a4c37f5ba : SUCCESS in 2h 21m 18s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03b145439d724605a634d566229a9b52 : FAILURE in 2h 35m 36s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b0f78d3b8503409d9b9ec8a6d9cede3e : SUCCESS in 2h 43m 24s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9ef0edb720e24e99b1bfca74d0cf6498 : SUCCESS in 1h 35m 59s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7204977d2e149b8a272a42eb2e21355 : SUCCESS in 3h 09m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e35ed2feeb2341d592ae41de6424fdba : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a17eda5c8823409d81e2a8a606053af7 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":11},{"id":"dd59c35fefd12da2c014d2b4dfedde3e187ea4c6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-01 12:09:05.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"b50cd4bb61befef7cedb9f63a127df1cb2311a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 15:19:43.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f360d4d066c149ec968e41825c0b208e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab3de5e3188749b8aaafd9154dcc5071 : SUCCESS in 11m 17s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/aea14e05922e4a4387394a895cf00c88 : SUCCESS in 1h 33m 19s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2caa4258600045588687af09b1fe2f2e : SUCCESS in 2h 56m 53s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/78e5a5c089a944e1b801d6546e29901e : SUCCESS in 3h 01m 55s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1b38faba36b4f12be6ef682d5e80a4c : SUCCESS in 2h 49m 21s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b037a34c124046328bdd77ec7c6a473e : SUCCESS in 2h 50m 44s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/04ef8ba01441487186b06d9b42917065 : SUCCESS in 1h 37m 01s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6aed184ef5f442c2ba892ea33bf72a05 : SUCCESS in 3h 00m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6522af6d3004639a1a4f7256895d955 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/239126952e124726a1a35608e380ce71 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":11},{"id":"6460863c980dca0a24bacf456eff570807ed26ab","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-09-16 09:53:24.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"f763cbf8292c2c3a08d5c510539c75e27b48bceb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 09:56:57.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"a57ff12e8eecce6301d440f113bebd93aa745941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 13:07:30.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77975a57fb1147a2a205d34284a7511c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abbc804a3f3b46c4a07fc7befe474446 : SUCCESS in 14m 36s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ff4136f570fa4a2a9e67769c9c902a58 : SUCCESS in 1h 47m 11s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/5e211a9f82b14ace8d0b2bee1e2003f9 : SUCCESS in 2h 59m 45s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/52ca77593ff646f38e76d93eb15e7092 : SUCCESS in 2h 33m 06s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8595a287a98c47448c15fe86aaa7cd20 : SUCCESS in 2h 50m 22s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/33dfdf3b9782489fb582382ec4740ebe : SUCCESS in 1h 48m 27s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0461931965fe4263bd5d7acc11e6bdbb : SUCCESS in 2h 27m 01s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fbd1e8774de4583933b8613a6e544fb : SUCCESS in 3h 05m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a342af144e447fae72d5783f8e36e0 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecfb51e22de845128ac036d6d75ed688 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":11},{"id":"4291974976d4ec20ac0471f8648521bdcaf2d434","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 13:07:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"8a96f750742974d69f5c506aa6be512abc2b577c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 13:08:28.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e3125908bea4b95a836543313b05bed\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f07c8ca0ec647b480e8969602a33c3d : SUCCESS in 45s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"66ccf10bea5cd8337d74feba7e7c4a03c224c1b3","revisions":{"b0b1603b5a94f438f4a0f4a37440f0f2a988d298":{"kind":"REWORK","_number":1,"created":"2026-06-04 18:37:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/16/991716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/16/991716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/1"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"subject":"Remove v1 API","message":"Remove v1 API\n\nRemoved from freezer-api in I1f25522e61eaaf4d01d7ec6055e0df65bfe38008\n\nChange-Id: Id950ff110925b5fda2cac88539678a7c5047dc88\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/b0b1603b5a94f438f4a0f4a37440f0f2a988d298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/b0b1603b5a94f438f4a0f4a37440f0f2a988d298"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"1c5db7373574beb37c34338d75eb7e66f4b6549c":{"kind":"REWORK","_number":2,"created":"2026-06-04 18:38:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/16/991716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/16/991716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/2"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:38:25.000000000","tz":0},"subject":"Remove v1 API","message":"Remove v1 API\n\nRemoved from freezer-api in I1f25522e61eaaf4d01d7ec6055e0df65bfe38008\n\nChange-Id: Id950ff110925b5fda2cac88539678a7c5047dc88\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/1c5db7373574beb37c34338d75eb7e66f4b6549c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/1c5db7373574beb37c34338d75eb7e66f4b6549c"}]},"branch":"refs/heads/master"},"6501d759104e24e9e5852876f07cec5892313d19":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-07-01 11:57:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/991716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/16/991716/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/3"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-07-01 11:57:35.000000000","tz":0},"subject":"Remove v1 API","message":"Remove v1 API\n\nRemoved from freezer-api in I1f25522e61eaaf4d01d7ec6055e0df65bfe38008\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nDepends-On: https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995668\nChange-Id: Id950ff110925b5fda2cac88539678a7c5047dc88\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/6501d759104e24e9e5852876f07cec5892313d19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/6501d759104e24e9e5852876f07cec5892313d19"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e7ba90a65a288e19da6e10fceec332e68116f92a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-07-01 14:47:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/991716/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/16/991716/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/4"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-07-01 14:47:22.000000000","tz":0},"subject":"Remove v1 API","message":"Remove v1 API\n\nRemoved from freezer-api in I1f25522e61eaaf4d01d7ec6055e0df65bfe38008\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nDepends-On: https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995696\nChange-Id: Id950ff110925b5fda2cac88539678a7c5047dc88\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e7ba90a65a288e19da6e10fceec332e68116f92a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e7ba90a65a288e19da6e10fceec332e68116f92a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"96bcee059649037031930d802d56180b08560723":{"kind":"REWORK","_number":5,"created":"2026-07-01 15:01:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/991716/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/16/991716/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/5"}}},"commit":{"parents":[{"commit":"e154108444b41516c1f5e23cdbff463012fd6acb","subject":"Merge \"Add internal TLS coverage for API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/e154108444b41516c1f5e23cdbff463012fd6acb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-07-01 15:01:09.000000000","tz":0},"subject":"Remove v1 API","message":"Remove v1 API\n\nRemoved from freezer-api in I1f25522e61eaaf4d01d7ec6055e0df65bfe38008\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nDepends-On: https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995696\nChange-Id: Id950ff110925b5fda2cac88539678a7c5047dc88\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/96bcee059649037031930d802d56180b08560723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/96bcee059649037031930d802d56180b08560723"}]},"branch":"refs/heads/master"},"d4d587557773e3368c5e192793a2d81531ae7b31":{"kind":"REWORK","_number":6,"created":"2026-07-02 08:37:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/991716/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/16/991716/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/6"}}},"commit":{"parents":[{"commit":"612c2b9d2938cdcdeaaa6ab51ad0e7dc05449b40","subject":"Add support for centralized scheduler deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/612c2b9d2938cdcdeaaa6ab51ad0e7dc05449b40"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-02 08:37:28.000000000","tz":120},"subject":"Remove v1 API","message":"Remove v1 API\n\nRemoved from freezer-api in I1f25522e61eaaf4d01d7ec6055e0df65bfe38008\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nDepends-On: https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995696\nChange-Id: Id950ff110925b5fda2cac88539678a7c5047dc88\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d4d587557773e3368c5e192793a2d81531ae7b31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d4d587557773e3368c5e192793a2d81531ae7b31"}]},"branch":"refs/heads/master"},"1237d4e1d9552e90256a6361a7e0591bdd076ea9":{"kind":"NO_CHANGE","_number":7,"created":"2026-07-14 15:22:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/991716/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/16/991716/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/7"}}},"commit":{"parents":[{"commit":"f2cd1e0e73f783d7ba7e3abfe53ce34ea2812370","subject":"Add support for centralized scheduler deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/f2cd1e0e73f783d7ba7e3abfe53ce34ea2812370"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-14 15:22:11.000000000","tz":0},"subject":"Remove v1 API","message":"Remove v1 API\n\nRemoved from freezer-api in I1f25522e61eaaf4d01d7ec6055e0df65bfe38008\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nDepends-On: https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995696\nChange-Id: Id950ff110925b5fda2cac88539678a7c5047dc88\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/1237d4e1d9552e90256a6361a7e0591bdd076ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/1237d4e1d9552e90256a6361a7e0591bdd076ea9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"612c2b9d2938cdcdeaaa6ab51ad0e7dc05449b40","ours":"d4d587557773e3368c5e192793a2d81531ae7b31","theirs":"f2cd1e0e73f783d7ba7e3abfe53ce34ea2812370","merge_strategy":"recursive","contains_conflicts":false}},"287e72a9e7d613bc1343f0dc54fd395a69f34196":{"kind":"NO_CHANGE","_number":8,"created":"2026-07-24 15:24:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/991716/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/16/991716/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/8"}}},"commit":{"parents":[{"commit":"3e8a62598a350ad73cc751634e3f55923ff396d8","subject":"Add support for centralized scheduler deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/3e8a62598a350ad73cc751634e3f55923ff396d8"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-24 15:24:40.000000000","tz":0},"subject":"Remove v1 API","message":"Remove v1 API\n\nRemoved from freezer-api in I1f25522e61eaaf4d01d7ec6055e0df65bfe38008\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nDepends-On: https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995696\nChange-Id: Id950ff110925b5fda2cac88539678a7c5047dc88\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/287e72a9e7d613bc1343f0dc54fd395a69f34196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/287e72a9e7d613bc1343f0dc54fd395a69f34196"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"f2cd1e0e73f783d7ba7e3abfe53ce34ea2812370","ours":"1237d4e1d9552e90256a6361a7e0591bdd076ea9","theirs":"3e8a62598a350ad73cc751634e3f55923ff396d8","merge_strategy":"recursive","contains_conflicts":false}},"0815561ea27bf12541b60f587cd8d89af0a95190":{"kind":"NO_CHANGE","_number":9,"created":"2026-08-20 11:49:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/991716/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/16/991716/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/9"}}},"commit":{"parents":[{"commit":"cb1eb1f8a1f6cc01ecbe1bdc495d4e616d8f22c4","subject":"Add support for centralized scheduler deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/cb1eb1f8a1f6cc01ecbe1bdc495d4e616d8f22c4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-20 11:49:21.000000000","tz":0},"subject":"Remove v1 API","message":"Remove v1 API\n\nRemoved from freezer-api in I1f25522e61eaaf4d01d7ec6055e0df65bfe38008\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nDepends-On: https://review.opendev.org/c/openstack/freezer-tempest-plugin/+/995696\nChange-Id: Id950ff110925b5fda2cac88539678a7c5047dc88\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/0815561ea27bf12541b60f587cd8d89af0a95190"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/0815561ea27bf12541b60f587cd8d89af0a95190"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3e8a62598a350ad73cc751634e3f55923ff396d8","ours":"287e72a9e7d613bc1343f0dc54fd395a69f34196","theirs":"cb1eb1f8a1f6cc01ecbe1bdc495d4e616d8f22c4","merge_strategy":"recursive","contains_conflicts":false}},"57dfa3f90f62e53605165e48b54568deefcfaee9":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-08-26 16:19:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/16/991716/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/16/991716/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/10"}}},"commit":{"parents":[{"commit":"cb1eb1f8a1f6cc01ecbe1bdc495d4e616d8f22c4","subject":"Add support for centralized scheduler deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/cb1eb1f8a1f6cc01ecbe1bdc495d4e616d8f22c4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-26 16:19:24.000000000","tz":0},"subject":"Remove v1 API","message":"Remove v1 API\n\nRemoved from freezer-api in I1f25522e61eaaf4d01d7ec6055e0df65bfe38008\n\nChange-Id: Id950ff110925b5fda2cac88539678a7c5047dc88\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/57dfa3f90f62e53605165e48b54568deefcfaee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/57dfa3f90f62e53605165e48b54568deefcfaee9"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"base":"3e8a62598a350ad73cc751634e3f55923ff396d8","ours":"287e72a9e7d613bc1343f0dc54fd395a69f34196","theirs":"cb1eb1f8a1f6cc01ecbe1bdc495d4e616d8f22c4","merge_strategy":"recursive","contains_conflicts":false}},"66ccf10bea5cd8337d74feba7e7c4a03c224c1b3":{"kind":"REWORK","_number":11,"created":"2026-08-31 16:29:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/991716/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/16/991716/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/16/991716/11"}}},"commit":{"parents":[{"commit":"523e1c11bfcd78aaf1c7b47289e12a416648146e","subject":"Implement freezer-scheduler clustering support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/523e1c11bfcd78aaf1c7b47289e12a416648146e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-04 18:37:27.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-31 16:28:49.000000000","tz":120},"subject":"Remove v1 API","message":"Remove v1 API\n\nRemoved from freezer-api in I1f25522e61eaaf4d01d7ec6055e0df65bfe38008\n\nChange-Id: Id950ff110925b5fda2cac88539678a7c5047dc88\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/66ccf10bea5cd8337d74feba7e7c4a03c224c1b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/66ccf10bea5cd8337d74feba7e7c4a03c224c1b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
