)]}'
{"id":"openstack%2Fopenstack-ansible-os_freezer~998718","triplet_id":"openstack%2Fopenstack-ansible-os_freezer~master~I29ae48983e8dc1f4bc39d93885c7f85936138445","project":"openstack/openstack-ansible-os_freezer","branch":"master","topic":"osa/freezer","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-01 12:08:41.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I29ae48983e8dc1f4bc39d93885c7f85936138445","subject":"Implement freezer-scheduler clustering support","status":"MERGED","created":"2026-07-26 09:15:54.000000000","updated":"2026-09-01 15:30:14.000000000","submitted":"2026-09-01 15:29:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998718-osa/freezer","meta_rev_id":"aea854c092e7928f30594e25d9cc97d746b94099","_number":998718,"virtual_id_number":998718,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 15:29:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-08-27 11:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-31 23:15:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2026-09-01 12:08:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-26 15:25:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-27 08:49:18.000000000","updated_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"reviewer":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"state":"CC"},{"updated":"2026-08-27 11:16:32.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-08-31 23:15:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"991286744a0db8f2d2214376a50b294d772f685f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-26 09:15:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0783f2a8ef55175745eb88727e8971f82cefa9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-26 09:24:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0290cb3f5bfeab8a1f2aac5e9e1f653db401d453","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-26 12:02:22.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3c19855540acdd8d21ab0962a3bd179210aca756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-26 15:25:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca72bfdb2a3f4183bac8d5dda8525051\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1991da59711943d79711c232d66abc95 : SUCCESS in 8m 05s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e029604271da4624b9d553f7f5b59043 : TIMED_OUT in 3h 19m 29s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/ebe0bfb82f8441b09a6e861e832b4cd7 : SUCCESS in 2h 04m 09s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c4e7000e1f49417eb0282d66855451d9 : SUCCESS in 2h 16m 49s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd82ec3757394ff9b74c5541b091c645 : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/1de66f6941b347a2b51970753020d3a6 : SUCCESS in 1h 22m 19s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/01bf41445f7744b7b1eb28ddf42d6b11 : SUCCESS in 2h 09m 05s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3d73646426d4052869072521c79196a : SUCCESS in 2h 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8f46b8136f46528d6c7b832af06abb : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a3847a65e184837b44ad56b9875483c : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"36e76210587a012d7734a7ea3c64c4ca1bb48b1c","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-07-27 08:49:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b2c81c3074950ba48f285540a0b54ec53a3c5883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-27 12:07:51.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"375c1bdc1304186a3f1c501b1813c0b65a22ac0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 15:45:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aaf805bfcd744647880fc218de392cb9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0104fcad4c0491d81a583163b000d48 : SUCCESS in 9m 44s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/78360bc135d243caba9bcfed4d05d69e : TIMED_OUT in 3h 12m 48s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/731f00ade9a241f687602a48b1c329ed : SUCCESS in 2h 42m 53s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/78106556054047f6826a7fc0ced431cf : SUCCESS in 2h 39m 44s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf1c0f89d2824f18a05a50f5263880a7 : TIMED_OUT in 3h 10m 51s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/24aa1cd9994246339bd9c42856e51693 : SUCCESS in 3h 00m 31s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7c66d56781be448b81e651cbb34b56df : TIMED_OUT in 3h 09m 05s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b8259a6ac5e443697474157ff1b92cd : SUCCESS in 2h 45m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341ceb6dded34f51a4e152cbf74fb240 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f38f1bb7f37349d4a2a82204c8fa6e0c : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":4},{"id":"45224f1cf2defbdeabbd8f075696bd1aecaba2c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-28 08:41:04.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"9a2c3c941cc8daa893bb84b076358b9d74da8471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 12:13:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4b1ff8ea3d94aa9931828affa1b2865\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbda8398b9f640e890ba491fbd38e2ec : SUCCESS in 11m 37s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/8d2d18b257714e83a2b290061df0f210 : TIMED_OUT in 3h 20m 57s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/96b66411d80b4690b8a912dfbfd4db42 : FAILURE in 3h 04m 28s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eb8ad1d271ef4c2fa75ccfd4dfa12ac3 : TIMED_OUT in 3h 07m 57s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebf47b67f1e94046b4c11f71272ac8d8 : FAILURE in 3h 04m 19s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f8963c8803cd443c981e5c1af6def77e : FAILURE in 3h 05m 53s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/70fd380a365d498a9930427beeeafc5d : TIMED_OUT in 3h 09m 47s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ad552cf1238469b92574d70ce8e1555 : FAILURE in 3h 08m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0286506beb2a4f3385553584ba88780d : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/245383f2369440dea9c1b79cd7d4f435 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":5},{"id":"f586c26289c0ac015eec397e1572d102088103e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-20 11:50:23.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"9bd16d18b5200859c6118b96593775958b840136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 14:00:25.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible/+/995111 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"59bd52bad95e4c7814c889c2abdea4e9b11ddf5f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-24 08:28:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0df1088210dffe2ef520c3dd18eace8f36700251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 11:47:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2e13f27221941e79fdf4dcdb9a3c239\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/147b2fb62b634d5ca813b9feba9b1970 : SUCCESS in 16m 38s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/741f39d2f1674063911b391145742edb : SUCCESS in 1h 44m 51s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/597bc1048274454ba8173744815ceea1 : SUCCESS in 2h 53m 06s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5c60121de71a4e56806017b405c901db : SUCCESS in 2h 28m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b26ed057010406e9b8bc3d61cd8d217 : SUCCESS in 3h 05m 24s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/83ca80043e79483fa8259966145bfc7a : SUCCESS in 1h 35m 18s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ed718d7395304d2881b3ebcbeb915b26 : SUCCESS in 1h 50m 22s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14177cb486a34017a96da5181d7d7d62 : SUCCESS in 2h 16m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a23f3c17e26e42d8a7852c90ce0b8784 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2371fbcdb5e34ddb9eb03de1d3d22932 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":6},{"id":"b57539f73bd749d8d1d8c15d9b376ab325f2d0ed","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-08-27 11:16:32.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"8b7f2e90c4554e3f25a8536d3e7efd297cb90c80","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 23:15:23.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"aeab891bfaa35d0a2d04207e98898d1d6bee1624","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-01 12:08:41.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e51fd93f11b4b103511f19b0e4f4bcf165539fb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:10:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"28dca482092821330c58ed16db3193b7a39e5b28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 15:29:10.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/943e87606191499fbf12c6b30e54852c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/712e3d8c43a04097acae896884f55d99 : SUCCESS in 11m 36s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/99a29f46d40d4256ba07694fd9876a82 : SUCCESS in 1h 34m 40s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/89736f99669f4401a7e2f56210b28780 : SUCCESS in 3h 13m 35s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fb837885f8d441209486938cf91e4f9b : SUCCESS in 2h 19m 55s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11d0213b3b3345a8b66e1160ad48dde2 : SUCCESS in 1h 29m 49s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/02e773c215a04b3db37d25eb6d49f06b : SUCCESS in 2h 19m 57s\n- openstack-ansible-deploy-aio_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3adb270ae3f440bab9b488ff6b545435 : SUCCESS in 1h 45m 13s\n- openstack-ansible-deploy-aio_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb9504a900e446d089f9ab42d39e0a2a : SUCCESS in 1h 32m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/722e748d322645f39fa640fc0fb7592e : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1944e080e6224041bf831abbf3b31f39 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":6},{"id":"dc67c1a55fdf585e8019103e53a0187c8baf8b3a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 15:29:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"aea854c092e7928f30594e25d9cc97d746b94099","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 15:30:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/727d9bbed01b42919d0e2afc4d9adc4f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d6aef8d5474698ae9f118bebf3cd5e : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"523e1c11bfcd78aaf1c7b47289e12a416648146e","revisions":{"de7cfa8b54896bda79680688b990ae891eb17b8c":{"kind":"REWORK","_number":1,"created":"2026-07-26 09:15:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/998718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/18/998718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/1"}}},"commit":{"parents":[{"commit":"d4827f262ed555c4d8f82a6441e2fc0bbe44001d","subject":"Add freezer-agent configuration file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d4827f262ed555c4d8f82a6441e2fc0bbe44001d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-26 09:14:13.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-26 09:15:43.000000000","tz":120},"subject":"Implement freezer-scheduler clustering support","message":"Implement freezer-scheduler clustering support\n\nThis adds configuration support for freezer scheduler coordination,\nwhich is required to avoid race conditions when running the scheduler in\nHA mode.\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/998522\nChange-Id: I29ae48983e8dc1f4bc39d93885c7f85936138445\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/de7cfa8b54896bda79680688b990ae891eb17b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/de7cfa8b54896bda79680688b990ae891eb17b8c"}]},"branch":"refs/heads/master"},"d2279615eed552657c0a65313b0ac0a01fd9e3b5":{"kind":"REWORK","_number":2,"created":"2026-07-26 09:24:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/998718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/18/998718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/2"}}},"commit":{"parents":[{"commit":"d4827f262ed555c4d8f82a6441e2fc0bbe44001d","subject":"Add freezer-agent configuration file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d4827f262ed555c4d8f82a6441e2fc0bbe44001d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-26 09:14:13.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-26 09:24:03.000000000","tz":120},"subject":"Implement freezer-scheduler clustering support","message":"Implement freezer-scheduler clustering support\n\nThis adds configuration support for freezer scheduler coordination,\nwhich is required to avoid race conditions when running the scheduler in\nHA mode.\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/998522\nChange-Id: I29ae48983e8dc1f4bc39d93885c7f85936138445\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d2279615eed552657c0a65313b0ac0a01fd9e3b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d2279615eed552657c0a65313b0ac0a01fd9e3b5"}]},"branch":"refs/heads/master"},"7ca7c08f0614e7a1763eb4791b70890477a6b203":{"kind":"REWORK","_number":3,"created":"2026-07-26 12:02:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/998718/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/18/998718/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/3"}}},"commit":{"parents":[{"commit":"d4827f262ed555c4d8f82a6441e2fc0bbe44001d","subject":"Add freezer-agent configuration file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d4827f262ed555c4d8f82a6441e2fc0bbe44001d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-26 09:14:13.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-26 12:02:22.000000000","tz":0},"subject":"Implement freezer-scheduler clustering support","message":"Implement freezer-scheduler clustering support\n\nThis adds configuration support for freezer scheduler coordination,\nwhich is required to avoid race conditions when running the scheduler in\nHA mode.\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/998522\nChange-Id: I29ae48983e8dc1f4bc39d93885c7f85936138445\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/7ca7c08f0614e7a1763eb4791b70890477a6b203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/7ca7c08f0614e7a1763eb4791b70890477a6b203"}]},"branch":"refs/heads/master"},"c71bf67cd2604b6b31a50a8113fd36b0c7cef869":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-07-27 12:07:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/998718/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/18/998718/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/4"}}},"commit":{"parents":[{"commit":"d4827f262ed555c4d8f82a6441e2fc0bbe44001d","subject":"Add freezer-agent configuration file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d4827f262ed555c4d8f82a6441e2fc0bbe44001d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-26 09:14:13.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-27 12:07:50.000000000","tz":0},"subject":"Implement freezer-scheduler clustering support","message":"Implement freezer-scheduler clustering support\n\nThis adds configuration support for freezer scheduler coordination,\nwhich is required to avoid race conditions when running the scheduler in\nHA mode.\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/998522\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/998794\nChange-Id: I29ae48983e8dc1f4bc39d93885c7f85936138445\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/c71bf67cd2604b6b31a50a8113fd36b0c7cef869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/c71bf67cd2604b6b31a50a8113fd36b0c7cef869"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fa8fe3c64fbabc58c0888b90e10684b4687bccc0":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-07-28 08:41:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/998718/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/18/998718/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/5"}}},"commit":{"parents":[{"commit":"d4827f262ed555c4d8f82a6441e2fc0bbe44001d","subject":"Add freezer-agent configuration file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/d4827f262ed555c4d8f82a6441e2fc0bbe44001d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-26 09:14:13.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-28 08:41:04.000000000","tz":0},"subject":"Implement freezer-scheduler clustering support","message":"Implement freezer-scheduler clustering support\n\nThis adds configuration support for freezer scheduler coordination,\nwhich is required to avoid race conditions when running the scheduler in\nHA mode.\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/998522\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/998794\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nChange-Id: I29ae48983e8dc1f4bc39d93885c7f85936138445\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/fa8fe3c64fbabc58c0888b90e10684b4687bccc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/fa8fe3c64fbabc58c0888b90e10684b4687bccc0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"523e1c11bfcd78aaf1c7b47289e12a416648146e":{"kind":"NO_CHANGE","_number":6,"created":"2026-08-20 11:50:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/998718/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_freezer","ref":"refs/changes/18/998718/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_freezer refs/changes/18/998718/6"}}},"commit":{"parents":[{"commit":"95a5cf0d11691f9dddeeafecbe1c38450dafeb3b","subject":"Add freezer-agent configuration file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/95a5cf0d11691f9dddeeafecbe1c38450dafeb3b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-26 09:14:13.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-20 11:50:23.000000000","tz":0},"subject":"Implement freezer-scheduler clustering support","message":"Implement freezer-scheduler clustering support\n\nThis adds configuration support for freezer scheduler coordination,\nwhich is required to avoid race conditions when running the scheduler in\nHA mode.\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/998522\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/998794\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995111\nChange-Id: I29ae48983e8dc1f4bc39d93885c7f85936138445\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/523e1c11bfcd78aaf1c7b47289e12a416648146e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_freezer/commit/523e1c11bfcd78aaf1c7b47289e12a416648146e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d4827f262ed555c4d8f82a6441e2fc0bbe44001d","ours":"fa8fe3c64fbabc58c0888b90e10684b4687bccc0","theirs":"95a5cf0d11691f9dddeeafecbe1c38450dafeb3b","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
