)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~606022","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Iff2c977de0c27f62a832170106831231a7438447","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"Iff2c977de0c27f62a832170106831231a7438447","subject":"use import_playbook instead of include","status":"MERGED","created":"2018-09-28 09:56:29.000000000","updated":"2019-10-08 13:43:28.000000000","submitted":"2019-10-08 13:42:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"606022-1570542126582-01385176","meta_rev_id":"5b0eadc88063a65949b7403a7847c1331cc54c05","_number":606022,"virtual_id_number":606022,"owner":{"_account_id":27382,"name":"brandon zhao","email":"zhaolong@inspur.com","username":"brandonzhao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 13:42:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-09-20 17:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-20 17:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-20 17:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-09-20 17:30:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 19:27:44.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-20 13:26:36.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-09-20 17:30:33.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-08 13:42:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08b92e28333c31eb89bc75da1382a8ac5182f6a9","author":{"_account_id":27382,"name":"brandon zhao","email":"zhaolong@inspur.com","username":"brandonzhao"},"date":"2018-09-28 09:56:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"772fd8c0c42bc19d66e36fbd2b4b4b1e2fb25d4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 14:18:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/22/606022/1/check/build-openstack-sphinx-docs/b5b7d83/html/ : SUCCESS in 2m 51s\n- openstack-ansible-linters http://logs.openstack.org/22/606022/1/check/openstack-ansible-linters/629de12/ : SUCCESS in 6m 10s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/22/606022/1/check/openstack-ansible-functional-centos-7/00d9339/ : SUCCESS in 36m 51s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/22/606022/1/check/openstack-ansible-functional-opensuse-150/ce3d85b/ : SUCCESS in 45m 11s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/22/606022/1/check/openstack-ansible-functional-opensuse-423/9f29fd8/ : SUCCESS in 30m 59s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/22/606022/1/check/openstack-ansible-functional-ubuntu-bionic/446759f/ : SUCCESS in 36m 26s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/22/606022/1/check/openstack-ansible-functional-ubuntu-xenial/8d67595/ : SUCCESS in 32m 22s\n- openstack-tox-docs http://logs.openstack.org/22/606022/1/check/openstack-tox-docs/d31307c/html/ : SUCCESS in 2m 38s\n- openstack-ansible-nfs_glance http://logs.openstack.org/22/606022/1/check/openstack-ansible-nfs_glance/5c786f5/ : SUCCESS in 36m 56s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/22/606022/1/check/openstack-ansible-v2_registry_enabled/157fdae/ : SUCCESS in 33m 14s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/22/606022/1/check/openstack-ansible-functional-distro_install-ubuntu-xenial/d696500/ : SUCCESS in 33m 47s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/22/606022/1/check/openstack-ansible-functional-distro_install-centos-7/075894e/ : SUCCESS in 33m 06s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/22/606022/1/check/openstack-ansible-functional-distro_install-opensuse-423/c6d3dcb/ : SUCCESS in 32m 31s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/22/606022/1/check/openstack-ansible-functional-distro_install-opensuse-150/2a3cd30/ : SUCCESS in 44m 22s","accounts_in_message":[],"_revision_number":1},{"id":"91f8d957f5adeb634a76a91a9a001a290337c05b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-19 16:43:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9b630b4bb639e1253a4d8575fefbe5bc47c45aa2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-19 19:27:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b6539545f23c799a9c30e49f34ce6ad9d3e8f4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:28:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1675721b050b369907693c08775fbc5fc3202e00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 21:40:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/18151097185c4b1a99b1ed50001d1c04 : SUCCESS in 8m 09s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/edf31f84d24d421b83a305df281acdc4 : SUCCESS in 1h 27m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/604b2dc064a441ccbd32e533381b2a7f : SUCCESS in 1h 30m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c768a0f5d28f4a1a955fd5e4edb6a192 : SUCCESS in 1h 34m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2624ca5c2574044a84a91d449afda0c : SUCCESS in 1h 26m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d927f23d6adf48e498391462ae7e312a : SUCCESS in 1h 11m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35e3a3e82d4b43de97ec3f97056c6df5 : SUCCESS in 49m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd9c83ee0144475bcad4ed84c85fc45 : SUCCESS in 3m 50s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/9c68c1508440457299ee83f8effff7be : FAILURE in 35m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b7726dbb99a5dc99fa9ff65a5f39f60c0c149c0b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-20 07:07:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0f1132a72eb67ca7acc3af151c82b971fa6eb690","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 09:38:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/500dc82074134cbba29759df781d4867 : SUCCESS in 4m 40s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/5d69ca2905f44f289c00078fa117e5b2 : SUCCESS in 1h 21m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0e831f1be2924a57b167510ca83bf5ed : SUCCESS in 1h 56m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/559837b043de43fe99edf764cdd09459 : SUCCESS in 1h 34m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ce621d96f9e4cdbb187207f6243e00f : SUCCESS in 1h 17m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3f44fae73dfa405c94bca0d7c675baf5 : SUCCESS in 1h 24m 23s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b899b23af6b340cf8e0274c107129047 : FAILURE in 36m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/140f163afdf24913a167c3e3eb0c1ba9 : SUCCESS in 1h 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50d6dd7639b14e748b7f8dbdda855990 : SUCCESS in 4m 21s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/895d2e6970194238b8c5321f56841b6d : FAILURE in 40m 35s","accounts_in_message":[],"_revision_number":1},{"id":"89f0b58f0c947a029e2f4240c0d915a3a9b8f0b1","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-09-20 13:26:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9a1a9105586163fb9025de5a42b665d15dc9b089","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 15:57:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e7925fe5c17e438ea5b324e54485c9b5 : SUCCESS in 4m 27s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2e3c2800aca4488b812c5dc23bb35552 : SUCCESS in 1h 22m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3441ea707d454065ac359093e1397fbf : SUCCESS in 1h 40m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8e8c14335299474e936706bec6a7b206 : SUCCESS in 1h 27m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/877f94f77e6d42adab0f6cf613e9b049 : SUCCESS in 1h 12m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/baf479ec299840f799bb8eb9a306c912 : SUCCESS in 1h 20m 52s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cc55fcb51d7f48b8a5a5008b3132efab : FAILURE in 29m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15408f0b4d9d4537ac4670f389092dc0 : SUCCESS in 1h 02m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170789886e3b4b89a12f682654d17880 : SUCCESS in 3m 34s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/c925f9780db5429bb2a672ebe7d5163a : FAILURE in 44m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f478f6a001802082127c17d05bdebd4a0e29d9fd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 17:30:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"dd4ca9e222e9fedf8eaae5bd84acb1709ab9e513","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 17:30:33.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e617cc92b2c42d11e92a2e8d3bacc26997452f74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 19:20:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7b228bc3b901414ebaefb256a54c0ac3 : SUCCESS in 5m 44s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e7c57107292d4020822cc3d5c91b76e4 : SUCCESS in 1h 22m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0bc10cd674a748c5aeb038344fee3412 : SUCCESS in 1h 40m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c2ab27fdccb944c58da3050c706aded2 : SUCCESS in 1h 27m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/965fe86c1b0b4e57a564ede197c1f212 : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/53ad6d74666e47228a9f060715b0b709 : SUCCESS in 1h 13m 50s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cf27ee0e4aed4295beda3f50bde23a34 : FAILURE in 1h 03m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2152c280a6a44d6184f9c10759dd9c97 : SUCCESS in 58m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/232bacd17bfd41b594abb5a9f38301d9 : SUCCESS in 3m 14s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/e813b6a9e76d496bbc08ebcf3a2d4ec7 : FAILURE in 35m 17s","accounts_in_message":[],"_revision_number":2},{"id":"8445d8a6b31f589e3fbf3c375ecde08f989a8c62","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 21:54:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"840f635ea72553cd8b7ed11d317d727d5140c70c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 23:42:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/68d7e25d0ace47ea884db271f9348c83 : SUCCESS in 6m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/151e0087b12748b0a31cc8fcf0b741ab : SUCCESS in 1h 27m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cda113daf7144c2ab1323e4d6e50606a : SUCCESS in 1h 37m 53s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ccf97c26837a4d67ab99812bc69f92e4 : SUCCESS in 1h 31m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e2a49cb5d1446af9c6320ad6c4219ab : SUCCESS in 1h 19m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aaf2042651cb4c418834df4ba488f590 : SUCCESS in 1h 07m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ff0f26e4008846448dcd5fa4e369c3f1 : FAILURE in 51m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8dd1220f480492aa652254a74e1862f : SUCCESS in 58m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d368c69c0c04c42bd5fab89ec21bf53 : SUCCESS in 3m 39s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/4fae9ab53c0e4000ac8afcf6f75a78d6 : FAILURE in 46m 18s","accounts_in_message":[],"_revision_number":2},{"id":"43333c5b6323f2f555b86326f6b4076694bebeb5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-08 10:33:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"18bf084500cd03b34fc68fb00d4779a8784fe303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:10:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2e8fcba8ff7d40b9906f32cdc12b00de : SUCCESS in 5m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/81adb11ca2f64dd588880aacdbe5335b : SUCCESS in 1h 08m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/27d5b865781047b68cab830d24b51534 : SUCCESS in 1h 31m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/af52ba63df904e62a9f1166753fd7df8 : SUCCESS in 1h 24m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f729f501709d49668485531563b8eab7 : SUCCESS in 1h 02m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f06713ef973b46f5bb2c68c1d2cc4ea1 : SUCCESS in 1h 11m 35s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/43516ea5b42640e4b1466b3ae4c21b1d : FAILURE in 34m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7907d9586de04d428678b16dfa8328c7 : SUCCESS in 54m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5611d2d617c4aee9fcdbc68c33b2b9e : SUCCESS in 3m 20s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/cc5664ca055248cba892839f2f934fd5 : SUCCESS in 37m 00s","accounts_in_message":[],"_revision_number":2},{"id":"6dcba178cffcb30983e306f1a92fbe432c1ac454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 12:10:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"11e53b69729a5513c2700278b6a8dde821407098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 13:42:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/16102ec9d3344f7fbef6bcc5c2a864d5 : SUCCESS in 5m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/05ab15204d004035bf766602172a8423 : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/87a618a1a1484a8587b325aac401fa78 : SUCCESS in 1h 31m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a928328eb93442a585ee9607a40717f3 : SUCCESS in 1h 24m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cf786ea4976429d8cae27bef9a5df9b : SUCCESS in 1h 11m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d8dfc4a8dde147d5be605149f05320d0 : SUCCESS in 1h 09m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87194e190b984ebc968374a14ba748af : SUCCESS in 51m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09ea1312ae584c8bbfd85bdee4fdbef3 : SUCCESS in 2m 29s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/d369e6f8ad964466a592734fcdc7f55a : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":2},{"id":"44f0ec3670c8909e4a700e0001da1ab6e38a9eee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 13:42:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5b0eadc88063a65949b7403a7847c1331cc54c05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 13:43:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50799c6fcb3d4ed5a9d353024b7a36c2 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3c262af20fffa384cc5cd4b7a5854ee6b8ab2550","revisions":{"8d8cbc70a87f5b65143de4382258234354cafb24":{"kind":"REWORK","_number":1,"created":"2018-09-28 09:56:29.000000000","uploader":{"_account_id":27382,"name":"brandon zhao","email":"zhaolong@inspur.com","username":"brandonzhao"},"ref":"refs/changes/22/606022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/22/606022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/606022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/606022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/606022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/606022/1"}}},"commit":{"parents":[{"commit":"62e3c56ca37ab6059816b59198a793ebce9f74df","subject":"Merge \"Update messaging notification configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/62e3c56ca37ab6059816b59198a793ebce9f74df"}]}],"author":{"name":"brandonzhao","email":"zhaolong@inspur.com","date":"2018-09-28 09:53:54.000000000","tz":480},"committer":{"name":"brandonzhao","email":"zhaolong@inspur.com","date":"2018-09-28 09:55:22.000000000","tz":480},"subject":"use import_playbook instead of include","message":"use import_playbook instead of include\n\ninclude is marked as deprecated since ansible 2.4[0]\n\nSwitch to import_playbook as necessary\n\n[0] https://docs.ansible.com/ansible/2.4/include_module.html#deprecated\n\nChange-Id: Iff2c977de0c27f62a832170106831231a7438447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8d8cbc70a87f5b65143de4382258234354cafb24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8d8cbc70a87f5b65143de4382258234354cafb24"}]},"branch":"refs/heads/master"},"3c262af20fffa384cc5cd4b7a5854ee6b8ab2550":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-20 17:30:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/606022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/22/606022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/606022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/606022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/606022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/606022/2"}}},"commit":{"parents":[{"commit":"ae681ded4e6bc033c75f72b7a429a5b04dd0a676","subject":"Merge \"Allow venv python interpreter to be overridden\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ae681ded4e6bc033c75f72b7a429a5b04dd0a676"}]}],"author":{"name":"brandonzhao","email":"zhaolong@inspur.com","date":"2018-09-28 09:53:54.000000000","tz":480},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-09-20 17:30:30.000000000","tz":0},"subject":"use import_playbook instead of include","message":"use import_playbook instead of include\n\ninclude is marked as deprecated since ansible 2.4[0]\n\nSwitch to import_playbook as necessary\n\n[0] https://docs.ansible.com/ansible/2.4/include_module.html#deprecated\n\nChange-Id: Iff2c977de0c27f62a832170106831231a7438447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3c262af20fffa384cc5cd4b7a5854ee6b8ab2550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3c262af20fffa384cc5cd4b7a5854ee6b8ab2550"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
