)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~613924","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I7403ac9bd85ea3ed149e13cb57c51039602f6ba1","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"nfs_folder","hashtags":[],"change_id":"I7403ac9bd85ea3ed149e13cb57c51039602f6ba1","subject":"Turn the filesystem directory to be configurable","status":"MERGED","created":"2018-10-29 13:14:52.000000000","updated":"2019-10-09 15:47:56.000000000","submitted":"2019-10-09 15:46:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"613924-1570635977666-4f81ad49","meta_rev_id":"f93ee331483e6509aaab25df4777eccd208c53e8","_number":613924,"virtual_id_number":613924,"owner":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28561,"name":"Nikos Karandreas","email":"nick@intracom-telecom.com","username":"nikoskarandreas"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 15:46:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28561,"name":"Nikos Karandreas","email":"nick@intracom-telecom.com","username":"nikoskarandreas"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},{"value":2,"date":"2019-10-09 14:09:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-07 20:21:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28561,"name":"Nikos Karandreas","email":"nick@intracom-telecom.com","username":"nikoskarandreas"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},{"value":1,"date":"2019-10-09 14:09:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"date":"2019-09-30 10:30:37.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28561,"name":"Nikos Karandreas","email":"nick@intracom-telecom.com","username":"nikoskarandreas"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28561,"name":"Nikos Karandreas","email":"nick@intracom-telecom.com","username":"nikoskarandreas"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-30 08:11:54.000000000","updated_by":{"_account_id":28561,"name":"Nikos Karandreas","email":"nick@intracom-telecom.com","username":"nikoskarandreas"},"reviewer":{"_account_id":28561,"name":"Nikos Karandreas","email":"nick@intracom-telecom.com","username":"nikoskarandreas"},"state":"REVIEWER"},{"updated":"2018-10-30 19:56:55.000000000","updated_by":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"reviewer":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"state":"REVIEWER"},{"updated":"2018-11-09 09:28:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2018-11-09 09:30:21.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-09-30 10:30:37.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-07 20:21:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-09 14:09:21.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-10-09 15:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eabecfd14967ab3cfc51fa2f1a014f930f75c350","author":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"date":"2018-10-29 13:14:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be8f82522dfb4da12101a4b09e76ebdb7ac0aa4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-29 16:00:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/613924/1/check/openstack-ansible-functional-distro_install-centos-7/b390ea2/ : SUCCESS in 43m 52s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/613924/1/check/openstack-ansible-functional-distro_install-opensuse-150/2f7a506/ : SUCCESS in 32m 00s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/613924/1/check/openstack-ansible-functional-distro_install-opensuse-423/4de1bf5/ : SUCCESS in 27m 38s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/613924/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/5897281/ : FAILURE in 31m 13s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/613924/1/check/openstack-ansible-linters/5a1f7f1/ : SUCCESS in 5m 20s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/613924/1/check/openstack-ansible-functional-centos-7/1ad5dd2/ : SUCCESS in 38m 02s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/613924/1/check/openstack-ansible-functional-opensuse-150/6e447df/ : SUCCESS in 42m 27s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/613924/1/check/openstack-ansible-functional-opensuse-423/3750a4c/ : SUCCESS in 37m 04s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/613924/1/check/openstack-ansible-functional-ubuntu-bionic/c97935c/ : SUCCESS in 37m 35s\n- openstack-tox-docs http://logs.openstack.org/24/613924/1/check/openstack-tox-docs/5117f6a/html/ : SUCCESS in 2m 47s\n- openstack-ansible-nfs_glance http://logs.openstack.org/24/613924/1/check/openstack-ansible-nfs_glance/a5f5ef0/ : SUCCESS in 38m 36s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/24/613924/1/check/openstack-ansible-v2_registry_enabled/cf8afd0/ : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":1},{"id":"dbd9bbaca18f91f3f1fd173da775083e75a378fd","author":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"date":"2018-10-30 07:21:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f87f525440bd5e2eab396dbc5e97dac368517fcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-30 08:14:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/613924/2/check/openstack-ansible-functional-distro_install-centos-7/91b66f0/ : SUCCESS in 30m 11s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/613924/2/check/openstack-ansible-functional-distro_install-opensuse-150/3b089ec/ : SUCCESS in 41m 56s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/613924/2/check/openstack-ansible-functional-distro_install-opensuse-423/29abea6/ : SUCCESS in 27m 00s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/613924/2/check/openstack-ansible-functional-distro_install-ubuntu-bionic/b03244b/ : FAILURE in 26m 58s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/613924/2/check/openstack-ansible-linters/472be93/ : SUCCESS in 4m 31s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/613924/2/check/openstack-ansible-functional-centos-7/7af6976/ : SUCCESS in 37m 36s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/613924/2/check/openstack-ansible-functional-opensuse-150/2cf53ba/ : SUCCESS in 51m 41s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/613924/2/check/openstack-ansible-functional-opensuse-423/d22dc3d/ : SUCCESS in 33m 55s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/613924/2/check/openstack-ansible-functional-ubuntu-bionic/15f48bb/ : SUCCESS in 46m 11s\n- openstack-tox-docs http://logs.openstack.org/24/613924/2/check/openstack-tox-docs/52f7fe8/html/ : SUCCESS in 2m 38s\n- openstack-ansible-nfs_glance http://logs.openstack.org/24/613924/2/check/openstack-ansible-nfs_glance/19298f7/ : FAILURE in 12m 49s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/24/613924/2/check/openstack-ansible-v2_registry_enabled/11ea6c3/ : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c90df16a37abb495f2c6f212a9a20d9860f14ba3","author":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"date":"2018-10-30 08:19:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9795bddd23f5001df170514390a087d329a88492","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-30 09:25:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/613924/3/check/openstack-ansible-functional-distro_install-centos-7/dd9c10b/ : SUCCESS in 30m 49s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/613924/3/check/openstack-ansible-functional-distro_install-opensuse-150/92c2a15/ : SUCCESS in 44m 41s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/613924/3/check/openstack-ansible-functional-distro_install-opensuse-423/76e12f1/ : FAILURE in 8m 28s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/613924/3/check/openstack-ansible-functional-distro_install-ubuntu-bionic/79fe8d9/ : FAILURE in 25m 40s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/613924/3/check/openstack-ansible-linters/6b6ee0f/ : SUCCESS in 6m 22s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/613924/3/check/openstack-ansible-functional-centos-7/de246a1/ : SUCCESS in 40m 40s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/613924/3/check/openstack-ansible-functional-opensuse-150/27b7218/ : SUCCESS in 54m 40s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/613924/3/check/openstack-ansible-functional-opensuse-423/e9a50f2/ : FAILURE in 55m 31s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/613924/3/check/openstack-ansible-functional-ubuntu-bionic/d7fddf3/ : SUCCESS in 33m 30s\n- openstack-tox-docs http://logs.openstack.org/24/613924/3/check/openstack-tox-docs/8ce1968/html/ : SUCCESS in 3m 52s\n- openstack-ansible-nfs_glance http://logs.openstack.org/24/613924/3/check/openstack-ansible-nfs_glance/b5d9f1f/ : SUCCESS in 33m 36s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/24/613924/3/check/openstack-ansible-v2_registry_enabled/9bff3e7/ : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":3},{"id":"d4f40deeef1a1538155ec6443717e6b95c1778f2","author":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"date":"2018-10-30 10:05:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"01afc746f5efdc5ab3ad5493f27e8fc3718a6271","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-30 10:57:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/613924/4/check/openstack-ansible-functional-distro_install-centos-7/bf43ba9/ : SUCCESS in 41m 10s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/613924/4/check/openstack-ansible-functional-distro_install-opensuse-150/c57327b/ : SUCCESS in 41m 20s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/613924/4/check/openstack-ansible-functional-distro_install-opensuse-423/54ca249/ : SUCCESS in 27m 03s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/613924/4/check/openstack-ansible-functional-distro_install-ubuntu-bionic/0ad9e88/ : FAILURE in 31m 25s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/613924/4/check/openstack-ansible-linters/4208a22/ : SUCCESS in 5m 07s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/613924/4/check/openstack-ansible-functional-centos-7/f40236e/ : SUCCESS in 37m 35s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/613924/4/check/openstack-ansible-functional-opensuse-150/8db21f6/ : SUCCESS in 45m 35s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/613924/4/check/openstack-ansible-functional-opensuse-423/bdb4d48/ : FAILURE in 33m 36s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/613924/4/check/openstack-ansible-functional-ubuntu-bionic/96dfbd6/ : SUCCESS in 41m 08s\n- openstack-tox-docs http://logs.openstack.org/24/613924/4/check/openstack-tox-docs/95af919/html/ : SUCCESS in 3m 50s\n- openstack-ansible-nfs_glance http://logs.openstack.org/24/613924/4/check/openstack-ansible-nfs_glance/97a81a3/ : SUCCESS in 37m 09s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/24/613924/4/check/openstack-ansible-v2_registry_enabled/901cabe/ : SUCCESS in 30m 51s","accounts_in_message":[],"_revision_number":4},{"id":"b5cd8849ee6880414bbdd5e200d2c6a909a45d74","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2018-10-30 19:56:55.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d44015ce70ae7682789206cc52ec741e69efaf88","author":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"date":"2018-10-31 16:10:07.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"da951ba59891c2a630493a1728d97b2dd9e33de0","author":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"date":"2018-11-07 14:47:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f86a7699a7eb9d3401f7e989f1d9a3106d93dc44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 15:36:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/613924/5/check/openstack-ansible-functional-distro_install-centos-7/9d60e42/ : SUCCESS in 30m 03s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/613924/5/check/openstack-ansible-functional-distro_install-opensuse-150/ad856f5/ : SUCCESS in 44m 55s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/613924/5/check/openstack-ansible-functional-distro_install-opensuse-423/603d8eb/ : SUCCESS in 26m 06s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/613924/5/check/openstack-ansible-functional-distro_install-ubuntu-bionic/42eb2f1/ : SUCCESS in 31m 01s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/613924/5/check/openstack-ansible-linters/eca5560/ : SUCCESS in 8m 27s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/613924/5/check/openstack-ansible-functional-centos-7/2e6130e/ : SUCCESS in 40m 32s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/613924/5/check/openstack-ansible-functional-opensuse-150/7a77b8b/ : SUCCESS in 45m 25s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/613924/5/check/openstack-ansible-functional-opensuse-423/443cb12/ : SUCCESS in 32m 15s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/613924/5/check/openstack-ansible-functional-ubuntu-bionic/9cd48c3/ : SUCCESS in 36m 21s\n- openstack-tox-docs http://logs.openstack.org/24/613924/5/check/openstack-tox-docs/710878f/html/ : SUCCESS in 3m 43s\n- openstack-ansible-nfs_glance http://logs.openstack.org/24/613924/5/check/openstack-ansible-nfs_glance/ee38795/ : SUCCESS in 35m 41s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/24/613924/5/check/openstack-ansible-v2_registry_enabled/87223f8/ : SUCCESS in 32m 06s","accounts_in_message":[],"_revision_number":5},{"id":"3bf71ee3ecb360af9d04d5daa094473a27b6a693","author":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"date":"2018-11-07 16:11:25.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"5ef22c7732eab06f29e5989e9dc36e7efe0492d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 17:10:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/613924/6/check/openstack-ansible-functional-distro_install-centos-7/c7c9fc0/ : SUCCESS in 39m 10s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/613924/6/check/openstack-ansible-functional-distro_install-opensuse-150/54ca8b9/ : SUCCESS in 38m 52s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/613924/6/check/openstack-ansible-functional-distro_install-opensuse-423/802108a/ : SUCCESS in 28m 27s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/613924/6/check/openstack-ansible-functional-distro_install-ubuntu-bionic/dd28fa4/ : SUCCESS in 31m 37s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/613924/6/check/openstack-ansible-linters/487e187/ : SUCCESS in 11m 42s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/613924/6/check/openstack-ansible-functional-centos-7/7010544/ : SUCCESS in 38m 54s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/613924/6/check/openstack-ansible-functional-opensuse-150/be88781/ : SUCCESS in 47m 07s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/613924/6/check/openstack-ansible-functional-opensuse-423/cef313b/ : SUCCESS in 44m 14s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/613924/6/check/openstack-ansible-functional-ubuntu-bionic/30f73a5/ : SUCCESS in 33m 45s\n- openstack-tox-docs http://logs.openstack.org/24/613924/6/check/openstack-tox-docs/f2145f4/html/ : SUCCESS in 6m 22s\n- openstack-ansible-nfs_glance http://logs.openstack.org/24/613924/6/check/openstack-ansible-nfs_glance/0cad167/ : SUCCESS in 40m 51s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/24/613924/6/check/openstack-ansible-v2_registry_enabled/2bd8a3e/ : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":6},{"id":"260fb392a3bc28c1be07a25018a345cf2a979c4c","author":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"date":"2018-11-09 08:48:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7842f4053ba0afb85b5d9bfc87e621bb4d02e48c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 09:36:34.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-centos-7/3392202/ : SUCCESS in 31m 12s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-opensuse-150/6c48230/ : SUCCESS in 30m 54s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-opensuse-423/86f4caa/ : SUCCESS in 25m 05s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-ubuntu-bionic/62e615a/ : SUCCESS in 27m 25s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/613924/7/check/openstack-ansible-linters/f45a4ac/ : SUCCESS in 6m 21s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-centos-7/7b20a84/ : SUCCESS in 36m 09s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-opensuse-150/61b5543/ : SUCCESS in 46m 29s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-opensuse-423/d5abb93/ : SUCCESS in 32m 23s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-ubuntu-bionic/14b8214/ : SUCCESS in 33m 08s\n- openstack-tox-docs http://logs.openstack.org/24/613924/7/check/openstack-tox-docs/b48ad66/html/ : SUCCESS in 2m 57s\n- openstack-ansible-nfs_glance http://logs.openstack.org/24/613924/7/check/openstack-ansible-nfs_glance/ba71aaf/ : FAILURE in 34m 55s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/24/613924/7/check/openstack-ansible-v2_registry_enabled/4b2fbb7/ : SUCCESS in 30m 30s","accounts_in_message":[],"_revision_number":7},{"id":"7ce0023308535f2ce589cae32994e665f3e7b063","author":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"date":"2018-11-09 09:49:15.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"c94ffeaac16ce1426ab2a5b78d50d23bf69ff61a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 10:50:41.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-centos-7/d1d41bb/ : SUCCESS in 31m 00s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-opensuse-150/f46095b/ : SUCCESS in 42m 32s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-opensuse-423/f37dce7/ : SUCCESS in 29m 03s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-ubuntu-bionic/4d84e80/ : SUCCESS in 34m 13s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/613924/7/check/openstack-ansible-linters/579aadd/ : SUCCESS in 5m 29s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-centos-7/66a62e9/ : SUCCESS in 34m 26s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-opensuse-150/791843b/ : SUCCESS in 47m 12s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-opensuse-423/4949e7b/ : SUCCESS in 40m 29s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-ubuntu-bionic/0842120/ : SUCCESS in 44m 16s\n- openstack-tox-docs http://logs.openstack.org/24/613924/7/check/openstack-tox-docs/4e92cac/html/ : SUCCESS in 2m 58s\n- openstack-ansible-nfs_glance http://logs.openstack.org/24/613924/7/check/openstack-ansible-nfs_glance/f08d9cd/ : SUCCESS in 32m 43s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/24/613924/7/check/openstack-ansible-v2_registry_enabled/4bebcb6/ : FAILURE in 37m 10s","accounts_in_message":[],"_revision_number":7},{"id":"145bc3013ab98ce4b7ab90348c1041156ae62db8","author":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"date":"2018-11-09 10:51:36.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"7d346f6e2c4c74d34021836e959a44445c055663","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 11:38:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-centos-7/dd2aa20/ : SUCCESS in 29m 25s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-opensuse-150/197ab3e/ : SUCCESS in 41m 45s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-opensuse-423/64b19c8/ : SUCCESS in 26m 02s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-distro_install-ubuntu-bionic/b3dc3a2/ : SUCCESS in 28m 03s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/24/613924/7/check/openstack-ansible-linters/b711f53/ : SUCCESS in 6m 14s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-centos-7/99fbd5d/ : SUCCESS in 39m 26s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-opensuse-150/b551d6b/ : SUCCESS in 43m 04s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-opensuse-423/9831c32/ : SUCCESS in 33m 23s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/613924/7/check/openstack-ansible-functional-ubuntu-bionic/eb94cfd/ : SUCCESS in 33m 20s\n- openstack-tox-docs http://logs.openstack.org/24/613924/7/check/openstack-tox-docs/53279f6/html/ : SUCCESS in 2m 36s\n- openstack-ansible-nfs_glance http://logs.openstack.org/24/613924/7/check/openstack-ansible-nfs_glance/7e1ac09/ : SUCCESS in 35m 38s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/24/613924/7/check/openstack-ansible-v2_registry_enabled/ba73893/ : SUCCESS in 37m 29s","accounts_in_message":[],"_revision_number":7},{"id":"c61cff639ae6e5516fa458b434eed328a4ca6fb3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-19 16:45:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"88021a6bc23f0acbf0579e5d818a1432b5ce898e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 18:22:29.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/da3a2a0448a94f27acefe3ff5d148145 : SUCCESS in 5m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a0cdbfd8b70d4270b783fa3ecbb4d34b : SUCCESS in 1h 25m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4e6285048d8b4c6f8f64bc3dfe73591d : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/48ac305894c04855a21b8df0daf1e438 : SUCCESS in 1h 26m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3526b1adedbf43d3821d9f6c92cf112f : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7b74544b3179453789551eb0672c7d11 : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a46d8cdd69e04d7eb782310aae823854 : FAILURE in 29m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12ce061af4ff4d1babf4ddd3b14ed058 : SUCCESS in 48m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c326af138276455485b27d4893a5df13 : SUCCESS in 2m 58s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/0ca18f81dbce42efa41dbd5de7bdd8ee : FAILURE in 40m 36s","accounts_in_message":[],"_revision_number":8},{"id":"357dea5d19d49de205fde67e7d2d458382156501","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 09:02:07.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"b498f57415561aa2ccbb1f07a4c4d1fce4d656e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 10:58:51.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/44121d998ba5448f837cea24f6730a9c : SUCCESS in 6m 48s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/09bb6826c3c3449d974f56db6cd6f8b2 : SUCCESS in 1h 21m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5459a1e8c1d74b879c9b9eb296dfdff9 : SUCCESS in 1h 49m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/98fd19839421471ea20d15a1b14075ad : FAILURE in 1h 13m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6194748114a43c18413333f30dc1021 : SUCCESS in 1h 15m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/70c9bb353bac431586136162ddc3c0d2 : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a2c70b3829d24702becd96596e1972aa : FAILURE in 31m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1517faebc5a74a938e261dc66c2d8fd5 : SUCCESS in 57m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea3dc77be726483c8d2330f880584f39 : SUCCESS in 4m 18s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/2cd08d4ac0cc46f99ad084e9a9fcfeb6 : FAILURE in 37m 25s","accounts_in_message":[],"_revision_number":8},{"id":"e035491547fc38ed81d87ee9f573011ee51a3abb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 17:31:29.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"b5952942f03fd2f10a3986d0a4865ed795557b18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 19:26:57.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/37bd3d6bf8a24caeac3ecbc4a50bf78a : SUCCESS in 5m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/027a595d5d1d40b38728c9baa1389e2e : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3db37569a03e4e018915c8cce5c0968f : SUCCESS in 1h 47m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ccf66742b85a4caaaeaf7ffc626ac21a : SUCCESS in 1h 25m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66a62cfaec3c4a54a6c8c34ef15fb1b7 : SUCCESS in 1h 13m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8a051db39e014fbf8cc9a38b3a393384 : SUCCESS in 1h 18m 52s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/81dd5a36e338444d825bf37c4683b582 : FAILURE in 22m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28f2d305f7024753952495fee001af27 : SUCCESS in 1h 17m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/743141e6a89c453ab9a72d3ef76bca43 : SUCCESS in 4m 46s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/1430880e8a6d46c2963de132955cda3c : FAILURE in 50m 16s","accounts_in_message":[],"_revision_number":8},{"id":"c7027dcebd1dfbeb201bf0926284d227ddb38fa1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-21 11:06:15.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"f80e71214a08de2454d2a484e9bbde20f83609d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 13:07:12.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ff8d0037b78243349de91955624110cc : SUCCESS in 6m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8fbcb3d08e7b4d048ec615c794d90d88 : SUCCESS in 1h 15m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7486ceffe49c45b39c230b9d541cf8f4 : SUCCESS in 1h 52m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f576602a617241b7971f92e6931f6bb9 : SUCCESS in 1h 23m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0f6c0b38755493aa482351a888b79fa : SUCCESS in 1h 09m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1007120b71b342ca9d89eb32326c12b3 : SUCCESS in 1h 09m 11s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/673741ac21d842169d3c57e1c975eb69 : FAILURE in 53m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6459c46f44c545f78f7bf0b6db22880a : SUCCESS in 55m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512c10829b98439395e52ac5d469454a : SUCCESS in 4m 03s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/451a866d23a743aba9f9cf20da8b7242 : FAILURE in 41m 29s","accounts_in_message":[],"_revision_number":8},{"id":"de921bb8825d74880e1a20105cf1e9cebebd539f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-30 10:30:37.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"84829988be7957ba4870f549f8dc90161ff11503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 12:14:31.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/49f68c3c4a8f43c2a9da19c99800e439 : SUCCESS in 4m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/682280c3acce4b4f914c44bb3a27f418 : SUCCESS in 1h 13m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cffe3ec6c6f64394a5d8abce06849384 : SUCCESS in 1h 38m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4807ca0389e947238b4952fd0e87830d : SUCCESS in 1h 18m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/023ab2c27df843a3a7111e0e174de0ea : SUCCESS in 1h 07m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f42abc38c22b4aa0b613c20e4e21123b : SUCCESS in 1h 13m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/76e295c83c1d42229265f87562552e25 : SUCCESS in 1h 11m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29ea0943761447098240b6b872bd97e0 : SUCCESS in 59m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d59abf31c84917a133338a30ea0feb : SUCCESS in 4m 53s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/80ad73acabb54c39a7d5f151bc5b275d : FAILURE in 48m 18s","accounts_in_message":[],"_revision_number":8},{"id":"5720ca2e3087da9d6709abe36b6cb79ae6367235","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-07 17:37:13.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"533e9f0961419bbcde3002c5de83ded870239ffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 19:46:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a19def275aee4786b63bdf438ac2469e : SUCCESS in 7m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/627a123c9aa9485cadfebe8333d6cc6b : SUCCESS in 1h 14m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2f7268014b4546399faf41055636dd7b : SUCCESS in 1h 24m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c1c0086e927e42ab9fb378b36e067645 : SUCCESS in 1h 50m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60f361cb3cd545c3b70ef090b4344246 : SUCCESS in 1h 22m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d345081c1ba64cc5a80850c8e8bd5873 : SUCCESS in 1h 00m 27s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/99859df231d64702a19d78ec167beb1f : FAILURE in 32m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51168c8e70864c28b89ef849d43c70cc : SUCCESS in 58m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7b05f09daf44499abec05647ba691d : SUCCESS in 3m 38s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/66b207d16be64740abd56f996120722f : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":8},{"id":"d105183f5ea3ae2bc2755429231c5d1b5723779f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-07 20:21:22.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"bf569e662fc759f48e8ca4575bdf14adf45e41ac","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-10-09 14:09:21.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"7b38ce99105cc3d622c6c273d7cf02cd75f46933","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 14:09:33.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2eeb94a647e199b6a4192975fce368a3dae43d13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:46:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"00361b336b2a243c2ad7c69355c33a7e7f26c4da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:46:17.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6c4dc32ce7f54440ba7bbe441e997468 : SUCCESS in 6m 14s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e2eeebd6ff9a432e8843f1189e20e227 : SUCCESS in 1h 26m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/71003191800e4bb384c493b60ced932b : SUCCESS in 1h 31m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1c743d60c5a849d2a2c3d898b207ce89 : SUCCESS in 1h 16m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fb9a5603e3d461aa000720c2ce3c7ac : SUCCESS in 1h 00m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ba56c171dab9407b953ca798e1691a01 : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d862eff2fbac4c7b9d2ebdb5b45ea6b4 : SUCCESS in 51m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc77793e5e8040c884b16a7b4f0a1c22 : SUCCESS in 3m 09s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/45286f7eb69547f6b983745e5b0d9e9e : SUCCESS in 42m 58s","accounts_in_message":[],"_revision_number":8},{"id":"fe55e7adccbbca7acc4a8964e96cc23266462de8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:47:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a45f43d0f94475a8e553d98d2078923 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"c27d0123ab8c70dbb558f2378f181dc52c0fb534","revisions":{"3ff9b0a121ab6f0144f40a01e0ed68d9746be5ac":{"kind":"REWORK","_number":1,"created":"2018-10-29 13:14:52.000000000","uploader":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"ref":"refs/changes/24/613924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/613924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/1"}}},"commit":{"parents":[{"commit":"d28c0dced7adbb8b1d2a2f01ca9163350b358591","subject":"Merge \"Add cron as a required package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d28c0dced7adbb8b1d2a2f01ca9163350b358591"}]}],"author":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-29 13:13:54.000000000","tz":120},"committer":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-29 13:14:50.000000000","tz":120},"subject":"[WIP]Update permissions and path of NFS folder","message":"[WIP]Update permissions and path of NFS folder\n\nAfter running of glance role, the owner and group of NFS local folder\nis root instead of glance.\nMoreover, the location that the image is going to be registered is\nunder to \u0027v2/images\u0027.\n\nThe ownership of the local NFS folder is updated and the folder of\nthe \u0027images\u0027 has been moved under folder \u0027v2\u0027.\n\nChange-Id: I7403ac9bd85ea3ed149e13cb57c51039602f6ba1\nSigned-off-by: Panagiotis Karalis \u003cpkaralis@intracom-telecom.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3ff9b0a121ab6f0144f40a01e0ed68d9746be5ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3ff9b0a121ab6f0144f40a01e0ed68d9746be5ac"}]},"branch":"refs/heads/master"},"2840cd92ba3aeafd8a80d066bf2fef0895cf20e2":{"kind":"REWORK","_number":2,"created":"2018-10-30 07:21:46.000000000","uploader":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"ref":"refs/changes/24/613924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/613924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/2"}}},"commit":{"parents":[{"commit":"d28c0dced7adbb8b1d2a2f01ca9163350b358591","subject":"Merge \"Add cron as a required package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d28c0dced7adbb8b1d2a2f01ca9163350b358591"}]}],"author":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-29 13:13:54.000000000","tz":120},"committer":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-30 07:21:25.000000000","tz":120},"subject":"[WIP]Update permissions and path of NFS folder","message":"[WIP]Update permissions and path of NFS folder\n\nAfter running of glance role, the owner and group of NFS local folder\nis root instead of glance.\nMoreover, the location that the image is going to be registered is\nunder to \u0027v2/images\u0027.\n\nThe ownership of the local NFS folder is updated and the folder of\nthe \u0027images\u0027 has been moved under folder \u0027v2\u0027.\n\nChange-Id: I7403ac9bd85ea3ed149e13cb57c51039602f6ba1\nSigned-off-by: Panagiotis Karalis \u003cpkaralis@intracom-telecom.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2840cd92ba3aeafd8a80d066bf2fef0895cf20e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2840cd92ba3aeafd8a80d066bf2fef0895cf20e2"}]},"branch":"refs/heads/master"},"23e20832015fb0c320cb97f18dd2a25455cbed72":{"kind":"REWORK","_number":3,"created":"2018-10-30 08:19:33.000000000","uploader":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"ref":"refs/changes/24/613924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/613924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/3"}}},"commit":{"parents":[{"commit":"d28c0dced7adbb8b1d2a2f01ca9163350b358591","subject":"Merge \"Add cron as a required package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d28c0dced7adbb8b1d2a2f01ca9163350b358591"}]}],"author":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-29 13:13:54.000000000","tz":120},"committer":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-30 08:19:09.000000000","tz":120},"subject":"[WIP]Update permissions and path of NFS folder","message":"[WIP]Update permissions and path of NFS folder\n\nAfter running of glance role, the owner and group of NFS local folder\nis root instead of glance.\nMoreover, the location that the image is going to be registered is\nunder to \u0027v2/images\u0027.\n\nThe ownership of the local NFS folder is updated and the folder of\nthe \u0027images\u0027 has been moved under folder \u0027v2\u0027.\n\nChange-Id: I7403ac9bd85ea3ed149e13cb57c51039602f6ba1\nSigned-off-by: Panagiotis Karalis \u003cpkaralis@intracom-telecom.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/23e20832015fb0c320cb97f18dd2a25455cbed72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/23e20832015fb0c320cb97f18dd2a25455cbed72"}]},"branch":"refs/heads/master"},"147ad78151c066f39653bcbb29675ecf4d2c416a":{"kind":"REWORK","_number":4,"created":"2018-10-30 10:05:05.000000000","uploader":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"ref":"refs/changes/24/613924/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/613924/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/4"}}},"commit":{"parents":[{"commit":"d28c0dced7adbb8b1d2a2f01ca9163350b358591","subject":"Merge \"Add cron as a required package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d28c0dced7adbb8b1d2a2f01ca9163350b358591"}]}],"author":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-29 13:13:54.000000000","tz":120},"committer":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-30 10:01:45.000000000","tz":120},"subject":"[WIP]Update permissions and path of NFS folder","message":"[WIP]Update permissions and path of NFS folder\n\nAfter running of glance role, the owner and group of NFS local folder\nis root instead of glance.\nMoreover, the location that the image is going to be registered is\nunder to \u0027v2/images\u0027.\n\nThe ownership of the local NFS folder is updated and the folder of\nthe \u0027images\u0027 has been moved under folder \u0027v2\u0027.\nThe solution keeps the backward compatibility aspects.\n\nChange-Id: I7403ac9bd85ea3ed149e13cb57c51039602f6ba1\nSigned-off-by: Panagiotis Karalis \u003cpkaralis@intracom-telecom.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/147ad78151c066f39653bcbb29675ecf4d2c416a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/147ad78151c066f39653bcbb29675ecf4d2c416a"}]},"branch":"refs/heads/master"},"ec0ae6244648d9305d60872156327b900d8e920e":{"kind":"REWORK","_number":5,"created":"2018-11-07 14:47:53.000000000","uploader":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"ref":"refs/changes/24/613924/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/613924/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/5"}}},"commit":{"parents":[{"commit":"d28c0dced7adbb8b1d2a2f01ca9163350b358591","subject":"Merge \"Add cron as a required package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d28c0dced7adbb8b1d2a2f01ca9163350b358591"}]}],"author":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-29 13:13:54.000000000","tz":120},"committer":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-11-07 14:45:17.000000000","tz":120},"subject":"Turn the filesystem directory to be configurable","message":"Turn the filesystem directory to be configurable\n\nThis patch has been started as solution of bug, but finally an\nalternative solution has been followed without the involving\nof this repo.\n\nTherefore, this path could be stand as an improvement for future use.\n\nMaking the filesystem directory configurable, we are able to store the\nimage in the different directory or a new path under\nglance_system_user_home, which new directory name or path could be\ndefined in the scenario as a new variable like other related\nvariables without any impact on glance role.\n\nChange-Id: I7403ac9bd85ea3ed149e13cb57c51039602f6ba1\nSigned-off-by: Panagiotis Karalis \u003cpkaralis@intracom-telecom.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ec0ae6244648d9305d60872156327b900d8e920e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ec0ae6244648d9305d60872156327b900d8e920e"}]},"branch":"refs/heads/master"},"d6dd9cb0c7915a52601246ac06f0442cd69de6f9":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2018-11-07 16:11:25.000000000","uploader":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"ref":"refs/changes/24/613924/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/613924/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/6"}}},"commit":{"parents":[{"commit":"d28c0dced7adbb8b1d2a2f01ca9163350b358591","subject":"Merge \"Add cron as a required package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d28c0dced7adbb8b1d2a2f01ca9163350b358591"}]}],"author":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-29 13:13:54.000000000","tz":120},"committer":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-11-07 16:07:43.000000000","tz":120},"subject":"Turn the filesystem directory to be configurable","message":"Turn the filesystem directory to be configurable\n\nThis patch had been prepared as solution of a bug, but finally an\nalternative solution has been followed without the involving\nof this repo.\n\nTherefore, this path could be stand as an improvement for future use.\n\nMaking the filesystem directory configurable, we are able to store the\nimage in the different directory or a new path under\nglance_system_user_home repo, which new directory name or path could be\ndefined in the scenario as a new variable, like other related\nvariables, without any impact on glance role.\n\nChange-Id: I7403ac9bd85ea3ed149e13cb57c51039602f6ba1\nSigned-off-by: Panagiotis Karalis \u003cpkaralis@intracom-telecom.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d6dd9cb0c7915a52601246ac06f0442cd69de6f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d6dd9cb0c7915a52601246ac06f0442cd69de6f9"}]},"branch":"refs/heads/master"},"cd9c0f77d4c20fc52999e09350501e95c7c18aa8":{"kind":"REWORK","_number":7,"created":"2018-11-09 08:48:00.000000000","uploader":{"_account_id":27749,"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","username":"pkaralis"},"ref":"refs/changes/24/613924/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/613924/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/7"}}},"commit":{"parents":[{"commit":"d28c0dced7adbb8b1d2a2f01ca9163350b358591","subject":"Merge \"Add cron as a required package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d28c0dced7adbb8b1d2a2f01ca9163350b358591"}]}],"author":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-29 13:13:54.000000000","tz":120},"committer":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-11-09 08:34:00.000000000","tz":120},"subject":"Turn the filesystem directory to be configurable","message":"Turn the filesystem directory to be configurable\n\nUpdate the ownership of the directory about NFS mount point(s).\n\nThis patch could be also stand as an improvement for future use.\n\nMaking the filesystem directory configurable, we are able to store\nthe image in the different directory (or in a new path) under\nglance_system_user_home repo, which is able to be configured\ndynamically, for instance, via deployment of a scenario.\n\nChange-Id: I7403ac9bd85ea3ed149e13cb57c51039602f6ba1\nSigned-off-by: Panagiotis Karalis \u003cpkaralis@intracom-telecom.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cd9c0f77d4c20fc52999e09350501e95c7c18aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cd9c0f77d4c20fc52999e09350501e95c7c18aa8"}]},"branch":"refs/heads/master"},"c27d0123ab8c70dbb558f2378f181dc52c0fb534":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-09-19 16:45:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/613924/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/613924/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/613924/8"}}},"commit":{"parents":[{"commit":"ae681ded4e6bc033c75f72b7a429a5b04dd0a676","subject":"Merge \"Allow venv python interpreter to be overridden\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ae681ded4e6bc033c75f72b7a429a5b04dd0a676"}]}],"author":{"name":"Panagiotis Karalis","email":"pkaralis@intracom-telecom.com","date":"2018-10-29 13:13:54.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-19 16:45:04.000000000","tz":180},"subject":"Turn the filesystem directory to be configurable","message":"Turn the filesystem directory to be configurable\n\nUpdate the ownership of the directory about NFS mount point(s).\n\nThis patch could be also stand as an improvement for future use.\n\nMaking the filesystem directory configurable, we are able to store\nthe image in the different directory (or in a new path) under\nglance_system_user_home repo, which is able to be configured\ndynamically, for instance, via deployment of a scenario.\n\nChange-Id: I7403ac9bd85ea3ed149e13cb57c51039602f6ba1\nSigned-off-by: Panagiotis Karalis \u003cpkaralis@intracom-telecom.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c27d0123ab8c70dbb558f2378f181dc52c0fb534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c27d0123ab8c70dbb558f2378f181dc52c0fb534"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
