)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~614102","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I9109e8a2deddd0b59628d733227a9cb067301239","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"I9109e8a2deddd0b59628d733227a9cb067301239","subject":"missing document start \"---\"","status":"MERGED","created":"2018-10-30 07:38:30.000000000","updated":"2019-04-24 19:54:52.000000000","submitted":"2019-04-24 19:54:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"614102-1556135692977-dfe5b5ea","meta_rev_id":"0d9f6086cde6a222508258447c0a395a225a5389","_number":614102,"virtual_id_number":614102,"owner":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-24 19:54:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-11-07 10:22:38.000000000","_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-04-24 18:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-04-24 18:29:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 18:29:32.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-24 19:54:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c518d609b25d922ee14cc1b7de14bd59162d9d7","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-10-30 07:38:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3af6ae0d9edfdfe3f9d8183a50d97162a003b01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-30 08:30:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-centos-7/6919c51/ : FAILURE in 30m 47s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-opensuse-150/919b2f4/ : SUCCESS in 30m 05s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-opensuse-423/49e23c0/ : SUCCESS in 30m 17s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/117d7dd/ : FAILURE in 27m 46s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/02/614102/1/check/openstack-ansible-linters/993c97c/ : SUCCESS in 5m 11s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-centos-7/ef547a6/ : SUCCESS in 34m 08s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-opensuse-150/bfab87c/ : SUCCESS in 41m 42s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-opensuse-423/6fcf412/ : SUCCESS in 29m 59s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-ubuntu-bionic/d146459/ : SUCCESS in 47m 45s\n- openstack-tox-docs http://logs.openstack.org/02/614102/1/check/openstack-tox-docs/205f9bc/html/ : SUCCESS in 2m 55s\n- openstack-ansible-nfs_glance http://logs.openstack.org/02/614102/1/check/openstack-ansible-nfs_glance/795c7ab/ : SUCCESS in 36m 34s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/02/614102/1/check/openstack-ansible-v2_registry_enabled/c9ef772/ : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":1},{"id":"84bf275d43bf17d45d001ae7af13567385f16e42","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-11-07 10:22:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6c7d0069fe7f41bcb42fbacf67ff48a653f1be41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 11:21:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-centos-7/de51db6/ : SUCCESS in 35m 05s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-opensuse-150/5891da1/ : SUCCESS in 33m 04s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-opensuse-423/cfa3e27/ : SUCCESS in 29m 09s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/b650f0a/ : SUCCESS in 26m 36s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/02/614102/1/check/openstack-ansible-linters/a331324/ : SUCCESS in 7m 01s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-centos-7/b299437/ : SUCCESS in 36m 16s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-opensuse-150/8b99a71/ : SUCCESS in 50m 58s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-opensuse-423/79b2dcf/ : SUCCESS in 34m 18s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-ubuntu-bionic/01d37cc/ : SUCCESS in 41m 29s\n- openstack-tox-docs http://logs.openstack.org/02/614102/1/check/openstack-tox-docs/99f9054/html/ : SUCCESS in 3m 51s\n- openstack-ansible-nfs_glance http://logs.openstack.org/02/614102/1/check/openstack-ansible-nfs_glance/861a2eb/ : SUCCESS in 42m 55s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/02/614102/1/check/openstack-ansible-v2_registry_enabled/1eb2db6/ : SUCCESS in 34m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0bb08d49f33b0b5d4e5a315796d10e85b1ac6e7b","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-11-08 03:55:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"2f370f36aea7315988a15ea2fabb0d53e03a8dfa","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-12-10 23:04:06.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"cdc9811cc5e872c01c68a3d7b7845dc388a92a33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-11 00:10:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-centos-7/f063b15/ : SUCCESS in 40m 28s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-opensuse-150/37e6836/ : FAILURE in 20m 20s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-opensuse-423/8b09845/ : FAILURE in 18m 57s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/7b1858e/ : SUCCESS in 50m 52s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/02/614102/1/check/openstack-ansible-linters/c7b2c81/ : SUCCESS in 4m 28s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-centos-7/96a94df/ : SUCCESS in 40m 39s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-opensuse-150/ec61cb0/ : SUCCESS in 41m 01s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-opensuse-423/e23000f/ : SUCCESS in 42m 05s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/02/614102/1/check/openstack-ansible-functional-ubuntu-bionic/c1d4577/ : SUCCESS in 52m 19s\n- openstack-tox-docs http://logs.openstack.org/02/614102/1/check/openstack-tox-docs/f0efc6f/html/ : SUCCESS in 2m 35s\n- openstack-ansible-nfs_glance http://logs.openstack.org/02/614102/1/check/openstack-ansible-nfs_glance/382df1a/ : SUCCESS in 57m 02s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/02/614102/1/check/openstack-ansible-v2_registry_enabled/10569aa/ : SUCCESS in 38m 24s","accounts_in_message":[],"_revision_number":1},{"id":"eda00a75c5e9aee86a3b43c75386a7762e0c4b81","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 18:29:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d30fca452de9bfea087d74879a2b202d9b72c70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:29:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98c26564da851e1d57e3617a8217173088d14b20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:54:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0d9f6086cde6a222508258447c0a395a225a5389","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:54:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/02/614102/1/gate/openstack-ansible-linters/b883199/ : SUCCESS in 5m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/02/614102/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/672536f/ : SUCCESS in 1h 22m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/02/614102/1/gate/openstack-ansible-deploy-aio_metal-centos-7/58cb87f/ : SUCCESS in 1h 16m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/02/614102/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/317a0b2/ : SUCCESS in 1h 10m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/02/614102/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/082eeba/ : SUCCESS in 59m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/02/614102/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/7cfee4f/ : SUCCESS in 53m 18s\n- openstack-tox-docs http://logs.openstack.org/02/614102/1/gate/openstack-tox-docs/9cb6e12/html/ : SUCCESS in 5m 25s\n- openstack-ansible-nfs_glance http://logs.openstack.org/02/614102/1/gate/openstack-ansible-nfs_glance/743dd12/ : SUCCESS in 44m 37s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/02/614102/1/gate/openstack-ansible-v2_registry_enabled/15f8721/ : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b1d43bdf6d5582e2b5dee5346ae7d9f51756cc3","revisions":{"6b1d43bdf6d5582e2b5dee5346ae7d9f51756cc3":{"kind":"REWORK","_number":1,"created":"2018-10-30 07:38:30.000000000","uploader":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"ref":"refs/changes/02/614102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/02/614102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/02/614102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/02/614102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/02/614102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/02/614102/1"}}},"commit":{"parents":[{"commit":"d28c0dced7adbb8b1d2a2f01ca9163350b358591","subject":"Merge \"Add cron as a required package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d28c0dced7adbb8b1d2a2f01ca9163350b358591"}]}],"author":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-10-30 07:38:12.000000000","tz":0},"committer":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-10-30 07:38:28.000000000","tz":0},"subject":"missing document start \"---\"","message":"missing document start \"---\"\n\nChange-Id: I9109e8a2deddd0b59628d733227a9cb067301239\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6b1d43bdf6d5582e2b5dee5346ae7d9f51756cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6b1d43bdf6d5582e2b5dee5346ae7d9f51756cc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
