)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~654357","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Id327f8e91c299bdef40384f931befb9d1f2ef76d","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Id327f8e91c299bdef40384f931befb9d1f2ef76d","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 07:14:10.000000000","updated":"2019-04-22 12:38:25.000000000","submitted":"2019-04-22 12:38:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654357-1555936705855-61dee02a","meta_rev_id":"1bf18b1858d01a853328143148dd0cb4ac0efe6d","_number":654357,"virtual_id_number":654357,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-22 12:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-22 10:54:41.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-22 10:54:41.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 10:54:41.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-22 12:38:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5598b86800456899a359cff28c0a999117673f09","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:14:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e017768ce76cb8300912dba42236d180c52f2e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:40:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/57/654357/1/check/openstack-ansible-linters/0629cbf/ : SUCCESS in 5m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/57/654357/1/check/openstack-ansible-deploy-aio_metal-debian-stable/f47aa4d/ : SUCCESS in 58m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/57/654357/1/check/openstack-ansible-deploy-aio_metal-centos-7/fe04271/ : SUCCESS in 1h 13m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/57/654357/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/4617edd/ : SUCCESS in 1h 03m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/57/654357/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3eda9e0/ : SUCCESS in 59m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/57/654357/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/f519751/ : SUCCESS in 58m 22s\n- openstack-tox-docs http://logs.openstack.org/57/654357/1/check/openstack-tox-docs/928b017/html/ : SUCCESS in 2m 38s\n- openstack-ansible-nfs_glance http://logs.openstack.org/57/654357/1/check/openstack-ansible-nfs_glance/1df92e4/ : SUCCESS in 33m 42s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/57/654357/1/check/openstack-ansible-v2_registry_enabled/726c260/ : SUCCESS in 28m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a17779c5bf22a0af323a3cada73d052d15765daf","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 10:54:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fdb43c5ecb9f2eca0fa3c5ccdb2f570f57028158","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 10:54:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6c7a802ebd0111dccd60ad0611ad09f8c587324","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:38:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5b5fd8457cd4b2927c2b0e8d1edf0ccb4fe424d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:38:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/57/654357/1/gate/openstack-ansible-linters/a2f6187/ : SUCCESS in 4m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/57/654357/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/93b1975/ : SUCCESS in 1h 04m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/57/654357/1/gate/openstack-ansible-deploy-aio_metal-centos-7/e0a9cd5/ : SUCCESS in 1h 19m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/57/654357/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/aac9554/ : SUCCESS in 1h 05m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/57/654357/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/bddce06/ : SUCCESS in 55m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/57/654357/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/bb9aff9/ : SUCCESS in 53m 32s\n- openstack-tox-docs http://logs.openstack.org/57/654357/1/gate/openstack-tox-docs/40d48e7/html/ : SUCCESS in 2m 58s\n- openstack-ansible-nfs_glance http://logs.openstack.org/57/654357/1/gate/openstack-ansible-nfs_glance/1cc88c8/ : SUCCESS in 34m 14s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/57/654357/1/gate/openstack-ansible-v2_registry_enabled/cca6efe/ : SUCCESS in 31m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60bd033fe7e71442e8e6d1178083cabdea4aab75","revisions":{"60bd033fe7e71442e8e6d1178083cabdea4aab75":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:14:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/654357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/57/654357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/57/654357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/57/654357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/57/654357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/57/654357/1"}}},"commit":{"parents":[{"commit":"cc9822485e8764dfaa61161bb5a7b1da621d15bf","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cc9822485e8764dfaa61161bb5a7b1da621d15bf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:14:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:14:08.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Id327f8e91c299bdef40384f931befb9d1f2ef76d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/60bd033fe7e71442e8e6d1178083cabdea4aab75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/60bd033fe7e71442e8e6d1178083cabdea4aab75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
