)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~654511","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fstein~Iddd3e755a4dc0e135ed8d9427e23f1a140eb5f46","project":"openstack/openstack-ansible-os_glance","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Iddd3e755a4dc0e135ed8d9427e23f1a140eb5f46","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 15:38:10.000000000","updated":"2019-04-22 22:16:58.000000000","submitted":"2019-04-22 22:16:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654511-1555971418686-332a2ac9","meta_rev_id":"bd62a60fbad899dc93316ec907942b0efe777f5a","_number":654511,"virtual_id_number":654511,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-22 22:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-22 16:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-22 16:10:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 16:10:37.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-22 22:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"755519720e57cc1f8729714fbe8f378501de6d36","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 15:38:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78c42a4502b419292eac3ddb5d1cbb9e54cc5fd7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:10:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ffbc3d77ec3fd8516479b86678db3d2277912a42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:44:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/11/654511/1/check/openstack-ansible-linters/4fdfd2d/ : SUCCESS in 5m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/11/654511/1/check/openstack-ansible-deploy-aio_metal-debian-stable/04922db/ : FAILURE in 23m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/11/654511/1/check/openstack-ansible-deploy-aio_metal-centos-7/9f14235/ : FAILURE in 30m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/11/654511/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3edb7ae/ : FAILURE in 21m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/11/654511/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/f4157e8/ : FAILURE in 23m 43s\n- openstack-tox-docs http://logs.openstack.org/11/654511/1/check/openstack-tox-docs/f4f4ffe/html/ : SUCCESS in 2m 49s\n- openstack-ansible-nfs_glance http://logs.openstack.org/11/654511/1/check/openstack-ansible-nfs_glance/9bd4343/ : SUCCESS in 34m 02s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/11/654511/1/check/openstack-ansible-v2_registry_enabled/f8c9b42/ : SUCCESS in 29m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5f514d6859a1f64500ebafcfe9dd9771628e9cb2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 19:12:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f551d30cc07a5b34a30d4d123101c3bdc11728e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:49:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/11/654511/1/check/openstack-ansible-linters/10e3053/ : SUCCESS in 4m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/11/654511/1/check/openstack-ansible-deploy-aio_metal-debian-stable/4a7f4b2/ : SUCCESS in 52m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/11/654511/1/check/openstack-ansible-deploy-aio_metal-centos-7/a75dd56/ : SUCCESS in 1h 12m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/11/654511/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c3eca8c/ : SUCCESS in 1h 00m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/11/654511/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e0104c0/ : SUCCESS in 56m 16s\n- openstack-tox-docs http://logs.openstack.org/11/654511/1/check/openstack-tox-docs/b518bb5/html/ : SUCCESS in 3m 02s\n- openstack-ansible-nfs_glance http://logs.openstack.org/11/654511/1/check/openstack-ansible-nfs_glance/b9f6ac5/ : SUCCESS in 57m 13s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/11/654511/1/check/openstack-ansible-v2_registry_enabled/62bd389/ : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e2d3fe1d6eaa578e842ab9bf7bf6ec5a0f5afdba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:50:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27bd8c11813cfdfc47210df1017d497d83f9a811","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:16:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/11/654511/1/gate/openstack-ansible-linters/ee267b3/ : SUCCESS in 5m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/11/654511/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/23e3a7d/ : SUCCESS in 1h 01m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/11/654511/1/gate/openstack-ansible-deploy-aio_metal-centos-7/6a00909/ : SUCCESS in 1h 18m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/11/654511/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8ab0554/ : SUCCESS in 57m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/11/654511/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/cb7f3fe/ : SUCCESS in 52m 26s\n- openstack-tox-docs http://logs.openstack.org/11/654511/1/gate/openstack-tox-docs/ceeeb93/html/ : SUCCESS in 2m 43s\n- openstack-ansible-nfs_glance http://logs.openstack.org/11/654511/1/gate/openstack-ansible-nfs_glance/89035a1/ : SUCCESS in 34m 02s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/11/654511/1/gate/openstack-ansible-v2_registry_enabled/8ffe7a7/ : SUCCESS in 28m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d9587f1b228e4d720dd1fead38c75c06313079f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:16:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f48e5132b58c9bab1adcd2a43907c8d231af8126","revisions":{"f48e5132b58c9bab1adcd2a43907c8d231af8126":{"kind":"REWORK","_number":1,"created":"2019-04-22 15:38:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/654511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/11/654511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/11/654511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/11/654511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/11/654511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/11/654511/1"}}},"commit":{"parents":[{"commit":"fe104b64c13122e22bec9358324212ec281e4ca0","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/fe104b64c13122e22bec9358324212ec281e4ca0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:38:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:38:09.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iddd3e755a4dc0e135ed8d9427e23f1a140eb5f46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f48e5132b58c9bab1adcd2a43907c8d231af8126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f48e5132b58c9bab1adcd2a43907c8d231af8126"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
