)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~654906","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fstein~I915d36cb65235bec363645f1ebc9c7af502222fa","project":"openstack/openstack-ansible-os_glance","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I915d36cb65235bec363645f1ebc9c7af502222fa","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-23 09:23:59.000000000","updated":"2019-04-24 09:05:51.000000000","submitted":"2019-04-24 09:05:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654906-1556096751215-b16e9fe7","meta_rev_id":"4f71ade2f3707af2a17408c8886becfcebd48b64","_number":654906,"virtual_id_number":654906,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-24 09:05:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-23 20:22:42.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-23 10:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-23 10:50:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 10:50:19.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-23 20:22:42.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-04-24 09:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63c99abe5ebb829c4799c8fd3b5fd694816f5c2b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-23 09:23:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"530aa3c7e3fb8a958b50329547e081f419b986dc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-23 10:50:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54fda5296156eae2c3cb11883cdc309e8d17a73a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:52:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/06/654906/1/check/openstack-ansible-linters/0803b6b/ : SUCCESS in 6m 43s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_metal-debian-stable/ca47585/ : SUCCESS in 56m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_metal-centos-7/88fbad3/ : SUCCESS in 1h 18m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b4e037e/ : SUCCESS in 1h 06m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/efb1233/ : SUCCESS in 56m 42s\n- openstack-tox-docs http://logs.openstack.org/06/654906/1/check/openstack-tox-docs/224c79b/html/ : SUCCESS in 2m 31s\n- openstack-ansible-nfs_glance http://logs.openstack.org/06/654906/1/check/openstack-ansible-nfs_glance/94342c9/ : SUCCESS in 34m 34s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/06/654906/1/check/openstack-ansible-v2_registry_enabled/eea6efa/ : SUCCESS in 30m 17s","accounts_in_message":[],"_revision_number":1},{"id":"eec67b078b328b2f970eca95c504ce8a8a84deed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:52:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f1a8e11748ef79df66452a98834e8773c99e885","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:11:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/06/654906/1/gate/openstack-ansible-linters/0fb0984/ : SUCCESS in 5m 44s\n- openstack-ansible-deploy-aio_metal-debian-stable finger://ze06.openstack.org/e644aee04b744ee9b5bfbcfa1847e020 : RETRY_LIMIT in 42m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 finger://ze12.openstack.org/b303be075e864c889a47104aaf360e87 : RETRY_LIMIT in 35m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic finger://ze06.openstack.org/1f96eac89ed84166870c623f3263620f : RETRY_LIMIT in 44m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 finger://ze08.openstack.org/3f0d9dad2e9842d5a6a2cbde7be378ae : RETRY_LIMIT in 42m 37s\n- openstack-tox-docs http://logs.openstack.org/06/654906/1/gate/openstack-tox-docs/66b1f57/html/ : SUCCESS in 2m 42s\n- openstack-ansible-nfs_glance finger://ze11.openstack.org/a6a5edc4538d4fdfae080913a8a35730 : RETRY_LIMIT in 22m 23s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/06/654906/1/gate/openstack-ansible-v2_registry_enabled/7f8a258/ : RETRY_LIMIT in 37m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e741e44c46128e9feac01c856eb7f59f67de54b9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-23 20:22:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bca8387dc9e99302832fdff52c27dee573e19815","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 22:04:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/06/654906/1/check/openstack-ansible-linters/8593517/ : SUCCESS in 5m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_metal-debian-stable/33ff637/ : SUCCESS in 59m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_metal-centos-7/f486dcf/ : POST_FAILURE in 16m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/21365ec/ : SUCCESS in 56m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/ebd37cb/ : SUCCESS in 54m 07s\n- openstack-tox-docs http://logs.openstack.org/06/654906/1/check/openstack-tox-docs/002212c/html/ : SUCCESS in 2m 38s\n- openstack-ansible-nfs_glance http://logs.openstack.org/06/654906/1/check/openstack-ansible-nfs_glance/80405f3/ : SUCCESS in 34m 53s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/06/654906/1/check/openstack-ansible-v2_registry_enabled/192c0cd/ : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5ba91b50dcacbe169543546ceb65aad4d3b238df","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-24 05:57:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77d1b87c9a8347fac150f7b428e7f8045bbbfed3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 07:31:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/06/654906/1/check/openstack-ansible-linters/e72d4b8/ : SUCCESS in 6m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_metal-debian-stable/afa906a/ : SUCCESS in 56m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_metal-centos-7/47378df/ : SUCCESS in 1h 18m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e560d09/ : SUCCESS in 56m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/654906/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/d3fe5bd/ : SUCCESS in 53m 25s\n- openstack-tox-docs http://logs.openstack.org/06/654906/1/check/openstack-tox-docs/a8951b1/html/ : SUCCESS in 2m 44s\n- openstack-ansible-nfs_glance http://logs.openstack.org/06/654906/1/check/openstack-ansible-nfs_glance/49d1a44/ : SUCCESS in 34m 37s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/06/654906/1/check/openstack-ansible-v2_registry_enabled/bf2355f/ : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":1},{"id":"43fe0bc9d76b1d9434f3f6326723c395e3eb695c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 07:31:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e49830cf3ca6eb87887dfd49b4740f08a3a20b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 09:05:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8abec7c56bf24ff5e81374a9c9fc6b5d5c5df588","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 09:05:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/06/654906/1/gate/openstack-ansible-linters/eceb565/ : SUCCESS in 4m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/654906/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/5b82d5c/ : SUCCESS in 53m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/654906/1/gate/openstack-ansible-deploy-aio_metal-centos-7/e5f8e2b/ : SUCCESS in 1h 17m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/654906/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2417e21/ : SUCCESS in 1h 00m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/654906/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/76b7a6f/ : SUCCESS in 52m 36s\n- openstack-tox-docs http://logs.openstack.org/06/654906/1/gate/openstack-tox-docs/f53479f/html/ : SUCCESS in 2m 48s\n- openstack-ansible-nfs_glance http://logs.openstack.org/06/654906/1/gate/openstack-ansible-nfs_glance/827121f/ : SUCCESS in 34m 43s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/06/654906/1/gate/openstack-ansible-v2_registry_enabled/226d880/ : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2d4635b243f3c036596659fcd0d3a45418efbf1","revisions":{"a2d4635b243f3c036596659fcd0d3a45418efbf1":{"kind":"REWORK","_number":1,"created":"2019-04-23 09:23:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/654906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/06/654906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/654906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/654906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/654906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/654906/1"}}},"commit":{"parents":[{"commit":"f48e5132b58c9bab1adcd2a43907c8d231af8126","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f48e5132b58c9bab1adcd2a43907c8d231af8126"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:23:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:23:58.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I915d36cb65235bec363645f1ebc9c7af502222fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a2d4635b243f3c036596659fcd0d3a45418efbf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a2d4635b243f3c036596659fcd0d3a45418efbf1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
