)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~658020","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I9acc37ec1b8d14426b1103818cd64cbb687d7140","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I9acc37ec1b8d14426b1103818cd64cbb687d7140","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-09 11:34:31.000000000","updated":"2019-05-09 14:35:37.000000000","submitted":"2019-05-09 14:35:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658020-1557412537028-121c7fd4","meta_rev_id":"0c6be137510485e360565316abe97a9b063e440e","_number":658020,"virtual_id_number":658020,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-09 14:35:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-09 12:59:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-09 12:59:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:59:34.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-09 14:35:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a13f83aa440601f2e718020468aa091666372fa","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 11:34:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76265d74fd9b5f422599e58f2acd4374674071f3","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-09 12:59:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4cfcf5777dd1bff61f275d0471f071a6d58687a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:12:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/20/658020/1/check/openstack-ansible-linters/1c8d2df/ : SUCCESS in 5m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/20/658020/1/check/openstack-ansible-deploy-aio_metal-debian-stable/6c9125f/ : SUCCESS in 1h 03m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/20/658020/1/check/openstack-ansible-deploy-aio_metal-centos-7/9d3b5b0/ : SUCCESS in 1h 22m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/20/658020/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/4f99976/ : SUCCESS in 1h 11m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/20/658020/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/714b67f/ : SUCCESS in 52m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/20/658020/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/85da403/ : FAILURE in 38m 01s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/20/658020/1/check/openstack-tox-docs/46b3afd/html/ : SUCCESS in 2m 10s\n- build-openstack-releasenotes http://logs.openstack.org/20/658020/1/check/build-openstack-releasenotes/286f3d8/html/ : SUCCESS in 2m 06s\n- openstack-ansible-nfs_glance http://logs.openstack.org/20/658020/1/check/openstack-ansible-nfs_glance/81d8c4e/ : SUCCESS in 34m 23s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/20/658020/1/check/openstack-ansible-v2_registry_enabled/f3ac367/ : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e99be6227557eb8d228bb62ad26cfe8032dc1878","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:12:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67fae759ea395f27d0824dc99664d5ffe856f94d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 14:35:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/20/658020/1/gate/openstack-ansible-linters/3279c0d/ : SUCCESS in 5m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/20/658020/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/14a1cf5/ : SUCCESS in 1h 01m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/20/658020/1/gate/openstack-ansible-deploy-aio_metal-centos-7/c1feadf/ : SUCCESS in 1h 16m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/20/658020/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/48d688c/ : SUCCESS in 1h 07m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/20/658020/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b6a1d26/ : SUCCESS in 57m 36s\n- openstack-tox-docs http://logs.openstack.org/20/658020/1/gate/openstack-tox-docs/6ffade9/html/ : SUCCESS in 2m 36s\n- build-openstack-releasenotes http://logs.openstack.org/20/658020/1/gate/build-openstack-releasenotes/6701295/html/ : SUCCESS in 2m 23s\n- openstack-ansible-nfs_glance http://logs.openstack.org/20/658020/1/gate/openstack-ansible-nfs_glance/f824328/ : SUCCESS in 32m 15s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/20/658020/1/gate/openstack-ansible-v2_registry_enabled/1499f78/ : SUCCESS in 29m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d9cd93c54877ce83555157bc368f61157aa8c5d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 14:35:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3420ea814294b0d7b44e6acf9f326a1afa6ee0bc","revisions":{"3420ea814294b0d7b44e6acf9f326a1afa6ee0bc":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:34:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/20/658020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/20/658020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/658020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/658020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/658020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/658020/1"}}},"commit":{"parents":[{"commit":"81b529140b3f01bcb3dcc639a9071e7c260cdb86","subject":"Merge \"missing document start \"---\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/81b529140b3f01bcb3dcc639a9071e7c260cdb86"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:34:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:34:30.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I9acc37ec1b8d14426b1103818cd64cbb687d7140\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3420ea814294b0d7b44e6acf9f326a1afa6ee0bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3420ea814294b0d7b44e6acf9f326a1afa6ee0bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
