)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~658594","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fstein~I5c0d0cf5675b86421fe65d8c46c50ca4893b67ff","project":"openstack/openstack-ansible-os_glance","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I5c0d0cf5675b86421fe65d8c46c50ca4893b67ff","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:58:38.000000000","updated":"2019-05-12 03:24:21.000000000","submitted":"2019-05-12 03:24:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658594-1557631461072-7f1a7d1e","meta_rev_id":"d7afa9297a2658149d361ced5730405ab100aeb9","_number":658594,"virtual_id_number":658594,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 03:24:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 01:55:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-12 01:55:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 01:55:28.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 03:24:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d61e6957abaac982e07b2fdbfa36e453b6c368a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:58:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69ad6e164c873be64145d702a4f77257a38c3b9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:26:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/94/658594/1/check/openstack-ansible-linters/4d131bc/ : SUCCESS in 4m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/94/658594/1/check/openstack-ansible-deploy-aio_metal-debian-stable/92f2fcd/ : SUCCESS in 55m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/94/658594/1/check/openstack-ansible-deploy-aio_metal-centos-7/f6a8c7c/ : SUCCESS in 1h 15m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/94/658594/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6b275fa/ : SUCCESS in 53m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/94/658594/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/3d7e6b1/ : FAILURE in 36m 13s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/94/658594/1/check/openstack-tox-docs/31bda4f/html/ : SUCCESS in 3m 29s\n- build-openstack-releasenotes http://logs.openstack.org/94/658594/1/check/build-openstack-releasenotes/b4ef2fe/html/ : SUCCESS in 3m 24s\n- openstack-ansible-nfs_glance http://logs.openstack.org/94/658594/1/check/openstack-ansible-nfs_glance/6c9d911/ : SUCCESS in 38m 02s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/94/658594/1/check/openstack-ansible-v2_registry_enabled/2ae030b/ : SUCCESS in 32m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c8a154b3089ae89ac33ce4feccc29a363d682e9e","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 01:55:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4cd3602014bb679ff7a2673d7ca235dc4c79c32a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:55:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ec6e4ec9fb2a2fb1d39751a024c3ebc9bfc68b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:24:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/94/658594/1/gate/openstack-ansible-linters/814de28/ : SUCCESS in 4m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/94/658594/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/c01fd1d/ : SUCCESS in 57m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/94/658594/1/gate/openstack-ansible-deploy-aio_metal-centos-7/1b19786/ : SUCCESS in 1h 16m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/94/658594/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/7d4d956/ : SUCCESS in 55m 22s\n- openstack-tox-docs http://logs.openstack.org/94/658594/1/gate/openstack-tox-docs/8f903ee/html/ : SUCCESS in 2m 01s\n- build-openstack-releasenotes http://logs.openstack.org/94/658594/1/gate/build-openstack-releasenotes/7435df4/html/ : SUCCESS in 2m 16s\n- openstack-ansible-nfs_glance http://logs.openstack.org/94/658594/1/gate/openstack-ansible-nfs_glance/27c7ef4/ : SUCCESS in 34m 41s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/94/658594/1/gate/openstack-ansible-v2_registry_enabled/e11dfcd/ : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":1},{"id":"039d6a8e8b833f6672f33e04c8be817bb1e6d373","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:24:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22ca0ec85d5fd4e35bd77f731ce2685bb82d8d29","revisions":{"22ca0ec85d5fd4e35bd77f731ce2685bb82d8d29":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:58:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/658594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/94/658594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/94/658594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/94/658594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/94/658594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/94/658594/1"}}},"commit":{"parents":[{"commit":"a2d4635b243f3c036596659fcd0d3a45418efbf1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a2d4635b243f3c036596659fcd0d3a45418efbf1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:58:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:58:37.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I5c0d0cf5675b86421fe65d8c46c50ca4893b67ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/22ca0ec85d5fd4e35bd77f731ce2685bb82d8d29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/22ca0ec85d5fd4e35bd77f731ce2685bb82d8d29"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
