)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~664088","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fqueens~I1229f7ea246dba173a416479baebc85f180b9f7f","project":"openstack/openstack-ansible-os_glance","branch":"stable/queens","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I1229f7ea246dba173a416479baebc85f180b9f7f","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-07 23:24:15.000000000","updated":"2019-06-11 23:43:57.000000000","submitted":"2019-06-11 23:43:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664088-1560296637922-4683f4c9","meta_rev_id":"d1de7e093d26ee857caedeec91179a348b32c7de","_number":664088,"virtual_id_number":664088,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-11 23:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-11 20:53:07.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-06-11 22:28:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-06-11 22:28:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 20:53:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-11 22:28:32.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-11 23:43:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d898f262aab9d57f69b95ad08cf0628a973f9c0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:24:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bf05bd26fc5738072cdffa31ba777ad4cee8ac3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:32:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/88/664088/1/check/build-openstack-sphinx-docs/73b6cb7/html/ : SUCCESS in 3m 46s\n- openstack-ansible-linters http://logs.openstack.org/88/664088/1/check/openstack-ansible-linters/d465bd6/ : SUCCESS in 6m 13s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/88/664088/1/check/openstack-ansible-functional-centos-7/8fa1ff7/ : SUCCESS in 54m 15s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/88/664088/1/check/openstack-ansible-functional-opensuse-423/75a4dc8/ : FAILURE in 9m 09s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/88/664088/1/check/openstack-ansible-functional-ubuntu-xenial/9de7ade/ : SUCCESS in 25m 46s\n- openstack-ansible-nfs_glance http://logs.openstack.org/88/664088/1/check/openstack-ansible-nfs_glance/c3dd61a/ : SUCCESS in 25m 50s\n- openstack-ansible-v1_api_enabled http://logs.openstack.org/88/664088/1/check/openstack-ansible-v1_api_enabled/2350b6a/ : SUCCESS in 28m 02s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/88/664088/1/check/openstack-ansible-v2_registry_enabled/c27d9b2/ : SUCCESS in 29m 49s","accounts_in_message":[],"_revision_number":1},{"id":"465871f7e631bf19fd0f0df744f9633ac5e06f96","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 20:53:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a92c2d8a658341c4dbcd3abf43342d0b7dd9d7b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:02:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/88/664088/1/check/build-openstack-sphinx-docs/3d00b8b/html/ : SUCCESS in 3m 32s\n- openstack-ansible-linters http://logs.openstack.org/88/664088/1/check/openstack-ansible-linters/2c72633/ : SUCCESS in 4m 52s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/88/664088/1/check/openstack-ansible-functional-centos-7/f28a722/ : SUCCESS in 52m 58s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/88/664088/1/check/openstack-ansible-functional-ubuntu-xenial/b9d9a24/ : SUCCESS in 25m 36s\n- openstack-ansible-nfs_glance http://logs.openstack.org/88/664088/1/check/openstack-ansible-nfs_glance/faed1c2/ : SUCCESS in 28m 01s\n- openstack-ansible-v1_api_enabled http://logs.openstack.org/88/664088/1/check/openstack-ansible-v1_api_enabled/8e15282/ : SUCCESS in 27m 39s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/88/664088/1/check/openstack-ansible-v2_registry_enabled/6ae961e/ : SUCCESS in 27m 31s","accounts_in_message":[],"_revision_number":1},{"id":"39e0c6d6f61404c7de7c65ebbb0138eba681473f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-11 22:28:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5947b3055e0b37c15ff47a2cb0fd7c83cb1542ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:28:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f5b651f9d921c1856f67a49a669239da56fd8c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 23:43:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/88/664088/1/gate/build-openstack-sphinx-docs/6ae8d03/html/ : SUCCESS in 4m 54s\n- openstack-ansible-linters http://logs.openstack.org/88/664088/1/gate/openstack-ansible-linters/8cad5f6/ : SUCCESS in 5m 52s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/88/664088/1/gate/openstack-ansible-functional-centos-7/7f3165a/ : SUCCESS in 1h 01m 56s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/88/664088/1/gate/openstack-ansible-functional-ubuntu-xenial/cecc512/ : SUCCESS in 25m 21s\n- openstack-ansible-nfs_glance http://logs.openstack.org/88/664088/1/gate/openstack-ansible-nfs_glance/ce21d03/ : SUCCESS in 26m 26s\n- openstack-ansible-v1_api_enabled http://logs.openstack.org/88/664088/1/gate/openstack-ansible-v1_api_enabled/61139c5/ : SUCCESS in 26m 28s\n- openstack-ansible-v2_registry_enabled http://logs.openstack.org/88/664088/1/gate/openstack-ansible-v2_registry_enabled/f825035/ : SUCCESS in 27m 36s","accounts_in_message":[],"_revision_number":1},{"id":"30c7652817982c5ac47a338f31128cb6da3d269c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 23:43:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d9fb1d70c8b87c6c0a4a0abd9f8d734494c8f3f","revisions":{"7d9fb1d70c8b87c6c0a4a0abd9f8d734494c8f3f":{"kind":"REWORK","_number":1,"created":"2019-06-07 23:24:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/664088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/88/664088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/664088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/664088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/664088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/664088/1"}}},"commit":{"parents":[{"commit":"ce26a9294fa244183ce22f51f3c66845d706aee5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ce26a9294fa244183ce22f51f3c66845d706aee5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:24:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:24:11.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I1229f7ea246dba173a416479baebc85f180b9f7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7d9fb1d70c8b87c6c0a4a0abd9f8d734494c8f3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7d9fb1d70c8b87c6c0a4a0abd9f8d734494c8f3f"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
