)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~664328","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fpike~I62564c909e564bcb4afa2268aa2871e9ec337f5e","project":"openstack/openstack-ansible-os_glance","branch":"stable/pike","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I62564c909e564bcb4afa2268aa2871e9ec337f5e","subject":"Updated from OpenStack Ansible Tests","status":"ABANDONED","created":"2019-06-10 15:44:40.000000000","updated":"2020-10-02 12:26:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e8574a96c7494daf77dacf486fe246c886bd27e3","_number":664328,"virtual_id_number":664328,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-10 17:36:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:36:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57328c14da745ffc7357f53f515922ceda65b8af","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:44:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf512519d3cbedd0f007227075dcd6de029edda1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:36:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/28/664328/1/check/build-openstack-sphinx-docs/597efc3/html/ : SUCCESS in 4m 18s\n- openstack-ansible-linters http://logs.openstack.org/28/664328/1/check/openstack-ansible-linters/5f90414/ : SUCCESS in 5m 38s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/28/664328/1/check/openstack-ansible-functional-centos-7/adcfc9c/ : SUCCESS in 43m 47s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/28/664328/1/check/openstack-ansible-functional-opensuse-423/e7b610a/ : FAILURE in 9m 46s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/28/664328/1/check/openstack-ansible-functional-ubuntu-xenial/e5b9997/ : SUCCESS in 32m 16s\n- openstack-ansible-nfs_glance http://logs.openstack.org/28/664328/1/check/openstack-ansible-nfs_glance/e7c18cb/ : SUCCESS in 32m 37s\n- openstack-ansible-v2_api_only http://logs.openstack.org/28/664328/1/check/openstack-ansible-v2_api_only/8f26f5e/ : SUCCESS in 37m 47s\n- openstack-ansible-v2_only_registry_enabled http://logs.openstack.org/28/664328/1/check/openstack-ansible-v2_only_registry_enabled/0d68510/ : SUCCESS in 25m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5e8c77720421199fd2643f7cf9ee71d5d07dd188","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 12:26:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e90754ebf40b458bcedd29731ed70866a4c2c2df","revisions":{"e90754ebf40b458bcedd29731ed70866a4c2c2df":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:44:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/664328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/28/664328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/664328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/664328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/664328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/664328/1"}}},"commit":{"parents":[{"commit":"d903264841508ec99800356c9cf43ec7800a33ad","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d903264841508ec99800356c9cf43ec7800a33ad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:38.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I62564c909e564bcb4afa2268aa2871e9ec337f5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e90754ebf40b458bcedd29731ed70866a4c2c2df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e90754ebf40b458bcedd29731ed70866a4c2c2df"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
