)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~666107","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ice4c8057d7d521ec35069934ca5b5d7f38017cc1","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ice4c8057d7d521ec35069934ca5b5d7f38017cc1","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-18 18:15:43.000000000","updated":"2019-06-18 21:55:42.000000000","submitted":"2019-06-18 21:55:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666107-1560894942430-ca750131","meta_rev_id":"da19a9be08e75be1dad2a82b6fc2bfb31d528852","_number":666107,"virtual_id_number":666107,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-18 21:55:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-18 18:54:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-18 20:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-18 20:09:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 18:54:30.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-18 20:09:45.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-18 21:55:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"041fc25633f1ce7118ab19d897ee815657b4922c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-18 18:15:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4c2ccb460de41fb4f169ba5b2374e7d765431d3","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-18 18:54:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ddd0f3fd40647bc3fab9d45496de2143ab4ea8df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:55:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/07/666107/1/check/openstack-ansible-linters/c6386c6/ : SUCCESS in 5m 51s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/07/666107/1/check/openstack-ansible-deploy-aio_metal-debian-stable/70ad12e/ : SUCCESS in 59m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/07/666107/1/check/openstack-ansible-deploy-aio_metal-centos-7/f1647c2/ : SUCCESS in 1h 31m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/07/666107/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/2b9183f/ : SUCCESS in 1h 09m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/07/666107/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9b8d8a0/ : SUCCESS in 57m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/07/666107/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/7c18ea7/ : SUCCESS in 59m 25s\n- openstack-tox-docs http://logs.openstack.org/07/666107/1/check/openstack-tox-docs/8ffda8e/html/ : SUCCESS in 3m 11s\n- openstack-ansible-nfs_glance http://logs.openstack.org/07/666107/1/check/openstack-ansible-nfs_glance/af19ffc/ : SUCCESS in 36m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a3f4e2f32cd9d4036b0de4b0d7a91134aa624eb3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-18 20:09:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b81053af24b113ced0ff5859af4dfd88340878e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 20:09:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c1c7677fb6406edff95ed8b969f5fe6a8328693","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:55:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"925920a20de8dfbedac0926051eaac03b54835eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:55:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/07/666107/1/gate/openstack-ansible-linters/e1891fe/ : SUCCESS in 5m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/07/666107/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/7147187/ : SUCCESS in 1h 09m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/07/666107/1/gate/openstack-ansible-deploy-aio_metal-centos-7/28f8edb/ : SUCCESS in 1h 19m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/07/666107/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/e40bed4/ : SUCCESS in 1h 08m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/07/666107/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9a3e9f2/ : SUCCESS in 1h 37m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/07/666107/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/00e4b7b/ : SUCCESS in 1h 02m 23s\n- openstack-tox-docs http://logs.openstack.org/07/666107/1/gate/openstack-tox-docs/b0dde34/html/ : SUCCESS in 2m 41s\n- openstack-ansible-nfs_glance http://logs.openstack.org/07/666107/1/gate/openstack-ansible-nfs_glance/ebf5c95/ : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b","revisions":{"96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b":{"kind":"REWORK","_number":1,"created":"2019-06-18 18:15:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/666107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/07/666107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/07/666107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/07/666107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/07/666107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/07/666107/1"}}},"commit":{"parents":[{"commit":"58461f77bcfe9883567cbc81dc013cdb27e24625","subject":"Merge \"Replace git.openstack.org with opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/58461f77bcfe9883567cbc81dc013cdb27e24625"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:15:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:15:42.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ice4c8057d7d521ec35069934ca5b5d7f38017cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
