)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~667704","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I55162bc2bf3cbb8858950e4abcf60a3de9929008","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"bug/1833725","hashtags":[],"change_id":"I55162bc2bf3cbb8858950e4abcf60a3de9929008","subject":"Add missing pre-reqs for glance cinder store","status":"MERGED","created":"2019-06-26 18:40:09.000000000","updated":"2019-07-30 20:32:56.000000000","submitted":"2019-07-30 20:32:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667704-1564518776708-8da850ff","meta_rev_id":"c43e0c8adf6efa33e0acf81a8ba3c3211e2d5673","_number":667704,"virtual_id_number":667704,"owner":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-07-30 20:32:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},{"value":2,"date":"2019-07-30 17:11:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-30 18:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-07-30 18:44:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 18:40:24.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-06-26 20:06:17.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-07-10 07:53:12.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-30 17:11:40.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-30 18:44:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-30 20:32:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"991e68d62ac1acd2310010ac930f7368547ea015","author":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"date":"2019-06-26 18:40:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d246bd448ac4795032bf1db1f920df8a6baeb14","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-06-26 18:40:24.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"a3e16a00272e10c28338f8f76ce775e90157377f","author":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"date":"2019-06-26 19:12:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1591352c5e7fff7f24f3be7ab49620dd6a010241","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-26 19:29:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(9 comments)\n\nLooks like a lot of loose ends fixed up here. Mostly whitespace to correct.","accounts_in_message":[],"_revision_number":2},{"id":"3195911a477f860ed0b1a4f489a4bfdba39f8638","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-26 19:29:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0a7423fbdafcc1b61d82c3a06f97759d5e70f92","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-06-26 20:06:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bff027fa0b1e5b14c5710e12f3cf2b43d0a82bba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 20:36:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/04/667704/2/check/openstack-ansible-linters/c060501/ : FAILURE in 5m 24s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/04/667704/2/check/openstack-ansible-deploy-aio_metal-debian-stable/7015ea6/ : SUCCESS in 1h 09m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/04/667704/2/check/openstack-ansible-deploy-aio_metal-centos-7/f6b1827/ : SUCCESS in 1h 16m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/04/667704/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/7921590/ : FAILURE in 1h 06m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/04/667704/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/814cff8/ : SUCCESS in 59m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/04/667704/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/3c3d8af/ : SUCCESS in 59m 31s\n- openstack-tox-docs http://logs.openstack.org/04/667704/2/check/openstack-tox-docs/b2ebca9/html/ : SUCCESS in 4m 26s\n- openstack-ansible-nfs_glance http://logs.openstack.org/04/667704/2/check/openstack-ansible-nfs_glance/14e714e/ : SUCCESS in 35m 34s","accounts_in_message":[],"_revision_number":2},{"id":"aaacfcc925725847fc4ff9d105fe32c99339ba2d","author":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"date":"2019-06-26 22:03:21.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nWill deal with whitespace issues","accounts_in_message":[],"_revision_number":2},{"id":"38a91b2f40c0c6f20e630e62dbbb877d21ad1571","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 06:32:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d71e2df6160a48468eaf41bf99369f573e3697e4","author":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"date":"2019-06-27 06:52:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"83fc4ebc20348644c55256de18b5d1c211e20f63","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-27 07:22:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nI think commit message need to be improved by being more clear on the \"Why\" we are doing this.\n\nNext to that I have a few questions about how it\u0027s implemented.","accounts_in_message":[],"_revision_number":2},{"id":"64206dc01a15e534b0d84e904e3cccd5a44f29eb","author":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"date":"2019-06-27 16:38:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f75acc6eec07d59aa769364bd0c486718450a845","author":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"date":"2019-06-27 16:45:46.000000000","message":"Patch Set 2:\n\n(12 comments)\n\nCommit message fleshed out as requested, futher discussion may be necessary on some of the remaining comments","accounts_in_message":[],"_revision_number":2},{"id":"eb4c073f188f1bbc2a55882ca1f5d0745a152a12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 18:08:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/04/667704/3/check/openstack-ansible-linters/d7c373e/ : FAILURE in 5m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/04/667704/3/check/openstack-ansible-deploy-aio_metal-debian-stable/ab4c1db/ : SUCCESS in 1h 04m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/04/667704/3/check/openstack-ansible-deploy-aio_metal-centos-7/55f5337/ : SUCCESS in 1h 22m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/04/667704/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/7467d7d/ : SUCCESS in 1h 10m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/04/667704/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/38c233e/ : SUCCESS in 57m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/04/667704/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/4280607/ : SUCCESS in 56m 31s\n- openstack-tox-docs http://logs.openstack.org/04/667704/3/check/openstack-tox-docs/0f77e7a/html/ : SUCCESS in 2m 56s\n- openstack-ansible-nfs_glance http://logs.openstack.org/04/667704/3/check/openstack-ansible-nfs_glance/7ee7318/ : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":3},{"id":"f2ad52fd0839ef9e3e12ce52ff00d9c3db147e37","author":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"date":"2019-06-27 20:32:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"96f176ffd21f092cae2ce49c82d7f1848a8816c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 21:57:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/04/667704/4/check/openstack-ansible-linters/ca16783/ : SUCCESS in 5m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/04/667704/4/check/openstack-ansible-deploy-aio_metal-debian-stable/456547e/ : SUCCESS in 59m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/04/667704/4/check/openstack-ansible-deploy-aio_metal-centos-7/22ced99/ : SUCCESS in 1h 18m 42s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/04/667704/4/check/openstack-ansible-deploy-aio_metal-opensuse-150/b2d8c2b/ : SUCCESS in 1h 10m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/04/667704/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3481ba2/ : SUCCESS in 55m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/04/667704/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/221e55d/ : SUCCESS in 1h 02m 26s\n- openstack-tox-docs http://logs.openstack.org/04/667704/4/check/openstack-tox-docs/50c3124/html/ : SUCCESS in 2m 25s\n- openstack-ansible-nfs_glance http://logs.openstack.org/04/667704/4/check/openstack-ansible-nfs_glance/5a0dc97/ : SUCCESS in 37m 26s","accounts_in_message":[],"_revision_number":4},{"id":"a4af36de88eec7c7854a42b0c6a3877af585c88e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-10 07:53:12.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"60791d4e12a2edbcf7816a2e4ce71175a7b24bd6","author":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"date":"2019-07-30 14:05:07.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nResponding to last comments, some questions in return","accounts_in_message":[],"_revision_number":4},{"id":"32d0ea89bd88adf9a22791bd9ea7a83470e43ec1","author":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"date":"2019-07-30 14:53:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"84dab6bd766887310cdbd01d20f5c5372d48d151","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 16:24:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/04/667704/5/check/openstack-ansible-linters/6084ca4/ : SUCCESS in 5m 13s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/04/667704/5/check/openstack-ansible-deploy-aio_metal-debian-stable/c222d27/ : SUCCESS in 1h 00m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/04/667704/5/check/openstack-ansible-deploy-aio_metal-centos-7/cdc25b8/ : SUCCESS in 1h 24m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/04/667704/5/check/openstack-ansible-deploy-aio_metal-opensuse-15/0d01d19/ : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/04/667704/5/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/71bee09/ : SUCCESS in 1h 05m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/04/667704/5/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e9b59af/ : SUCCESS in 1h 03m 17s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/04/667704/5/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/6316294/ : FAILURE in 41m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/04/667704/5/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/9ef8e03/ : SUCCESS in 53m 10s\n- openstack-tox-docs http://logs.openstack.org/04/667704/5/check/openstack-tox-docs/323db44/html/ : SUCCESS in 3m 14s\n- openstack-ansible-nfs_glance http://logs.openstack.org/04/667704/5/check/openstack-ansible-nfs_glance/759a105/ : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":5},{"id":"0b0421f77b9108aa66c0fd03d4c9089e3c741f02","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-30 17:11:40.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fc2591e726f861a2249be56cf498cd38fe37ea46","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-30 18:44:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6d1622721e497ff0190dc357494b5d36f6b9bb5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 18:45:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4d2b8adb45a86ed19504cde4322ddf8f9d987725","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:32:56.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/04/667704/5/gate/openstack-ansible-linters/3c44752/ : SUCCESS in 5m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/04/667704/5/gate/openstack-ansible-deploy-aio_metal-debian-stable/948e6d0/ : SUCCESS in 1h 08m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/04/667704/5/gate/openstack-ansible-deploy-aio_metal-centos-7/759fb2a/ : SUCCESS in 1h 24m 42s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/04/667704/5/gate/openstack-ansible-deploy-aio_metal-opensuse-15/fcd9a05/ : SUCCESS in 1h 21m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/04/667704/5/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8546f81/ : SUCCESS in 59m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/04/667704/5/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/7ae8feb/ : SUCCESS in 1h 03m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/04/667704/5/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/472a440/ : SUCCESS in 44m 23s\n- openstack-tox-docs http://logs.openstack.org/04/667704/5/gate/openstack-tox-docs/b172120/html/ : SUCCESS in 3m 19s\n- openstack-ansible-nfs_glance http://logs.openstack.org/04/667704/5/gate/openstack-ansible-nfs_glance/126ff9a/ : SUCCESS in 34m 32s","accounts_in_message":[],"_revision_number":5},{"id":"179d5df24a32083017c9d1ca9b0f1a44f46102ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:32:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2636d7eaea6a22ccde8510b2de3d52de269a6445","revisions":{"75a357419c7d33587da7e281404307e3cbc6bcf4":{"kind":"REWORK","_number":1,"created":"2019-06-26 18:40:09.000000000","uploader":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"ref":"refs/changes/04/667704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/04/667704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/1"}}},"commit":{"parents":[{"commit":"96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b"}]}],"author":{"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","date":"2019-06-26 18:24:42.000000000","tz":60},"committer":{"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","date":"2019-06-26 18:39:46.000000000","tz":60},"subject":"Add oslo.rootwrap to glance_pip_packages","message":"Add oslo.rootwrap to glance_pip_packages\n\nThere is a dependency on olso.rootwrapper when using cinder store\nwith glance, however this was not included in glance_pip_packages\n\nChange-Id: I55162bc2bf3cbb8858950e4abcf60a3de9929008\nPartial-Bug: #1833725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/75a357419c7d33587da7e281404307e3cbc6bcf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/75a357419c7d33587da7e281404307e3cbc6bcf4"}]},"branch":"refs/heads/master"},"38283bf0063637af567f1a7a95a55103476b51e7":{"kind":"REWORK","_number":2,"created":"2019-06-26 19:12:36.000000000","uploader":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"ref":"refs/changes/04/667704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/04/667704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/2"}}},"commit":{"parents":[{"commit":"96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b"}]}],"author":{"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","date":"2019-06-26 18:24:42.000000000","tz":60},"committer":{"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","date":"2019-06-26 19:07:53.000000000","tz":60},"subject":"Add missing pre-reqs for glance cinder store","message":"Add missing pre-reqs for glance cinder store\n\nThere are a number of missing dependencies when using cinder store\nwith glance:\n\n - olso.rootwrapper is was not included in glance_pip_packages\n\n - files/rootwrap.d/glance_cinder_store.filters\n\n - glance user is not added to sudoers\n\nglance_pip_packages updated, missing files now dropped in to\nrequired locations by glance_post_install.yml task\n\nChange-Id: I55162bc2bf3cbb8858950e4abcf60a3de9929008\nPartial-Bug: #1833725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/38283bf0063637af567f1a7a95a55103476b51e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/38283bf0063637af567f1a7a95a55103476b51e7"}]},"branch":"refs/heads/master"},"6cf9d31d024bf21cd5ad47ae1f0efe4929e89da8":{"kind":"REWORK","_number":3,"created":"2019-06-27 16:38:20.000000000","uploader":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"ref":"refs/changes/04/667704/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/04/667704/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/3"}}},"commit":{"parents":[{"commit":"96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b"}]}],"author":{"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","date":"2019-06-26 18:24:42.000000000","tz":60},"committer":{"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","date":"2019-06-27 16:34:21.000000000","tz":60},"subject":"Add missing pre-reqs for glance cinder store","message":"Add missing pre-reqs for glance cinder store\n\nThere are a number of missing dependencies in the role when using cinder\nstore with glance. Specifically rootwrap is required for elevating access\nwhen using os-brick to connect to cinder iscsi/fc volume back end storage.\nThis patch addresses the following:\n\n - olso.rootwrap is not included in glance_pip_packages\n\n - files/rootwrap.d/glance_cinder_store.filters is missing\n\n - glance user is not added to sudoers\n\nglance_pip_packages updated, missing rootwrap.d and sudoer files now dropped in to\nTheir required locations by glance_post_install.yml task\n\nChange-Id: I55162bc2bf3cbb8858950e4abcf60a3de9929008\nCloses-Bug: #1833725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6cf9d31d024bf21cd5ad47ae1f0efe4929e89da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6cf9d31d024bf21cd5ad47ae1f0efe4929e89da8"}]},"branch":"refs/heads/master"},"bdce898942387f292804ee2626345f77071ae077":{"kind":"REWORK","_number":4,"created":"2019-06-27 20:32:19.000000000","uploader":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"ref":"refs/changes/04/667704/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/04/667704/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/4"}}},"commit":{"parents":[{"commit":"96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b"}]}],"author":{"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","date":"2019-06-26 18:24:42.000000000","tz":60},"committer":{"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","date":"2019-06-27 20:31:30.000000000","tz":60},"subject":"Add missing pre-reqs for glance cinder store","message":"Add missing pre-reqs for glance cinder store\n\nThere are a number of missing dependencies in the role when using cinder\nstore with glance. Specifically rootwrap is required for elevating access\nwhen using os-brick to connect to cinder iscsi/fc volume back end storage.\nThis patch addresses the following:\n\n - olso.rootwrap is not included in glance_pip_packages\n\n - files/rootwrap.d/glance_cinder_store.filters is missing\n\n - glance user is not added to sudoers\n\nglance_pip_packages updated, missing rootwrap.d and sudoer files now dropped in to\nTheir required locations by glance_post_install.yml task\n\nChange-Id: I55162bc2bf3cbb8858950e4abcf60a3de9929008\nCloses-Bug: #1833725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/bdce898942387f292804ee2626345f77071ae077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/bdce898942387f292804ee2626345f77071ae077"}]},"branch":"refs/heads/master"},"2636d7eaea6a22ccde8510b2de3d52de269a6445":{"kind":"REWORK","_number":5,"created":"2019-07-30 14:53:03.000000000","uploader":{"_account_id":30204,"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","username":"ceemac"},"ref":"refs/changes/04/667704/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/04/667704/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/04/667704/5"}}},"commit":{"parents":[{"commit":"96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/96f0d5bd4ae796fdb3e9f52a68ed88c193fae72b"}]}],"author":{"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","date":"2019-06-26 18:24:42.000000000","tz":60},"committer":{"name":"Craig McIntyre","email":"craig.mcintyre@k3btg.com","date":"2019-07-30 14:52:38.000000000","tz":60},"subject":"Add missing pre-reqs for glance cinder store","message":"Add missing pre-reqs for glance cinder store\n\nThere are a number of missing dependencies in the role when using cinder\nstore with glance. Specifically rootwrap is required for elevating access\nwhen using os-brick to connect to cinder iscsi/fc volume back end storage.\nThis patch addresses the following:\n\n - olso.rootwrap is not included in glance_pip_packages\n\n - files/rootwrap.d/glance_cinder_store.filters is missing\n\n - glance user is not added to sudoers\n\nglance_pip_packages updated, missing rootwrap.d and sudoer files now dropped in to\nTheir required locations by glance_post_install.yml task\n\nChange-Id: I55162bc2bf3cbb8858950e4abcf60a3de9929008\nCloses-Bug: #1833725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2636d7eaea6a22ccde8510b2de3d52de269a6445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2636d7eaea6a22ccde8510b2de3d52de269a6445"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
