)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~670890","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I12c5a117d9ca508f24a36a477d2d71c36e6c8c96","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"use_journalctl","hashtags":[],"change_id":"I12c5a117d9ca508f24a36a477d2d71c36e6c8c96","subject":"Use systemd-journald instead of log files","status":"MERGED","created":"2019-07-15 17:40:55.000000000","updated":"2019-07-16 21:10:50.000000000","submitted":"2019-07-16 21:10:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670890-1563311450342-e47bf6dd","meta_rev_id":"faba457fd9ada43c600b34ac0c39c3571121e698","_number":670890,"virtual_id_number":670890,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-07-16 21:10:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-07-16 16:20:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-16 19:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-16 19:24:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 16:20:09.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-07-16 19:24:59.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-16 21:10:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24980d0a6e769a38250e9d76d0dc1a2b77a891a8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-15 17:40:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cb7859ce090a0ae763590f3522832e0dfa1ca35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 18:49:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/90/670890/1/check/openstack-ansible-linters/1838fd4/ : SUCCESS in 6m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/90/670890/1/check/openstack-ansible-deploy-aio_metal-debian-stable/be75dcd/ : FAILURE in 32m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/90/670890/1/check/openstack-ansible-deploy-aio_metal-centos-7/52dd89f/ : FAILURE in 38m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/90/670890/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/1160774/ : FAILURE in 34m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/90/670890/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/fffcded/ : FAILURE in 33m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/90/670890/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5eec031/ : SUCCESS in 59m 31s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/90/670890/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/1405c19/ : SUCCESS in 1h 02m 21s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/90/670890/1/check/openstack-tox-docs/bf030ec/html/ : SUCCESS in 3m 03s\n- openstack-ansible-nfs_glance http://logs.openstack.org/90/670890/1/check/openstack-ansible-nfs_glance/756ff9f/ : FAILURE in 29m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f46162670c245c645f215ec84df6473156e77257","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-15 19:39:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8204d661b63536f2f6e70a97a800789a8989bd01","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-15 19:46:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"41b7b9649fad13c892ed9b74543da7ea2793db19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 21:17:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/90/670890/3/check/openstack-ansible-linters/f22aaed/ : SUCCESS in 5m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/90/670890/3/check/openstack-ansible-deploy-aio_metal-debian-stable/29e324a/ : SUCCESS in 1h 12m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/90/670890/3/check/openstack-ansible-deploy-aio_metal-centos-7/4d3adb1/ : SUCCESS in 1h 25m 10s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/90/670890/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/c3c1416/ : SUCCESS in 1h 12m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/90/670890/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9c35384/ : SUCCESS in 52m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/90/670890/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8e6f0fc/ : SUCCESS in 1h 00m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/90/670890/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/341b6a5/ : SUCCESS in 52m 16s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/90/670890/3/check/openstack-tox-docs/5a2f6aa/html/ : SUCCESS in 2m 49s\n- openstack-ansible-nfs_glance http://logs.openstack.org/90/670890/3/check/openstack-ansible-nfs_glance/f989924/ : SUCCESS in 42m 50s","accounts_in_message":[],"_revision_number":3},{"id":"089e27d6a5bddfa9c2de7fdb0e6d396eefa2e723","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-16 12:40:58.000000000","message":"Patch Set 3: Code-Review-1\n\ndisable-logging so uwsgi doesnt interfere too much","accounts_in_message":[],"_revision_number":3},{"id":"d7a600e521139d1aeece688399b4c1646b778b02","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 13:09:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"161f9759108eeea28b68f56b37fa938fc401624f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 14:35:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/90/670890/4/check/openstack-ansible-linters/bbccbc3/ : SUCCESS in 5m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/90/670890/4/check/openstack-ansible-deploy-aio_metal-debian-stable/5f25831/ : SUCCESS in 1h 04m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/90/670890/4/check/openstack-ansible-deploy-aio_metal-centos-7/1060fef/ : SUCCESS in 1h 20m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/90/670890/4/check/openstack-ansible-deploy-aio_metal-opensuse-15/02e7ea1/ : SUCCESS in 1h 07m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/90/670890/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5ab6ced/ : SUCCESS in 58m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/90/670890/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/dc2b543/ : SUCCESS in 1h 03m 44s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/90/670890/4/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/c4e5642/ : SUCCESS in 53m 35s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/90/670890/4/check/openstack-tox-docs/ab8dd15/html/ : SUCCESS in 2m 27s\n- openstack-ansible-nfs_glance http://logs.openstack.org/90/670890/4/check/openstack-ansible-nfs_glance/519e611/ : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":4},{"id":"6594b2c9423dedc7b3093c2a60f0b12c3cf7b8aa","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-07-16 16:20:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e286ead3ef92dd0141416ea50fe15be8bf8c151a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-16 19:24:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6d31ac2039d64a47bcd6a065b8f32ea4526e007d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 19:25:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ba58f28be42c284a8febcd49fb0f9528768b5580","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 21:10:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"1a5ed61fdee56ce3f70e85a9f4eebd8e0bd217c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 21:10:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/90/670890/4/gate/openstack-ansible-linters/9de0be1/ : SUCCESS in 6m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/90/670890/4/gate/openstack-ansible-deploy-aio_metal-debian-stable/8f04745/ : SUCCESS in 1h 10m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/90/670890/4/gate/openstack-ansible-deploy-aio_metal-centos-7/3af0f11/ : SUCCESS in 1h 24m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/90/670890/4/gate/openstack-ansible-deploy-aio_metal-opensuse-15/1e85eea/ : SUCCESS in 1h 09m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/90/670890/4/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a11938c/ : SUCCESS in 1h 09m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/90/670890/4/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/8d82cfb/ : SUCCESS in 57m 50s\n- openstack-tox-docs http://logs.openstack.org/90/670890/4/gate/openstack-tox-docs/29357ca/html/ : SUCCESS in 2m 55s\n- openstack-ansible-nfs_glance http://logs.openstack.org/90/670890/4/gate/openstack-ansible-nfs_glance/1013034/ : SUCCESS in 37m 18s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"002d2a91ea2dc0cb9a8147101afe419b0b58b531","revisions":{"e10c223950c87ce66bdd3a704c2fba3b5b9bea03":{"kind":"REWORK","_number":1,"created":"2019-07-15 17:40:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/670890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/90/670890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/1"}}},"commit":{"parents":[{"commit":"30c000c226c3fcc181ec523b3fa5e010b1f5cf6a","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/30c000c226c3fcc181ec523b3fa5e010b1f5cf6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-15 17:40:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-15 17:40:40.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald.\n\nChange-Id: I12c5a117d9ca508f24a36a477d2d71c36e6c8c96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e10c223950c87ce66bdd3a704c2fba3b5b9bea03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e10c223950c87ce66bdd3a704c2fba3b5b9bea03"}]},"branch":"refs/heads/master"},"8bbb8169ceb63b4f2a577ecd917a4663c21f84b2":{"kind":"REWORK","_number":2,"created":"2019-07-15 19:39:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/670890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/90/670890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/2"}}},"commit":{"parents":[{"commit":"30c000c226c3fcc181ec523b3fa5e010b1f5cf6a","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/30c000c226c3fcc181ec523b3fa5e010b1f5cf6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-15 17:40:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-07-15 19:39:03.000000000","tz":0},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald.\n\nChange-Id: I12c5a117d9ca508f24a36a477d2d71c36e6c8c96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8bbb8169ceb63b4f2a577ecd917a4663c21f84b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8bbb8169ceb63b4f2a577ecd917a4663c21f84b2"}]},"branch":"refs/heads/master"},"c3070b11bdeea8bfe99a6a5ff381f427dc77fe8b":{"kind":"REWORK","_number":3,"created":"2019-07-15 19:46:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/670890/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/90/670890/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/3"}}},"commit":{"parents":[{"commit":"30c000c226c3fcc181ec523b3fa5e010b1f5cf6a","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/30c000c226c3fcc181ec523b3fa5e010b1f5cf6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-15 17:40:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-15 19:46:44.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald.\n\nChange-Id: I12c5a117d9ca508f24a36a477d2d71c36e6c8c96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c3070b11bdeea8bfe99a6a5ff381f427dc77fe8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c3070b11bdeea8bfe99a6a5ff381f427dc77fe8b"}]},"branch":"refs/heads/master"},"002d2a91ea2dc0cb9a8147101afe419b0b58b531":{"kind":"REWORK","_number":4,"created":"2019-07-16 13:09:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/670890/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/90/670890/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/90/670890/4"}}},"commit":{"parents":[{"commit":"30c000c226c3fcc181ec523b3fa5e010b1f5cf6a","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/30c000c226c3fcc181ec523b3fa5e010b1f5cf6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-15 17:40:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 13:09:28.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald. We also disable uwsgi logging, since it dublicates\nrequests that are logged by service itself.\n\nChange-Id: I12c5a117d9ca508f24a36a477d2d71c36e6c8c96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/002d2a91ea2dc0cb9a8147101afe419b0b58b531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/002d2a91ea2dc0cb9a8147101afe419b0b58b531"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
