)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~677309","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Iaee4b8e18c8f2173e3d4c7ef9d7d1f4e65d59985","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Iaee4b8e18c8f2173e3d4c7ef9d7d1f4e65d59985","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-08-19 22:20:43.000000000","updated":"2019-08-20 16:50:48.000000000","submitted":"2019-08-20 16:49:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677309-1566319762861-72204814","meta_rev_id":"650353aeb83f827232b58cf26ac09a78a30c6c96","_number":677309,"virtual_id_number":677309,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-20 16:49:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-08-20 10:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-08-20 10:56:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 10:56:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-08-20 16:49:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd9f234fb1c4a55073ec5508c72fda635f8123f5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-19 22:20:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50320404a3fe0c2ce331512505e9aca09fd222c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 23:57:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/29ed85ce7ac148f4ab955396462b7512 : SUCCESS in 6m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ead402a9cad2472e99db9c0cf2c7da66 : POST_FAILURE in 1h 31m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1597ce2b7e5d42099cb79ce7dfb7e215 : SUCCESS in 1h 22m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ce3d305940854587b72cf3700f05e4e8 : SUCCESS in 1h 13m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c866f83f1ed4f3f8ecae59778bd02b7 : SUCCESS in 57m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1fc63caaaacf4b7e983d38ca4e92fd9a : SUCCESS in 1h 01m 38s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/43653e5c1d22483caa69778feb72e39a : SUCCESS in 53m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15db6402aeef40b0857a5f2451eadb94 : POST_FAILURE in 1h 22m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a5d46780c9435fb564ad88bc552cd3 : SUCCESS in 4m 23s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/4bb2b942243e4f2db0b84911cf6450bb : SUCCESS in 37m 17s","accounts_in_message":[],"_revision_number":1},{"id":"067c70fffb64c3efa6f16b6405589d06c57db321","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-20 03:05:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a24b5adcf98e1e339f5649f924d88dd90387ae7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 04:50:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6556ce60e9ad45efbd2e44de01e3fd4c : SUCCESS in 5m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/5ba4ddcc1f0746db9ab623071a97464b : SUCCESS in 1h 06m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/de61ac5517d64b1a8342c5a23c9a2d89 : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8535b861056c48e2936d9af3a3d8defb : SUCCESS in 1h 12m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63241a42adb546948fb9c513cf79f631 : SUCCESS in 1h 04m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ae82ebbdd5b94df88c7a25bfcee3fcbb : SUCCESS in 1h 08m 28s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4866f6688cf240b09265804ee3fa1c02 : SUCCESS in 57m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39a91641bcb74b5f803b17d39191e0cd : SUCCESS in 1h 02m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c71a4f701a9d4be0b909e3ffdb936b71 : SUCCESS in 4m 23s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/c7d1e44dfb224093b828413e2525da7f : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":2},{"id":"e139a8e57f26c84350e5421512e1e1b7717845bd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-20 10:56:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fc69708046cf728fa77f894bb7f78bac0915cd3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 10:56:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ba53acbc474a669de1eccfad8b1fb71314eeda7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 12:34:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/490303dc9b2744f3a37ce901a1189472 : SUCCESS in 6m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6c79d8ab3eeb434780d09a15aa0c4c3c : SUCCESS in 1h 30m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2aa77f8c30a24ac593b1ffd1d4be1c07 : SUCCESS in 1h 23m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/73ca28f10d05441da98de8083903cd14 : FAILURE in 39m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/381ada1f4c504e35a1d457c1b6d991df : SUCCESS in 59m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c46c84e6f4574f868603572877739db0 : SUCCESS in 1h 03m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be94529f8d8a444e84b92beb4b41be7a : SUCCESS in 1h 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/689fdba9863d46659aca928c7881b62d : SUCCESS in 3m 14s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/7ea4ecf1173a4547bcacce530dbe5afa : SUCCESS in 39m 37s","accounts_in_message":[],"_revision_number":2},{"id":"fc21ce84a40122d0e7c4ece23997cab2423404c1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-20 13:21:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"14537ca563450942ee8eb01e65142430dd2f8b2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 15:06:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/81b67254216643dda5355d6fc5adc713 : SUCCESS in 6m 13s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/eefcdf11816a4b9f90dc118da5dc857a : SUCCESS in 1h 08m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fda728aa9242409b88c6ae4f14bec4bf : SUCCESS in 1h 16m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b99eb108049a4733961a5d4abaa532e3 : SUCCESS in 1h 16m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b090fdd016384e889a0530974ff16961 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3e7a9525bebc412ea9b4dae83ade885c : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d37ef576cf254676af457912dbc23bf8 : SUCCESS in 59m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ffb365c2b984b3e85b6ea8168e7d0fb : SUCCESS in 46m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1b7b26147c43d18d0491239296a998 : SUCCESS in 3m 49s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/00ef17e595bd418b85296590f3eb3969 : SUCCESS in 44m 15s","accounts_in_message":[],"_revision_number":2},{"id":"a524cb0c18af44bf44a83a5fbd82af14bfb2cd34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 15:06:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"194aba6299026b34048e00e3b1b101a1f4e93c71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 16:49:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"092d936198a28065a9acd8cecb36a5473e0d4085","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 16:49:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/789c293c77054e5bb92bbb37e177789e : SUCCESS in 8m 46s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2bf3f5aae62146998771fd066ee5bd2a : SUCCESS in 1h 04m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/735f7cfb665f43b4a4a0746a5fe52147 : SUCCESS in 1h 36m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cddda82649f548cc94e9f908464edeec : SUCCESS in 1h 15m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07db659c063d4c41a19f5c1e67cef823 : SUCCESS in 1h 02m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2e564cdd9bdd4df6802c5d9d7c4902e0 : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c264bc31fa24d00a225a87ffe8e8d1c : SUCCESS in 48m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3fb43f726c54754b5ea61d9f36681dd : SUCCESS in 4m 42s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/cbcd9cbb49644f8ea8d70f7aea239e58 : SUCCESS in 35m 11s","accounts_in_message":[],"_revision_number":2},{"id":"1f4b9f31ac056094954eed906d23f66681226c22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 16:50:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a645955727b8439fb41dd985e7f5e3f2 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63","revisions":{"e10226add9bed149b65b0013631969b624b7e889":{"kind":"REWORK","_number":1,"created":"2019-08-19 22:20:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/677309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/09/677309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/09/677309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/09/677309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/09/677309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/09/677309/1"}}},"commit":{"parents":[{"commit":"871bbc8e7f9a19dd2700dc4d4d24771aec54f449","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/871bbc8e7f9a19dd2700dc4d4d24771aec54f449"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-19 22:20:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-19 22:20:41.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iaee4b8e18c8f2173e3d4c7ef9d7d1f4e65d59985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e10226add9bed149b65b0013631969b624b7e889"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e10226add9bed149b65b0013631969b624b7e889"}]},"branch":"refs/heads/master"},"4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63":{"kind":"REWORK","_number":2,"created":"2019-08-20 03:05:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/677309/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/09/677309/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/09/677309/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/09/677309/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/09/677309/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/09/677309/2"}}},"commit":{"parents":[{"commit":"871bbc8e7f9a19dd2700dc4d4d24771aec54f449","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/871bbc8e7f9a19dd2700dc4d4d24771aec54f449"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-20 03:05:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-20 03:05:56.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iaee4b8e18c8f2173e3d4c7ef9d7d1f4e65d59985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
