)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~677978","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Frocky~I6d2f7e2c341e197a5d7b391319e1cd4ea12f0719","project":"openstack/openstack-ansible-os_glance","branch":"stable/rocky","topic":"goodbye_42.3","hashtags":[],"change_id":"I6d2f7e2c341e197a5d7b391319e1cd4ea12f0719","subject":"Remove openSUSE 42.3 job","status":"MERGED","created":"2019-08-22 11:45:38.000000000","updated":"2019-08-27 13:55:00.000000000","submitted":"2019-08-27 13:55:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677978-1566914100697-6c2ca45c","meta_rev_id":"a292427f755754eb689023041c98e61203540266","_number":677978,"virtual_id_number":677978,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-27 13:55:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2019-08-27 08:46:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-27 12:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-27 13:02:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-08-27 13:02:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 08:46:13.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-27 12:20:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-08-27 13:02:35.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-27 13:55:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"292b93a2281367239bb885d5245cc7e2886fbee9","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-22 11:45:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87fedea2fe77902cbe6d4034ca57e208b02e75c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 12:35:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0affc8eb08fe464bb316f5004c457063 : SUCCESS in 11m 16s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/3da3b7fd9c31473a97cf4ec3f103ff17 : POST_FAILURE in 37m 52s\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/79360b9376b642c986090d2e7530aab6 : POST_FAILURE in 11m 06s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69bd0a04ddb34f1c98f0d0061f05ff69 : SUCCESS in 32m 15s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d63f4b02239146daa143dfbd6835f0c4 : SUCCESS in 32m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/be318929711e42d79004eed97785960d : SUCCESS in 5m 07s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/94e917208b65420d9455b88aea3b9918 : SUCCESS in 44m 25s\n- openstack-ansible-v2_registry_enabled https://zuul.opendev.org/t/openstack/build/cc375e9c91aa46f48232375b3676a57b : SUCCESS in 32m 38s\n- openstack-ansible-functional-distro_install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4740a1bbfd5e4f6090487ae4609d9da3 : SUCCESS in 31m 12s\n- openstack-ansible-functional-distro_install-centos-7 https://zuul.opendev.org/t/openstack/build/70f4afd086ae4a0b81c390e233243690 : POST_FAILURE in 37m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c53d037700534819788757de53f74bf9b18cffa0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-27 08:46:13.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck - post_failure should be fixed now","accounts_in_message":[],"_revision_number":1},{"id":"779c7faa3f7744489f35400ee7ae867c8a6fe5e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 09:50:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f9e9d1f334f64793ae809554692ee4c7 : SUCCESS in 7m 31s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b1c72450af2843589ea57484deb58b4d : SUCCESS in 1h 02m 19s\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/1540f37b9d5e4ae68a9f8fd4e4793ee5 : RETRY_LIMIT in 3m 19s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d218bb7f17c94cbd9a2892fe55a985b8 : SUCCESS in 39m 32s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/72ccc9fe3c1a4686bcbbd3707aa263ac : SUCCESS in 35m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7e0f8159929c43b4a572bdd982bff1f1 : SUCCESS in 4m 54s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/4c497546523f4f76a821a9a583ddb9cc : SUCCESS in 41m 50s\n- openstack-ansible-v2_registry_enabled https://zuul.opendev.org/t/openstack/build/dc029fc627a14ef5a0ccd6d62192c748 : SUCCESS in 36m 05s\n- openstack-ansible-functional-distro_install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ca52404be6e40ac8a1ed1236af93110 : SUCCESS in 34m 40s\n- openstack-ansible-functional-distro_install-centos-7 https://zuul.opendev.org/t/openstack/build/b1321ab6427d48d58892dca2b1650ccf : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":1},{"id":"31a545080184c19ac9202fdaee2d7ffe39783da0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-27 12:10:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a63a70df1bc457ef98b5f0c1df783d21cf622e7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-27 12:13:17.000000000","message":"Removed Code-Review+2 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"fa109af10c2ceec478a6e6317776979ebaffabdf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-27 12:18:03.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027d love to get supported versions updated in role meta as well","accounts_in_message":[],"_revision_number":1},{"id":"59edbf7f6c8363128bb7c0349d207c4dc9a0733a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-27 12:20:50.000000000","message":"Removed Code-Review-1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"3263682e2c333cba1b031e76e57096b7e6b26d40","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-27 12:20:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"786e9a411ce8f411c90517166b967ebb9174f11b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-27 13:02:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a91d2f3d54b2d3827c0190f86de88b12c531ccc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:02:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3cc36bcf4bdcb11273a3224e78dfdb44a208be35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:55:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/501e55733eee4918ab89e3afd985e23c : SUCCESS in 5m 20s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/16cf6ed064294493b04769212d5cc2f5 : SUCCESS in 44m 40s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6def6d4586774a8f82def3f9f8c077fc : SUCCESS in 37m 20s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/7b41e2d7efa3421697b6ea00f929e74f : SUCCESS in 33m 31s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e2ea4406ddd54e838242f9161834eaec : SUCCESS in 6m 03s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/7ea424eb5c6340049de6b768bb5bc85d : SUCCESS in 34m 40s\n- openstack-ansible-v2_registry_enabled https://zuul.opendev.org/t/openstack/build/18124e331c6f4e24b720f4d40519d3fe : SUCCESS in 30m 30s\n- openstack-ansible-functional-distro_install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22cdc9dbb8fc41c2a7a8e589441c4109 : SUCCESS in 37m 16s\n- openstack-ansible-functional-distro_install-centos-7 https://zuul.opendev.org/t/openstack/build/cbc0a3707c624bc29dcfafc87acfdb5e : SUCCESS in 36m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f2104f46599e07e48a45bff0e6b22aed35c7f9a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:55:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ed3f3031d9f5f29b84d44c5f6b419fae05e18d8","revisions":{"9ed3f3031d9f5f29b84d44c5f6b419fae05e18d8":{"kind":"REWORK","_number":1,"created":"2019-08-22 11:45:38.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/78/677978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/78/677978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/677978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/677978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/677978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/677978/1"}}},"commit":{"parents":[{"commit":"82f5bfb87260d673fd2935e1e25db09376478cd3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/82f5bfb87260d673fd2935e1e25db09376478cd3"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 11:45:17.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 11:45:17.000000000","tz":120},"subject":"Remove openSUSE 42.3 job","message":"Remove openSUSE 42.3 job\n\nopenSUSE 42.3 is dead, remove the job, we test with openSUSE 15 now (as\non master). This allows to remove the nodes from CI.\n\nDepends-On: https://review.opendev.org/677947\nChange-Id: I6d2f7e2c341e197a5d7b391319e1cd4ea12f0719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/9ed3f3031d9f5f29b84d44c5f6b419fae05e18d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/9ed3f3031d9f5f29b84d44c5f6b419fae05e18d8"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
