)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~678055","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I6f129940e55130c289d94138171cee54dbd28fc1","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"ansible-role-uwsgi","hashtags":[],"change_id":"I6f129940e55130c289d94138171cee54dbd28fc1","subject":"Start using uWSGI role","status":"MERGED","created":"2019-08-22 16:27:40.000000000","updated":"2019-09-06 15:31:38.000000000","submitted":"2019-09-06 15:30:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678055-1567783822002-71632aeb","meta_rev_id":"651bded9157be761eb84d971bbbb3109759f5928","_number":678055,"virtual_id_number":678055,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"date":"2019-09-05 15:58:48.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-06 15:30:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-09-06 02:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-06 13:54:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-06 13:54:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 02:06:24.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-09-06 13:54:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-06 15:30:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"273baba00e1280285e8cf5bf1f097a2305dcbf19","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-22 16:27:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bf3d48aabc23194cc5c9879028b303d0fa1e10c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 16:31:29.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"3b2523f97ed4925982771b2a1f21b6a7897b2727","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-22 16:32:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"09d41238518616c3c3e3ef7122288d2fc62b6715","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 17:16:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a297760a60254a00be29f838a197ea25 : SUCCESS in 6m 39s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9ef4a10661a04774b5f8d16efb8d31df : FAILURE in 38m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2c29aeaf1bfd4b50ae8c38305f611517 : FAILURE in 43m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d348f809f17b440b87113bb12475715c : FAILURE in 32m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24949c8f7c304733a360b561ba7c4e23 : FAILURE in 36m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3b91a5efedcd49d8b44c0f76f954a32d : FAILURE in 39m 02s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/57aafddb35814fa2a6e7498030f12ccf : FAILURE in 29m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/336dc74cc58841f8bbaadc5b37a74813 : FAILURE in 25m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e13d1a744474e54a7a8c35ec72815ec : SUCCESS in 3m 12s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/c5dbc3cb05a74e74b1bc4d06950690fc : FAILURE in 27m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2adde94239539ec9d1eec7b4336ec493e756432c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-22 17:40:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43d41207867a349b9d81f45e7fbb745b0f06b86c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 19:35:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f679c9703a2341aa9e10dc114c874a7c : SUCCESS in 7m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3dce64e101f94e5b9699dda60c1c7835 : SUCCESS in 1h 20m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/792341a08c5345b09f5d72e71b26835b : SUCCESS in 1h 47m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4feb0f238b3e4588ba71aecffc6fbc9e : SUCCESS in 1h 23m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb00b3be5cdc472e8d4a7eb33504d79d : SUCCESS in 1h 10m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/828c3699ade0483e8b4a694189ba8a75 : FAILURE in 1h 05m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/092a31ecad354bc7b7d45a949f76749a : FAILURE in 46m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81cfb14cc8aa4f779cce4aafa9a77cde : FAILURE in 41m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ddbf1c56a15469b90e2f29b06ec5c31 : SUCCESS in 4m 11s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/1bc5ab63fa3944ffa6ff5f912aeae081 : SUCCESS in 42m 16s","accounts_in_message":[],"_revision_number":2},{"id":"04eda85e6341be63bfae25a6c6abf4736f9aff9e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-22 20:29:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"60297fa00ed9ad75af6764e9e657648a9c1b8c5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:37:50.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"52507a3265c8b0c84157575a28bfc33f4ddf0167","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-22 20:38:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"61932d91baf59db57036d5feb0c0740962d2cb48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 22:18:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0fc41d6dafcc4707a943467000dd80e7 : SUCCESS in 7m 48s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/078c366ca3c548a59f8b6e4c9060392a : FAILURE in 1h 01m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/46cc4eaa7e044dd0b388233ce5113ef5 : FAILURE in 1h 33m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/90791e667aea42ae8f1ab8aeccb2ce50 : FAILURE in 1h 17m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f2ed38d3f024968aaaa02605e87034b : FAILURE in 54m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3c9738d0312749e084a377479977a2cd : SUCCESS in 54m 07s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1b1d3fcb2fe849f584fc51b991dad62a : SUCCESS in 54m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89ba9093c2484062a9a8980b68d4aed2 : SUCCESS in 58m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ab7c14b9bc24d549be9654c4d7c2f87 : SUCCESS in 3m 24s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/d1de6f0db9b24629a717b263303d1a0a : FAILURE in 31m 00s","accounts_in_message":[],"_revision_number":3},{"id":"9205008a1162396688f8667270891676419cded6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-23 01:59:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9b4af737f2dcea5d2463a9dba0cab60f544db095","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 03:38:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/28386ee1dfda4af6bcce254c1a84654e : SUCCESS in 8m 47s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9ea26245f2db423d84f68f1e64e3ed5f : SUCCESS in 1h 12m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8a95ed5422fc46b6ac70378e3e9a9256 : SUCCESS in 1h 37m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/114756cb05414898899b134afd554312 : SUCCESS in 1h 25m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccfc6f89049c409583f2232a92fd6f3d : SUCCESS in 1h 16m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ce6b034e45f94ed4b4ac153084ded3d8 : FAILURE in 1h 03m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/217d42f0b14b494a8979372390bab1a9 : FAILURE in 52m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c6b78ae841b464480ba8c175297b906 : FAILURE in 47m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c9d89b15ac4c49a6b3e27b166d20d5 : SUCCESS in 4m 56s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/cdb5b9bc239a41bd82d9c1ece75281eb : SUCCESS in 45m 04s","accounts_in_message":[],"_revision_number":4},{"id":"a07221fe33949517946c10e4dbe029c294285c2b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-23 06:12:29.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9450276f3835c137626bc0c55831d284df964aa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 07:27:42.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"1b8b3d29bced0fb11c96f9457413a9b2f793cfce","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-23 07:29:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cea39b4de6602cdf3a7e16eff6796a08b45efb89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 09:16:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4c4fb9de05fe4044bce3972097b05448 : SUCCESS in 8m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/79b9e3018cdc451d88467de520c88964 : SUCCESS in 1h 08m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9a1c062920944d4dba5eb2df2ed734f8 : SUCCESS in 1h 27m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2740a542842f4a528271457361bffa77 : SUCCESS in 1h 28m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c094cca65404c6c81629f8ccb3073dc : SUCCESS in 1h 12m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3dbeee1ceeeb4712b746b5ab8157de62 : SUCCESS in 1h 07m 39s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5b7cd0fecb2d4bd795b525db27f00c54 : SUCCESS in 1h 00m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd1d0f0da0134d27b610ecb140d28d07 : SUCCESS in 53m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ca222e6b7494868a7b3b4857d669db4 : SUCCESS in 4m 11s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/f7bcd7212b614dd9b5e3e52751a5ee92 : SUCCESS in 43m 11s","accounts_in_message":[],"_revision_number":5},{"id":"0bc464a54b2a440e759e7716b69c74cd597d19ee","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-23 11:15:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"02e48e0f258433cd2b8880c370e6490e59ffa54d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-23 11:27:54.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"6dfb29a52b31e86087055370847233948fd66df1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 13:15:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/987e5a9cfafb4b3e9a63002fcde00fd3 : SUCCESS in 6m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/37052b6c980542dfa2c9c2e8a19e7f38 : SUCCESS in 1h 21m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8f31e20f50724154a186a45164c90aa4 : SUCCESS in 1h 54m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e70cca96fd83422f9421d765ceb4b69e : SUCCESS in 1h 25m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c957768be894044b53ddc1105c77054 : SUCCESS in 1h 10m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4d4ab020717d47cb81e98188464ea564 : SUCCESS in 1h 25m 24s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/06e74a488cb8499a8c9897ef357c9df7 : FAILURE in 40m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63f2b46a3f3f46089029b31163655e75 : SUCCESS in 54m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0327239dd1c4352bf3191184063f2df : SUCCESS in 5m 17s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/1f1e7bddae1a43e9a4bc8e531d3078c1 : SUCCESS in 39m 55s","accounts_in_message":[],"_revision_number":6},{"id":"e7e28b64099cb66bf96c0c533bfd2aa35eb49cb1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-23 15:37:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"20c3c4a37e80fb684cb6adc6068cc4b0e4306206","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 17:15:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ac270d53ac144bdaa36f389ec0ec8e95 : SUCCESS in 8m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8ca86dec57944015840bad9fc3a6f448 : SUCCESS in 1h 18m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1e4ccc5f705a436ababb57f55db29a72 : SUCCESS in 1h 25m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1d3888af73a247edbc2d85302700f59e : SUCCESS in 1h 26m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e488f2dfaa974880b9ca108f3093bfd8 : SUCCESS in 1h 01m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e28f912f6d8d4ab0a3765d2307a18dea : SUCCESS in 1h 10m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4c7b1834e5f84d8fbeebe7d3e6d68569 : FAILURE in 44m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2568384bebd419a9308e16422f5e23c : SUCCESS in 43m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab11c748e5b47f1b09382d8ac90ac08 : SUCCESS in 3m 31s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/9054fdce020e411b9de2d0ee1fb5a671 : SUCCESS in 45m 50s","accounts_in_message":[],"_revision_number":7},{"id":"af40c47467f5d32acac0319d3924a99f1b2f1b51","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-09-02 13:17:01.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"c316dd58971a805066995118346971f0ce8ba185","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-04 08:36:59.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"373bb1117dac1502079aeacfbf6b2bdb675bb263","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 08:37:11.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a6486e2e3ef1bb44bd659a4aefeb31639aa249c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 10:36:55.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9101524d1d5a46fc871b881e0467532d : FAILURE in 7m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/68604c332c7f431a93d536e8b1d4ead0 : SUCCESS in 1h 20m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/28eec25f12de416c80a892a6c9039c75 : SUCCESS in 1h 40m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d2e690da48534ab698f8c15ab257dff2 : RETRY_LIMIT in 5m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3f162449afa4f8aa90783b87b386515 : SUCCESS in 1h 08m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/47fecab979b440b784f77836602e7fbe : SUCCESS in 1h 21m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fb7824a169f422c990939f1446de3af : SUCCESS in 51m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6178402b3ed470d9866ef04b920657f : SUCCESS in 3m 38s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/949963a38d9045c69ebc122ba937968d : FAILURE in 6m 40s","accounts_in_message":[],"_revision_number":7},{"id":"79b249dbd131775079a8e1038bf531f08bcdf3b5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-04 12:32:25.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c372884075ae439a63d06eda75c81a5230d88f19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:26:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/225da7fa77e547e28da06e761cbb1d82 : SUCCESS in 6m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/cc4c03d540754cb4a383746e26f62294 : SUCCESS in 1h 23m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1d748a9f9a0f43d98fc0c93d4bea9f90 : SUCCESS in 1h 47m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6611905ca1e64ed6a8f474a6179b6ccf : RETRY_LIMIT in 5m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61e214db3e9c4a198abf2789a34e006c : SUCCESS in 1h 05m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aefa2f8094c44a61a813a08db59979a3 : SUCCESS in 1h 06m 39s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6eede8f84eef427580d60193165b3404 : RETRY_LIMIT in 5m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dd283f77bf1421591096e90a1d25448 : SUCCESS in 1h 02m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00009513315e43b6a32512b2c0ad19d4 : SUCCESS in 4m 17s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/bc1c904276a24c2c89883d33fe75ef59 : RETRY_LIMIT in 21m 37s","accounts_in_message":[],"_revision_number":8},{"id":"9febf116ef708b8b06c3d3f85bdfc34c5d2cff41","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 15:58:48.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"e07515a86d776e6409d4a1e7de0a0987aa31c38a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 19:32:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6db4387c03884eddbb5fdba08498d290 : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/960f3627ab354d19b37db6f50d758bde : SUCCESS in 1h 08m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3895bc5389a8421e8e907ccb84913979 : SUCCESS in 1h 32m 02s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/96a398a94eb445888ed62706321fe73f : SUCCESS in 1h 23m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7201e1266ad4c56ac41fd421c387ac5 : SUCCESS in 1h 07m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7c9183526d8d45908d4cfebb1b315619 : SUCCESS in 1h 07m 12s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ce65e84b6d4546ed9650caa3c7f42ac6 : FAILURE in 31m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/521d407ef43c45acbc9eaf047289d01c : SUCCESS in 47m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04b5c5f96dfa47abaec57013defeaa30 : SUCCESS in 4m 22s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/d7d74c4171e8475c8ded2c97dd20ba2b : SUCCESS in 36m 12s","accounts_in_message":[],"_revision_number":8},{"id":"41eb3a9dd61d649394dd2e40e6ef89424444b484","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-09-06 02:06:24.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"7379c4dbcf69758c974e4333edb891b56c8a8f3e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-06 13:54:00.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"37cb595c3af38c17a20ec4055a8fc5a2659f9f7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:54:12.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"783cdb6b13e0f8365f4bd18527310612c3b09db6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:30:21.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0754db4a192b4aff8cac3db97abc7a08 : SUCCESS in 7m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/23b944c9c27641829d7097af286ae9b3 : SUCCESS in 59m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0bae62e5b1f54c7c982f4ea13ecacc74 : SUCCESS in 1h 29m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/dc418c651eb74767b11892bc46fb36d6 : SUCCESS in 1h 23m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76b286dcb50c418387dba3b53475abd6 : SUCCESS in 1h 05m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3511c92b07a44df2883ead4f286e4f25 : SUCCESS in 1h 12m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3faaa0ada804c1c88076a34251453ff : SUCCESS in 56m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ce6395b51346cda6e10b2575453a46 : SUCCESS in 3m 32s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/069a963cfa254079ab3d858befe93e3c : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":8},{"id":"86cba235bdd27435f4e09d0aeb3984fc5a419538","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:30:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"2aa0e2fefa41c671f7d45010afee781a0a6d5360","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:31:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4649ed2fe90d4532874068a9789fdf98 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"4b2724b44bb252758c0f083ca1052b39e87c9462","revisions":{"46cb218de386faa8f8d9bf72a32ad65883b3d214":{"kind":"REWORK","_number":1,"created":"2019-08-22 16:27:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/678055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/55/678055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/1"}}},"commit":{"parents":[{"commit":"4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-22 16:27:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-22 16:27:25.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nDepends-On: https://review.opendev.org/678025/\nChange-Id: I6f129940e55130c289d94138171cee54dbd28fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/46cb218de386faa8f8d9bf72a32ad65883b3d214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/46cb218de386faa8f8d9bf72a32ad65883b3d214"}]},"branch":"refs/heads/master"},"f85f0374cb3324d50047f5b0de7f462a72315cb3":{"kind":"REWORK","_number":2,"created":"2019-08-22 17:40:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/678055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/55/678055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/2"}}},"commit":{"parents":[{"commit":"4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-22 16:27:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-22 17:40:33.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nDepends-On: https://review.opendev.org/678025/\nChange-Id: I6f129940e55130c289d94138171cee54dbd28fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f85f0374cb3324d50047f5b0de7f462a72315cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f85f0374cb3324d50047f5b0de7f462a72315cb3"}]},"branch":"refs/heads/master"},"0ab1f51c06c1d1c4e723185afcbb5e56e9f357f7":{"kind":"REWORK","_number":3,"created":"2019-08-22 20:29:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/678055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/55/678055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/3"}}},"commit":{"parents":[{"commit":"4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-22 16:27:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-22 20:29:44.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nDepends-On: https://review.opendev.org/678025/\nChange-Id: I6f129940e55130c289d94138171cee54dbd28fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0ab1f51c06c1d1c4e723185afcbb5e56e9f357f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0ab1f51c06c1d1c4e723185afcbb5e56e9f357f7"}]},"branch":"refs/heads/master"},"345b3abaad242ecd5ec2b4f8396fe3704a0273a9":{"kind":"REWORK","_number":4,"created":"2019-08-23 01:59:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/678055/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/55/678055/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/4"}}},"commit":{"parents":[{"commit":"4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-22 16:27:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-23 01:59:14.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nDepends-On: https://review.opendev.org/678025/\nChange-Id: I6f129940e55130c289d94138171cee54dbd28fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/345b3abaad242ecd5ec2b4f8396fe3704a0273a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/345b3abaad242ecd5ec2b4f8396fe3704a0273a9"}]},"branch":"refs/heads/master"},"c140b0c1a1a8d8a89afcc2331d59fc1e04c3f07e":{"kind":"REWORK","_number":5,"created":"2019-08-23 07:29:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/678055/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/55/678055/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/5"}}},"commit":{"parents":[{"commit":"4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-22 16:27:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-23 07:28:54.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nDepends-On: https://review.opendev.org/678025/\nChange-Id: I6f129940e55130c289d94138171cee54dbd28fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c140b0c1a1a8d8a89afcc2331d59fc1e04c3f07e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c140b0c1a1a8d8a89afcc2331d59fc1e04c3f07e"}]},"branch":"refs/heads/master"},"e827234f740ee18d554c312fa6196ef113a149aa":{"kind":"REWORK","_number":6,"created":"2019-08-23 11:15:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/678055/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/55/678055/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/6"}}},"commit":{"parents":[{"commit":"4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-22 16:27:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-23 11:13:18.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nDepends-On: https://review.opendev.org/678025/\nChange-Id: I6f129940e55130c289d94138171cee54dbd28fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e827234f740ee18d554c312fa6196ef113a149aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e827234f740ee18d554c312fa6196ef113a149aa"}]},"branch":"refs/heads/master"},"8a2bace61d8e2ed17d17b1da636727726956f3fe":{"kind":"REWORK","_number":7,"created":"2019-08-23 15:37:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/678055/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/55/678055/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/7"}}},"commit":{"parents":[{"commit":"4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-22 16:27:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-23 15:37:29.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nDepends-On: https://review.opendev.org/678025/\nChange-Id: I6f129940e55130c289d94138171cee54dbd28fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8a2bace61d8e2ed17d17b1da636727726956f3fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8a2bace61d8e2ed17d17b1da636727726956f3fe"}]},"branch":"refs/heads/master"},"4b2724b44bb252758c0f083ca1052b39e87c9462":{"kind":"REWORK","_number":8,"created":"2019-09-04 12:32:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/678055/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/55/678055/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/55/678055/8"}}},"commit":{"parents":[{"commit":"4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4c8dd7f0f3bc9423e11bcef325a0e1e5ff3e5c63"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-22 16:27:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-04 12:32:08.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nDepends-On: https://review.opendev.org/678025/\nChange-Id: I6f129940e55130c289d94138171cee54dbd28fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4b2724b44bb252758c0f083ca1052b39e87c9462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4b2724b44bb252758c0f083ca1052b39e87c9462"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
