)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~688461","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fpike~Ida90363b531d1ae61a6322c11495db7bb9335650","project":"openstack/openstack-ansible-os_glance","branch":"stable/pike","topic":"goodbye_42.3","hashtags":[],"change_id":"Ida90363b531d1ae61a6322c11495db7bb9335650","subject":"Remove openSUSE Leap 42.3  jobs","status":"MERGED","created":"2019-10-14 16:57:22.000000000","updated":"2019-10-14 20:27:48.000000000","submitted":"2019-10-14 20:27:48.000000000","submitter":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688461-1571084868665-0f5a3271","meta_rev_id":"9a8eaaa793626e23146a473b73217f408cb59070","_number":688461,"virtual_id_number":688461,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-10-14 20:27:43.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-14 17:34:37.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-10-14 17:30:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-10-14 17:30:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 17:30:51.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-14 17:34:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-14 20:27:43.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"934f0a04938415e2bbe31d4e20be2d062dff4960","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-14 16:57:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75b0ea6428b85af8b184522fb1c730dae6298cde","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-14 17:30:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3ec825f4418901595a7d02683c097f1399ea954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:34:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/565135b27dda434595ebe880462446f3 : SUCCESS in 4m 47s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bfd7a9dcfeb74c7c8086aed0dd0dc536 : SUCCESS in 8m 12s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c0006955dcda4a1ebc6ebb62318eb3e3 : FAILURE in 7m 44s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/08505225e776489284005ee195b3ae03 : SUCCESS in 25m 02s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/a64b742d473f41c38460f714e4002dd2 : SUCCESS in 24m 01s\n- openstack-ansible-v2_api_only https://zuul.opendev.org/t/openstack/build/79d5154754054c14b4c11b7038a3647a : SUCCESS in 21m 34s\n- openstack-ansible-v2_only_registry_enabled https://zuul.opendev.org/t/openstack/build/03d4ec82a8734f418e9155e76e5e8d4d : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b23aeeca1f8868f7be1b3ad1933f50c68ac05334","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-14 20:27:43.000000000","message":"Patch Set 1: Verified+2\n\nThe opensuse-423 nodeset is being removed.  This change is being force-merged to correct the resulting configuration error.","accounts_in_message":[],"_revision_number":1},{"id":"1d3c34485ef987abd27671f475ee640c7ddac4c4","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-14 20:27:48.000000000","message":"Change has been successfully merged by Clark Boylan","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf5c79a760e711e7fd47ed960b2577e2104cf768","revisions":{"cf5c79a760e711e7fd47ed960b2577e2104cf768":{"kind":"REWORK","_number":1,"created":"2019-10-14 16:57:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/61/688461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/61/688461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/688461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/688461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/688461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/688461/1"}}},"commit":{"parents":[{"commit":"d903264841508ec99800356c9cf43ec7800a33ad","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d903264841508ec99800356c9cf43ec7800a33ad"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 16:57:08.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 16:57:18.000000000","tz":120},"subject":"Remove openSUSE Leap 42.3  jobs","message":"Remove openSUSE Leap 42.3  jobs\n\nopenSUSE Leap 42.3 is EOL since June 2019, let\u0027s remove remaining jobs so\nthat we can remove Leap 42.3 from Opendev infrastructure.\n\nChange-Id: Ida90363b531d1ae61a6322c11495db7bb9335650\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cf5c79a760e711e7fd47ed960b2577e2104cf768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cf5c79a760e711e7fd47ed960b2577e2104cf768"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
