)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~688462","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fqueens~I8b4b1d48b1d8745f65db572ccb17543e99a81ad4","project":"openstack/openstack-ansible-os_glance","branch":"stable/queens","topic":"goodbye_42.3","hashtags":[],"change_id":"I8b4b1d48b1d8745f65db572ccb17543e99a81ad4","subject":"Remove openSUSE Leap 42.3  jobs","status":"MERGED","created":"2019-10-14 16:57:43.000000000","updated":"2019-10-14 18:59:28.000000000","submitted":"2019-10-14 18:59:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688462-1571079568508-29eb8155","meta_rev_id":"23f329b6d3d808cd17f35a7e95f4de86316a1e7a","_number":688462,"virtual_id_number":688462,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 18:59:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-10-14 17:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-10-14 17:30:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 17:30:45.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-14 18:59:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a83b1f1e13b23261ef68efd9aa941546ca6fad42","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-14 16:57:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d379c0acd688333eb3fb1e536b0275b935d4dc22","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-14 17:30:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94bafa1aa265f1ebf55f15a892fe1a045262c59b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:46:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/da1db83b058e483b96d657c92c8695a0 : SUCCESS in 5m 45s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8e20ccc2042a4a6fb7a15406b675c173 : SUCCESS in 6m 21s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/72a1e63d82a6405ea9682f518ae6e358 : SUCCESS in 45m 50s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/7c774b8c5de044e8a7254c3f8ffc156e : SUCCESS in 31m 55s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/b4331c26f1834707be4fd3913d5caeb3 : SUCCESS in 33m 16s\n- openstack-ansible-v1_api_enabled https://zuul.opendev.org/t/openstack/build/960325479d4d44b3ab265ad29f67f445 : SUCCESS in 29m 19s\n- openstack-ansible-v2_registry_enabled https://zuul.opendev.org/t/openstack/build/9e4202344fe14ff5b3ba4ab8fe4c656c : SUCCESS in 26m 50s","accounts_in_message":[],"_revision_number":1},{"id":"380e3d47c35151fd5b22746924b1fb24f3b6d7a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:54:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed616f16579e6215850cd25743eb59afd29c36e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 18:59:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f9a04e09cb7b0567f4a70b1f549403b67e3e8673","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 18:59:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8a19e7aa57d849618bd0cbaa8c8fbf87 : SUCCESS in 3m 37s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2cc56dfb2d754f07b6fb2c030b103e3c : SUCCESS in 6m 06s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1c4d3c6d369342519e50eaaa8c251401 : SUCCESS in 57m 26s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ea765051d4e74ef89927d068c08c0ad4 : SUCCESS in 29m 02s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/caff85d80b394d64a72abfb8fd285920 : SUCCESS in 30m 41s\n- openstack-ansible-v1_api_enabled https://zuul.opendev.org/t/openstack/build/5ced985e3f034e7e800b1843ecad1879 : SUCCESS in 27m 34s\n- openstack-ansible-v2_registry_enabled https://zuul.opendev.org/t/openstack/build/12f1336836e84afea50d4686c68c91ff : SUCCESS in 27m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6770101a2bdc5ad52deb4bd5727878af3d918565","revisions":{"6770101a2bdc5ad52deb4bd5727878af3d918565":{"kind":"REWORK","_number":1,"created":"2019-10-14 16:57:43.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/62/688462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/62/688462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/688462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/688462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/688462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/688462/1"}}},"commit":{"parents":[{"commit":"7d9fb1d70c8b87c6c0a4a0abd9f8d734494c8f3f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7d9fb1d70c8b87c6c0a4a0abd9f8d734494c8f3f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 16:57:37.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 16:57:37.000000000","tz":120},"subject":"Remove openSUSE Leap 42.3  jobs","message":"Remove openSUSE Leap 42.3  jobs\n\nopenSUSE Leap 42.3 is EOL since June 2019, let\u0027s remove remaining jobs so\nthat we can remove Leap 42.3 from Opendev infrastructure.\n\nChange-Id: I8b4b1d48b1d8745f65db572ccb17543e99a81ad4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6770101a2bdc5ad52deb4bd5727878af3d918565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6770101a2bdc5ad52deb4bd5727878af3d918565"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
