)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~688944","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ib8def3441b6c3a634cbb53a44d295b71586fed8a","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"Ib8def3441b6c3a634cbb53a44d295b71586fed8a","subject":"Drop duplicated keys","status":"MERGED","created":"2019-10-16 14:32:33.000000000","updated":"2019-10-18 18:37:47.000000000","submitted":"2019-10-18 18:36:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688944-1571423790987-f62c7b6a","meta_rev_id":"475f75171bb568534488a6ee16c1d85f43b54376","_number":688944,"virtual_id_number":688944,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"date":"2019-10-18 15:38:54.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 18:36:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2019-10-17 13:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-16 20:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2019-10-17 13:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 20:24:34.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-17 13:37:31.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-10-18 18:36:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88dec0219dabe1d90dbe65765b368a14bb2b02a8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-16 14:32:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0e4e56578dc82ac2f8a3cdf86bf0f5239c7a1ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:21:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3a78d8ea411a4f6387ee92d9cff74842 : SUCCESS in 7m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3e67002690964a1a8176f24c48c8fa9b : SUCCESS in 1h 28m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2f1e3f87b7ab486eb025a7209b5ee8de : SUCCESS in 1h 41m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fb9d76f75cde450dafbe537f982acbec : SUCCESS in 1h 35m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f535950d94b4011a1cc99a19507f012 : SUCCESS in 1h 05m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/58cdcffff59440a78dc0c0041d4e1a52 : SUCCESS in 1h 13m 50s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fee890a4c7fb45e1a041abce815a7bdb : FAILURE in 44m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0af92c94145f405f9259ebe1e1f6173b : SUCCESS in 52m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4226a0129c0b439ea8f8a92ebec727ac : SUCCESS in 3m 58s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/82908785eec84aae88c6bde8495ab25a : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":1},{"id":"de66e218a0f6cbed5424853924bc0b5936639339","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-16 20:24:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7ddcc8159e3099fee1fc8db9eba8375fc6643290","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-10-17 13:37:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62ca76ba62ca23681f2cc34ae49a64e4b3e2a8b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 13:37:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"774db3e196ea2df51bc8800d7b7d55cb2bbaaf62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 15:22:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d6af6657c93f4a899df08b497d6bd5cc : SUCCESS in 5m 52s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2f30ca13e84a48f4879f981a19d28ae5 : SUCCESS in 1h 20m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4c68c42ad499445088aa978f48252bb8 : FAILURE in 1h 39m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/88d57b59e073402fb421abf75e25d616 : FAILURE in 50m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aed93f6ae0c243e59d187d5d422d5583 : SUCCESS in 1h 31m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1f71ae5f01154cb4bf842afe95d472b3 : FAILURE in 1h 02m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/609301673df34712afe2c2531555217b : SUCCESS in 55m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1c773ffeba24c678bb8a707b7651df7 : SUCCESS in 2m 48s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/eefee0acbff943a3b62f3f5f1d641b9b : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9ed91472eba95bef613560264484ae90bc57f5b4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 21:10:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6bfb9e5596ea7d16261567a247222572b82c3525","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 22:59:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/074d2e753bb34ea289434a00182dbf2d : SUCCESS in 7m 11s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f4b11d365ef7422ea5739267a197ebef : SUCCESS in 1h 14m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e7f574041c61497880321bc466911f68 : SUCCESS in 1h 46m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4d54ea0e69be45d9ac2105ae88d69270 : FAILURE in 57m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9532fb50568a482b9db2faf3e47e04a4 : SUCCESS in 1h 11m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1e52c67069294349b911ad1696ac9c7e : SUCCESS in 1h 07m 23s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e020dfa5bf8d4fa498860115d5b524be : FAILURE in 43m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e3bc3cbc3204f689722ac3fffe6b311 : SUCCESS in 49m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3530fdfc508348abbd8ab60a0b33aee4 : SUCCESS in 4m 23s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/72843feb386140a38aef720777a2f4ca : SUCCESS in 38m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3f1c14b30162ee374486c74cb70b2af9a69fa198","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-10-18 12:46:33.000000000","message":"Patch Set 1:\n\nNo provider of +genisoimage : is that normal for opensuse15?","accounts_in_message":[],"_revision_number":1},{"id":"df3ec099ac14680c1153ce326424f776befd2926","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-10-18 12:46:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e12d4e21f7c5e9c9a420b33b9172fa7726f1524b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 14:28:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0f3c75a25ab9482d981b884fbe2933b0 : SUCCESS in 4m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/95f47724f6fc4d51b56a3e47a210a8cf : SUCCESS in 1h 27m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5f13c8c42b9248e496ae3152d8537ba4 : SUCCESS in 1h 35m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1e55db8317294c92ae597f6734bfa546 : FAILURE in 49m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e0c30cf22324b8fac8f7c204ac9d71d : SUCCESS in 1h 14m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/83e37c1c12a64904a102df2d61585de8 : SUCCESS in 1h 08m 18s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d3323a3d21c34fcc86982186746542df : FAILURE in 42m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b48e7f00e924a728c366d335a946347 : SUCCESS in 47m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e2cba35cd747269aa10054a4cdff16 : SUCCESS in 4m 07s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/d07f0e940e6d4884a718801b34e2ae4b : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6a103160704f2d436524fd3ee486fdef1d8cc5cc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-18 15:38:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"155f670a4149870f838fe17f1098802b7d3a80a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 17:08:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/23d33af1cedb4ba2ae93e5a67186806d : SUCCESS in 6m 05s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9de55f40e6a945aaa5c1dfa0a3217107 : SUCCESS in 1h 21m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e6281a1e57d1411db447a2294f46f839 : SUCCESS in 1h 27m 02s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/555ccdbb68904dc0bb57ae28ed099599 : SUCCESS in 1h 20m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb99b7dc3ef34baaa3b8f1cffac360df : SUCCESS in 1h 19m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5dd1fef716ab4b2381ed4cf8c7fb8112 : SUCCESS in 1h 14m 29s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e82257df3acc445d98b709d5d73825b9 : FAILURE in 34m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61a80fc61e5448ad8854daf34dcf5715 : SUCCESS in 55m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08bc3260bf7d4bb2b833d60d5b2ebe83 : SUCCESS in 3m 42s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/61cf0f0e2c184f779e42dda77838385f : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f17b5900f67e83a74564157b8ced3ccc187984fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 17:08:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9adda64281ba9c498927e8613e1c3212448d23f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:36:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e046239379ff43460eb2629c25ae9a76ae9658f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:36:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/af6cef4b2c364f37abc389839d792d8b : SUCCESS in 6m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/210863b605084b8fb1d028b7087fb0e4 : SUCCESS in 1h 12m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/369cc537b8264144a4ff7f7ccddfbb70 : SUCCESS in 1h 26m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3e890128677b463d9d62877e6c890355 : SUCCESS in 1h 18m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b37804ac721041439c945745cf218e3f : SUCCESS in 1h 07m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9986c3a37fa940f3a0251e9de80d3e14 : SUCCESS in 1h 04m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a3513f988404b37b206083ea18e382e : SUCCESS in 49m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5441994d14634d198c7f584868ee3a39 : SUCCESS in 3m 20s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/b0d30c1d3da9410f937802873a889e8f : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":1},{"id":"475f75171bb568534488a6ee16c1d85f43b54376","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:37:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d952c928c84de69be7ad3579d9fe56 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e966b4f4a48865b38385245867839b765b5feb8","revisions":{"4e966b4f4a48865b38385245867839b765b5feb8":{"kind":"REWORK","_number":1,"created":"2019-10-16 14:32:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/44/688944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/44/688944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/688944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/688944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/688944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/688944/1"}}},"commit":{"parents":[{"commit":"b1e8e2337aa1dc09ba799e9b60d45e255bd47c91","subject":"Merge \"Turn the filesystem directory to be configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b1e8e2337aa1dc09ba799e9b60d45e255bd47c91"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-16 14:32:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-16 14:32:20.000000000","tz":180},"subject":"Drop duplicated keys","message":"Drop duplicated keys\n\nChange-Id: Ib8def3441b6c3a634cbb53a44d295b71586fed8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4e966b4f4a48865b38385245867839b765b5feb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4e966b4f4a48865b38385245867839b765b5feb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
