)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~705680","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I71357a585a053d0b4e9316cc69c681caee2259da","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"I71357a585a053d0b4e9316cc69c681caee2259da","subject":"Reduce number of processes on small systems","status":"MERGED","created":"2020-02-04 13:38:48.000000000","updated":"2020-10-07 08:47:53.000000000","submitted":"2020-10-07 08:46:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705680-1602060379933-14270ba4","meta_rev_id":"cc8aa932b1f2b066e3d058c0344bec7e70066357","_number":705680,"virtual_id_number":705680,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-07 08:46:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-07 07:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-10-06 10:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-07 07:20:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 10:30:08.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-07 07:20:00.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-07 08:46:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0eea98f847a0f1ba80286c963382543259f6e483","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-04 13:38:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9457c87d0ff7a838bbcc89179c96ea9c415a2f4f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-02-04 15:08:12.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f1d3d08037da4a0c71f0d23798824f4f9e93ddaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 16:28:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f42772622fbf48ab87831e0b9e454716 : SUCCESS in 7m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/778ddec11a1c40739dded24522303ddb : SUCCESS in 1h 11m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0ff1e17997904ee490c0df1893faef0f : SUCCESS in 1h 42m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/160b3c0810314cc1a01c829cc1caf62d : SUCCESS in 1h 21m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/935915a10cf94b8ab1b96a8010f3b921 : SUCCESS in 1h 13m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/46543f8a701447d6a90a5c0e8963c3b6 : SUCCESS in 2h 44m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db6deb702b1e4cd786153ea020748b95 : SUCCESS in 1h 46m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3b44ec05997b425a94771adc532a2f7a : SUCCESS in 1h 09m 02s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/60935bd6e9534e628eaea32e8127c320 : FAILURE in 36m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24f5918ca1e04267b6c8b601415b661e : SUCCESS in 59m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bababb23ba14e8186674acef6cf9cd2 : SUCCESS in 4m 52s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/cba6123dd3a944eebbd4c7e32513418e : FAILURE in 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a8e474d5dc91d359a70ae243108bd42482007863","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-06 10:30:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6baebda5682ad88de59ad73b0e6449e384d015dd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-06 10:30:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5cc52050d4914670ffc18bd48c7d9c543d501899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 12:51:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a6ce502bc67f4427ad7b8c808a723dd6 : SUCCESS in 9m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dacd40d3ffd54baf813eb47179c2bd06 : SUCCESS in 1h 07m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ff6813e637e0481384820037a36bb3f2 : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5260cab6ae2c47e6967998193924e1d9 : SUCCESS in 1h 14m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d76317ffc8384dfdba13a3af270ff6a0 : SUCCESS in 1h 56m 41s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/832ed4c20b1d4c55a20275df83a90a24 : SUCCESS in 2h 07m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1e4960e80424291a41d8e37ce263574 : SUCCESS in 1h 37m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/39a3a62aec6247f58fe707c91c578bcd : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8be8b29f34641a7bf046fa8793f0040 : SUCCESS in 54m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d93d151ae3a4402e9866baa61e1207a6 : SUCCESS in 8m 21s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/4d3b413243d54ee48ae5f81a8ce67600 : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":2},{"id":"0f2cf756945b973158740e9b83cc57e12f8dbdca","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-07 07:20:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c783e04c4636373719516205a7aa85a05a705ca5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 07:20:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"abc6c6221046c99d31659e9b48f328e89e0d3e93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 08:46:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f495981bab364bca92f55266b641cd37 : SUCCESS in 8m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/02cdf2cf2b264bf883c75669e114cdcd : SUCCESS in 1h 06m 57s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/67bd9deebfeb4c189b59b3ab60b1a193 : SUCCESS in 1h 19m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a061f41ee9841dc81961163eb3aaf02 : SUCCESS in 1h 12m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a58ba857e2d4ba7aac5084a8d00335c : SUCCESS in 1h 05m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/152fa44e3a774035bebf66261c8a6101 : SUCCESS in 1h 03m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d2ae43aa5b2458b834e2ad454899287 : SUCCESS in 52m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a26b81e80904a41bb557e3c54657c70 : SUCCESS in 6m 15s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/ab46fdb618154d989b8de707e00d3af6 : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":2},{"id":"1dbb88e51103ab524a36a23fc46e34b5a295e612","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 08:46:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cc8aa932b1f2b066e3d058c0344bec7e70066357","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 08:47:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca158bb0d58b4ce3ba627a9ca8c160e5 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cbbd0dee61f939a15216bcf8f94d9d696ad03cc4","revisions":{"c8c9b5948e314e57b71a8d5e318a2977d96d4633":{"kind":"REWORK","_number":1,"created":"2020-02-04 13:38:48.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/80/705680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/80/705680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/705680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/705680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/705680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/705680/1"}}},"commit":{"parents":[{"commit":"61d1a602e087912fe99552e73fdb0d289c8a7b9c","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/61d1a602e087912fe99552e73fdb0d289c8a7b9c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-04 13:13:55.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-04 13:33:09.000000000","tz":0},"subject":"Reduce number of processes on small systems","message":"Reduce number of processes on small systems\n\nEven the most modest 4C/8T system would run with the maximum 16 processes\ndue to the calculation being VCPU*2.\n\nThis patch makes it clearer that glance_api_threads is the number of\nphysical CPU, and halves glance_wsgi_processes to physical CPU*2 for\na hyperthreaded CPU.\n\nChange-Id: I71357a585a053d0b4e9316cc69c681caee2259da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c8c9b5948e314e57b71a8d5e318a2977d96d4633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c8c9b5948e314e57b71a8d5e318a2977d96d4633"}]},"branch":"refs/heads/master"},"cbbd0dee61f939a15216bcf8f94d9d696ad03cc4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-06 10:30:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/80/705680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/80/705680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/705680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/705680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/705680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/705680/2"}}},"commit":{"parents":[{"commit":"3a9ebca90eb46bd988d84ef8935db912a0109e6b","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3a9ebca90eb46bd988d84ef8935db912a0109e6b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-04 13:13:55.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-06 10:30:32.000000000","tz":0},"subject":"Reduce number of processes on small systems","message":"Reduce number of processes on small systems\n\nEven the most modest 4C/8T system would run with the maximum 16 processes\ndue to the calculation being VCPU*2.\n\nThis patch makes it clearer that glance_api_threads is the number of\nphysical CPU, and halves glance_wsgi_processes to physical CPU*2 for\na hyperthreaded CPU.\n\nChange-Id: I71357a585a053d0b4e9316cc69c681caee2259da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cbbd0dee61f939a15216bcf8f94d9d696ad03cc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cbbd0dee61f939a15216bcf8f94d9d696ad03cc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
