)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~712061","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I3f365684542b390ea02c08ab56f76a447f65a814","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"I3f365684542b390ea02c08ab56f76a447f65a814","subject":"Fix glance policy.json deployment","status":"MERGED","created":"2020-03-10 12:25:55.000000000","updated":"2020-03-10 21:07:03.000000000","submitted":"2020-03-10 21:03:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712061-1583874226876-eb552905","meta_rev_id":"4154369d89e288995c62df90949c09483e89b5a6","_number":712061,"virtual_id_number":712061,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 21:03:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2020-03-10 19:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-10 18:21:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2020-03-10 19:24:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-10 18:21:06.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-10 19:24:51.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-03-10 21:03:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53de7078271864b7e9fbef182d627aecbd531ef3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-10 12:25:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad851d3880bb2a90dd397513efbe7b10eb47d754","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-10 12:27:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64d28b1048c19d0b76a82f16884eea51d19d58b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 14:17:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1b7d2fda355245a9840dc2bcb8f07d45 : SUCCESS in 8m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/153f10335e0e4a7eb3eae17b639f61e7 : SUCCESS in 1h 20m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c80f647e9cf049b883198690b8bd2cfc : SUCCESS in 1h 44m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f46ee061d49848489ddf482ae49e614e : SUCCESS in 1h 29m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17b0bb08e68f4547bdb33d53019907be : SUCCESS in 1h 31m 28s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a78e42ca79b44c3c9fde3b64c8a88f9f : FAILURE in 1h 35m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e1f3f2d41a64c6586280d59a2f73a3b : FAILURE in 1h 03m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9b4a7fb48c78488ab58f11ad44b04283 : SUCCESS in 1h 15m 05s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a1e075579a4346f681949f6615d16dc0 : FAILURE in 26m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73db39310dfa4c07b9f0a9915efe813e : SUCCESS in 52m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5674b069c10e41cf980e11c59448b596 : SUCCESS in 8m 41s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/f9541a1f06ae42d498d5ffd654eb2a53 : SUCCESS in 35m 03s","accounts_in_message":[],"_revision_number":2},{"id":"4bec55b2dd1f2b05bfe35c6fede7cf72694c0b36","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-10 18:21:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3479f34f39fc8ce69d95c0a897cfa3f32beef352","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-10 19:24:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"49400648ee0d3c3f0e753acfc946c697058f05a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 19:25:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a016bc882da6f2393a15801f5bfe523a5a8d15d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 21:03:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"89edcbb59c012134e28d438c825ce5bae1a7e049","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 21:03:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d06d1689c67c4fef99432992725e259d : SUCCESS in 6m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5851650ef5fc41bd89afc6cfd8115d80 : SUCCESS in 1h 04m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2c115a4ece434cbf8243987b9f83564f : SUCCESS in 1h 33m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8959a61a51cd4ebe864f4dcd09189581 : SUCCESS in 1h 29m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c482d1fcd5d4396ab345c29a554ca7c : SUCCESS in 1h 10m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/83fd9d0335ec4bb79732610f323dd592 : SUCCESS in 1h 09m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cb4ada33e9b4e34aeeec0e6e3bed77d : SUCCESS in 1h 20m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf46fb5fa47942fb9d3f45b5875ce6f1 : SUCCESS in 4m 41s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/e1fcd4870a434696b35a69fcd4565a0b : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":2},{"id":"4154369d89e288995c62df90949c09483e89b5a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 21:07:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b661d96b474b78a68555b1df911d73 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0f7971c9cba754ae11de5ea93e61e5c441cb7cc8","revisions":{"f4c519729cf2bb43b31e6cda0282c9f6d4913a8d":{"kind":"REWORK","_number":1,"created":"2020-03-10 12:25:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/61/712061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/61/712061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/712061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/712061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/712061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/712061/1"}}},"commit":{"parents":[{"commit":"c9d88043facb96cdf7f268d5dac45db2652088a7","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c9d88043facb96cdf7f268d5dac45db2652088a7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-03-10 12:25:44.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-03-10 12:25:44.000000000","tz":120},"subject":"Fix glance policy.json deployment","message":"Fix glance policy.json deployment\n\nGlance has dropped default policy.json [1] which\nwas used by \"smart sources\". We are fixing this by setting content\nto empty dict, that way the only content deployed will be the one\nprovided by overrides, so that won\u0027t change current behaviour.\nAdditionally `glance_policy_content` has been introduced, which\neventually is going to replace `glance_policy_overrides` in the future.\n\n[1] https://opendev.org/openstack/glance/commit/dd1975bd3ef690d8bf1272923f0eac7beab48c6d\n\nChange-Id: I3f365684542b390ea02c08ab56f76a447f65a814\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f4c519729cf2bb43b31e6cda0282c9f6d4913a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f4c519729cf2bb43b31e6cda0282c9f6d4913a8d"}]},"branch":"refs/heads/master"},"0f7971c9cba754ae11de5ea93e61e5c441cb7cc8":{"kind":"REWORK","_number":2,"created":"2020-03-10 12:27:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/61/712061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/61/712061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/712061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/712061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/712061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/61/712061/2"}}},"commit":{"parents":[{"commit":"c9d88043facb96cdf7f268d5dac45db2652088a7","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c9d88043facb96cdf7f268d5dac45db2652088a7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-03-10 12:25:44.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-03-10 12:27:33.000000000","tz":120},"subject":"Fix glance policy.json deployment","message":"Fix glance policy.json deployment\n\nGlance has dropped default policy.json [1] which\nwas used by \"smart sources\". We are fixing this by setting content\nto empty dict, that way the only content deployed will be the one\nprovided by overrides, so that won\u0027t change current behaviour.\nAdditionally `glance_policy_content` has been introduced, which\neventually is going to replace `glance_policy_overrides` in the future.\n\n[1] https://opendev.org/openstack/glance/commit/dd1975bd3ef690d8bf1272923f0eac7beab48c6d\n\nChange-Id: I3f365684542b390ea02c08ab56f76a447f65a814\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0f7971c9cba754ae11de5ea93e61e5c441cb7cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0f7971c9cba754ae11de5ea93e61e5c441cb7cc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
