)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~712688","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I19b74c3bc5119953256d3d8f2a98cb5f23787755","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"memcached-refactor","hashtags":[],"change_id":"I19b74c3bc5119953256d3d8f2a98cb5f23787755","subject":"Refactor memcached_servers","status":"MERGED","created":"2020-03-12 14:04:22.000000000","updated":"2020-03-14 14:51:29.000000000","submitted":"2020-03-14 14:50:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712688-1584197402854-eb552905","meta_rev_id":"e2c977423fa782c8008ae20a01cc597da211d36e","_number":712688,"virtual_id_number":712688,"owner":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-14 14:50:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2020-03-14 13:19:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-14 07:59:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":1,"date":"2020-03-14 13:19:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-14 07:59:51.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-14 13:19:22.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-03-14 14:50:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30e95583755132764eb6cf354cdb8b4de19cf76d","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-03-12 14:04:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"440adf5aac1fc8a5f757d2478d93435a87047446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 15:59:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/20496bf5f7a242789c3e68b85e931b9e : SUCCESS in 10m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fe11baebc4074cb48db2a52d31bc4683 : SUCCESS in 1h 06m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8e01bdfdc21a472fa08f0c486571c60a : FAILURE in 26m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a6c02e5711764cc9ac6792709b74e209 : SUCCESS in 1h 35m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/055c7b2b23e04f7e9b953a308839e54c : SUCCESS in 1h 24m 24s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/437d9b77f9944c49992ff8e0fe76cde6 : FAILURE in 9m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d5aabfeec4048ccb386a9280a779ad7 : SUCCESS in 1h 49m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/016850e529944dca8594c7d8127f551d : FAILURE in 26m 24s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/dbc706af552c4abd93bc87fa22be9c82 : FAILURE in 39m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37f1326be4f44507a7d8bb311aa54ecd : SUCCESS in 55m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ed98731f57e46af8e6f249686a267ac : SUCCESS in 5m 52s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/0c11495d440944da8f8c992b87b9a3c5 : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ba5c9ca3fc1625d4baeba07c44801aacf446a08a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-12 22:32:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3335cb3ebccfe8b503ec24e60b2ed10657cacaef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 00:37:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c1505317a1c84eaea401955487473bb8 : SUCCESS in 6m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ea3ca8c680a749df9f77734acf996a37 : SUCCESS in 1h 07m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/050350c042424c73816f0d3372094ee8 : SUCCESS in 1h 28m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f1ee6140a29143f08d004a1feb9d3110 : SUCCESS in 1h 22m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cf3ba2daebf41f6af637677d2587d36 : FAILURE in 1h 11m 28s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/134c4e3c42824335b05f691b37c1c8d3 : FAILURE in 9m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25d75a63f0214f5ba1d3abb17da6d3ba : FAILURE in 1h 59m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/20c6bb859e5a404fa76fbfa93e46df0c : SUCCESS in 1h 06m 29s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7fe58a3ac49740d18df2fb82a4c06974 : FAILURE in 33m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/765f62835cff4de7ae1969bb88b13802 : SUCCESS in 1h 02m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a0ce8886f446f799dd53f58e5cbfde : SUCCESS in 5m 38s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/747b8456b85940549a6ddfb05402e0e2 : SUCCESS in 41m 20s","accounts_in_message":[],"_revision_number":1},{"id":"eafd515270f075f299579deba73a3392f94eedc4","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-03-13 17:55:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0e2706ee1ddceaea2e3a01996f6851d5836e5ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 19:44:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/343f17e8631c4e3eb7f63858e25391b5 : SUCCESS in 7m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c5af7dc07bb84081879ee56e6e839867 : SUCCESS in 1h 01m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/18eb16e969fe4fc1999b2107a7b357d6 : SUCCESS in 1h 30m 28s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3ed2e79d0b6641b4a1b8bbaa267ea6f7 : SUCCESS in 1h 23m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07d0a3a51eca47edaf40ddcaa5f4892b : SUCCESS in 1h 14m 37s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b151c5ba37234a7b9007415e510986f6 : FAILURE in 8m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc2659b03c044d63a0bd9d2de757c32c : SUCCESS in 1h 46m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dbd4258d2f4a41b6a5e435f1709fa92d : SUCCESS in 1h 09m 49s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/88ba6ed09fac40de8669eb15a9693405 : FAILURE in 28m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bd807c63adf4185913e3d341a8e0125 : SUCCESS in 59m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b6539bb2cc64af4992593d4b9adf612 : SUCCESS in 5m 09s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/2ea63fe2db3a456bbefeecc53c4a2f41 : SUCCESS in 33m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9ee61526d59ccb31dada35eca37e73f4a80bc566","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-03-13 22:15:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"059e0a5531b9b59ba22f6e531183c0b957012082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 00:14:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e413b01789cb48fca38567ba180d1e0c : SUCCESS in 10m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7aafe11e16a74b3993b7f0f1a10acbe8 : SUCCESS in 1h 20m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0899539392a147dfac7c3ee0005c7b74 : SUCCESS in 1h 54m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ce9f8b087c274601895d97306243f6cd : SUCCESS in 1h 16m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc7792cc5aba46d6be560bfee70af2fb : SUCCESS in 1h 09m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f786c6607ff54c2b90e1274f8af55a74 : FAILURE in 9m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4af54d77b51843e682ee986b221632bd : SUCCESS in 1h 45m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/985d15f57e7c4684871cd268d8e86192 : SUCCESS in 1h 04m 46s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e6bc1e1ed6c94417929651ceeea43e76 : FAILURE in 30m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/590194ce710447ed9f3233af35ee1572 : SUCCESS in 54m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b9839b1b5e40898a4b83d8ace0cd14 : SUCCESS in 5m 25s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/d8780056318f4e0e877876b45a4e056c : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":2},{"id":"aca6a8a37267c9f56bcec82a9cb21b0c8ab02d31","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-14 07:59:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d1364cd611292b8904d1ff3ead9d5cb03537cd67","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-14 13:19:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6133e0b8b39836330257a26d497f79c0ad52bd7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 13:19:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"427f2ef3480e69fafe418e6ff9b88e12bee12363","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 14:50:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bfa62ca50fe74cf18ac94725ce844407 : SUCCESS in 9m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b6ce7569c3aa40fcbbb0ae4e692e2a67 : SUCCESS in 1h 07m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/797c114752464d69b1503cc1de2a0053 : SUCCESS in 1h 28m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6e2632cf260b460698645b2aebe737f6 : SUCCESS in 1h 17m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/069aaa73d86e4778aac8cae48f527a73 : SUCCESS in 1h 09m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c84aca72cb2f4b5597b2397113cbb4f9 : SUCCESS in 1h 06m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c1d141448924a888d97c7fa6c3d5638 : SUCCESS in 1h 00m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/606e8f8f840d418187f604b00180ef45 : SUCCESS in 5m 02s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/248354e2f50e4731a4e06514fc664eee : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":2},{"id":"6de4c489309a81bf0547e4e0db99cd90be938ced","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 14:50:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bc153fb0ac59d4c2722fa3fb77fca6b6532a9375","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 14:51:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91bbae1c735f4109a0cb7680cb791222 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9","revisions":{"d88324697cbf68c108579b13ee223bbcabb244dd":{"kind":"REWORK","_number":1,"created":"2020-03-12 14:04:22.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/88/712688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/88/712688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/712688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/712688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/712688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/712688/1"}}},"commit":{"parents":[{"commit":"7d38a01be3a8125bc8a830c70ab412c571b9fe3f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7d38a01be3a8125bc8a830c70ab412c571b9fe3f"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2020-03-12 14:03:41.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2020-03-12 14:03:41.000000000","tz":0},"subject":"Refactor memcached_servers","message":"Refactor memcached_servers\n\nThis patch aims to add a prefix for memcached_server\non each role to give the ability for deployers to\noverride the location of memcached cluster. I.e users\nwants to create a single memcached cluster with k8s\nfor each service.\n\nChange-Id: I19b74c3bc5119953256d3d8f2a98cb5f23787755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d88324697cbf68c108579b13ee223bbcabb244dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d88324697cbf68c108579b13ee223bbcabb244dd"}]},"branch":"refs/heads/master"},"0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9":{"kind":"REWORK","_number":2,"created":"2020-03-13 22:15:16.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/88/712688/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/88/712688/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/712688/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/712688/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/712688/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/88/712688/2"}}},"commit":{"parents":[{"commit":"7d38a01be3a8125bc8a830c70ab412c571b9fe3f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7d38a01be3a8125bc8a830c70ab412c571b9fe3f"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2020-03-12 14:03:41.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2020-03-13 22:14:53.000000000","tz":0},"subject":"Refactor memcached_servers","message":"Refactor memcached_servers\n\nThis patch aims to add a prefix for memcached_server\non each role to give the ability for deployers to\noverride the location of memcached cluster. I.e users\nwants to create a single memcached cluster with k8s\nfor each service.\n\nWe also add pymemcache based on [1]\n\n[1] https://review.opendev.org/711429\n\nChange-Id: I19b74c3bc5119953256d3d8f2a98cb5f23787755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
