)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~718473","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ib0f87d06eb4c946e6e7b10f20fb2ed176c553713","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa-collections","hashtags":[],"change_id":"Ib0f87d06eb4c946e6e7b10f20fb2ed176c553713","subject":"Use ansible openstack collection","status":"ABANDONED","created":"2020-04-08 14:29:20.000000000","updated":"2020-06-09 06:12:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c93edadd9e25a79d4a5490c26b0938196944a8c","_number":718473,"virtual_id_number":718473,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-20 15:03:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-06-08 08:17:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 15:03:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c651ca2236151bed8fc730cdabfba02d69836816","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-08 14:29:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3a2dbeb733b09b54eb4aa26725b9573c3ec9173","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-08 16:12:24.000000000","message":"Topic set to osa-collections","accounts_in_message":[],"_revision_number":1},{"id":"bab285a223baa50a5eb4f4250d183b6da9381222","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 16:19:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/84d59f6a730743b39190302a2b34555d : SUCCESS in 9m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/85355f731d724031aa26745ff2efb45b : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/83711e72f3e54d02bb9782951ac6d501 : SUCCESS in 1h 26m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e17315cd457642fa92064e79c96c70b3 : SUCCESS in 1h 31m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cecb5e6d7cbf4fbf9a373883fa7b7520 : SUCCESS in 1h 42m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3231d6d4a2204231a4eae451a3c77866 : FAILURE in 1h 32m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be1e34efd56e40b0810abb157f9ff5bd : FAILURE in 1h 08m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/31aee85a6eee41e1a44f5c3a3b9c2bfa : SUCCESS in 1h 08m 39s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/92f392ec5c3e4148af4599f16231610c : FAILURE in 35m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0d233ee1ebf45dabf26829b688a4d2e : SUCCESS in 55m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e74c0be98b9496db6e75b9d96b5587f : SUCCESS in 6m 36s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/948d34c55b82444a9767f8105f77f4c6 : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e74141c76e6f01aa540c9a34df8c443d51785f8d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-20 12:39:05.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8968bb2cd2f24d104dabdfad2dc632c5f649114b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-20 12:39:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5ffe3a98a0c7c8008e3a20133fde403e7d041a9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 15:03:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f201f7426c2d4cff80c002a71281e8a1 : SUCCESS in 9m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/84c5da42b349418b9c80b6752e34d74c : SUCCESS in 1h 07m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2c29fd84a77b496fa9130f6d963105a5 : SUCCESS in 1h 33m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2fff452d5154f319600c9c5496c6047 : SUCCESS in 1h 13m 11s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9c832a68a5e04496b40e77a7adfdb061 : SUCCESS in 2h 20m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2bb0dedf1d54372b677739de4ab20d1 : SUCCESS in 1h 51m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7446d103561e4e39a5d0a0205eda8185 : SUCCESS in 1h 05m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed1a7d1e6f434d0d8abc5557be0c7907 : SUCCESS in 1h 02m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e330a1fc2a14a69acdeaba3fcc6a15b : SUCCESS in 10m 13s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/fb4b6d5fc8bc46ac9a120f8612d3bf4c : SUCCESS in 41m 55s","accounts_in_message":[],"_revision_number":3},{"id":"437acfd1ab9f4504935107d6a3fb77e7bbdcc9ba","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 08:17:39.000000000","message":"Patch Set 3: Workflow-1\n\nThis is just a test patch","accounts_in_message":[],"_revision_number":3},{"id":"2c93edadd9e25a79d4a5490c26b0938196944a8c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-09 06:12:34.000000000","message":"Abandoned\n\nProposal bot has created this across all roles","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"77250117347e43345d74dfd5003002c57e5a9eca","revisions":{"f34c80d3ea5cbea1bba6341499232097e42f7176":{"kind":"REWORK","_number":1,"created":"2020-04-08 14:29:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/73/718473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/73/718473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/1"}}},"commit":{"parents":[{"commit":"0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-08 14:10:49.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-08 14:29:08.000000000","tz":60},"subject":"[DNM] test ansible collections","message":"[DNM] test ansible collections\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: Ib0f87d06eb4c946e6e7b10f20fb2ed176c553713\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f34c80d3ea5cbea1bba6341499232097e42f7176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f34c80d3ea5cbea1bba6341499232097e42f7176"}]},"branch":"refs/heads/master"},"151f67a6f964a13cedebc36873ce8ac99001864e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-20 12:39:05.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/73/718473/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/73/718473/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/2"}}},"commit":{"parents":[{"commit":"0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-08 14:10:49.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-20 12:39:01.000000000","tz":0},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: Ib0f87d06eb4c946e6e7b10f20fb2ed176c553713\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/151f67a6f964a13cedebc36873ce8ac99001864e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/151f67a6f964a13cedebc36873ce8ac99001864e"}]},"branch":"refs/heads/master"},"77250117347e43345d74dfd5003002c57e5a9eca":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-20 12:39:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/73/718473/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/73/718473/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/73/718473/3"}}},"commit":{"parents":[{"commit":"14834df845d985a34ed2e41eee4eea8c4f34bbf3","subject":"Merge \"[ussuri][goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/14834df845d985a34ed2e41eee4eea8c4f34bbf3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-08 14:10:49.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-20 12:39:09.000000000","tz":0},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: Ib0f87d06eb4c946e6e7b10f20fb2ed176c553713\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/77250117347e43345d74dfd5003002c57e5a9eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/77250117347e43345d74dfd5003002c57e5a9eca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
