)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~718530","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Icf572c656c24b646110ce3fd90727205c22eff15","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"Icf572c656c24b646110ce3fd90727205c22eff15","subject":"Make possible to run glance-api not with uwsgi","status":"MERGED","created":"2020-04-08 18:00:08.000000000","updated":"2020-07-22 13:50:32.000000000","submitted":"2020-04-27 15:29:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718530-1588001350302-81340386","meta_rev_id":"42e0ac80601a63c093f8e73492a9942a62783c9b","_number":718530,"virtual_id_number":718530,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 15:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":2,"date":"2020-04-27 10:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2020-04-27 10:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":1,"date":"2020-04-27 10:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 11:47:59.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-04-16 23:25:04.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-04-27 15:29:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b386f964c0605ac98c388102d930c8271e72ac99","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-08 18:00:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e56705f724677dfe925cdebef428dcd34c5547d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-08 18:43:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f3319a7d65b729a2727a0c18840eb588c501ddb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 21:39:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d3f582ebda5644c4bc285c8008f6a095 : SUCCESS in 9m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e63d297412064819a41aa29b15acab8a : SUCCESS in 1h 07m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/30ec4c8f5b2144aea4eb99932d0fa619 : SUCCESS in 1h 55m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6c49ca8ce29a492c9da8203ab70cb388 : RETRY_LIMIT in 9m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e46ca2560bd54609949c02435af81f72 : SUCCESS in 1h 36m 51s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b8962e583d9d4fda9426d13e8a43bdb0 : SUCCESS in 2h 46m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7de59762c1964c038cfa0baf15735d6c : SUCCESS in 2h 04m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/13405f08cd354424a5fd6cc7d8ed7e8d : SUCCESS in 1h 24m 01s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/51baa4e5332f4a259e39750a11067bc8 : RETRY_LIMIT in 9m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2388dab6dac4fe6b80d71a890296840 : SUCCESS in 1h 00m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b56c98dc76894633b83b738db0247b2d : SUCCESS in 7m 07s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/2c770e5b52c54c7084ad73ef661dc5cd : SUCCESS in 37m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a3c73c17b118ee723f8aff8e650be7c4e6cb39c1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-09 11:40:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"332a7208e7501aef6c9edaab6466b952df79313e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:11:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b84c88861b214c9484ecddf281b17a45 : SUCCESS in 10m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3f946ac831e64a7bb5453ed73ad3a295 : SUCCESS in 1h 09m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8ec587a0ea25452783a6b776a92a583a : SUCCESS in 1h 27m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9c71a44224ad43efbc67758618a8672b : SUCCESS in 1h 29m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f02282f08264ebb8724b2006c55eaca : SUCCESS in 1h 40m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/61dc1acc807845a1af586ecbdde2765b : SUCCESS in 2h 25m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6224ff5a68f6454ba810ff7065aa31ea : SUCCESS in 1h 54m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8ad9317c4d3a4d79b405c9c278eb8b66 : SUCCESS in 1h 22m 18s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/baf3fb8e10634024b27115327c1429f8 : FAILURE in 34m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77f0d3e4de3140d3840b28a04d6e9bf5 : SUCCESS in 58m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41085a66f1d4e53b1749adfec37bb4c : SUCCESS in 8m 20s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/6ed9c2963af34ce9a2a2975d2666ee4b : SUCCESS in 33m 47s","accounts_in_message":[],"_revision_number":2},{"id":"26ec300012850facd5226063fa0e55632fe18b78","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-04-13 18:05:02.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"934bf74087b43d1c4119fd1c6c56e91ed6d8a9d3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-13 20:21:24.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a6943e949710ab0af81b53fe6a3436a12b83df72","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-13 20:23:24.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"149ba91c0471a4bea25154474cb3605df3781a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 22:56:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8b66d11bc2ba478395564bd7c0d6961f : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/320733c3c30f47c98dd001711bfd8528 : SUCCESS in 56m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8e1b2464c9a64b02a295a3f9120ab464 : SUCCESS in 1h 21m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/98c92b6e6aa74677adcb61f8516c728e : SUCCESS in 1h 25m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c386c6567ea413abb245813a6fb8e02 : SUCCESS in 1h 12m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5bdea42e14e848d4bd22ed82822a486c : SUCCESS in 2h 30m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d11c6663317d401997d0033a36e51064 : SUCCESS in 1h 56m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/359c150bb4f84b3182876fb821c40ac5 : SUCCESS in 1h 08m 25s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/218f51446d4d40969fe4aa29aa3324f8 : FAILURE in 35m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/932455bcf6844e6b909025163f7a537e : SUCCESS in 58m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2199e672ab534ebb8116b514f90ac810 : SUCCESS in 4m 38s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/7300fa2647734a06a615e9c0499c4e71 : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":3},{"id":"5651826db2609849c907573fdce795ad385a32c6","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-04-14 11:47:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"18b896c33f29a44a47eba88abcf02c79a826b1cb","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-16 23:25:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"70529b7e2e80176eed5e525fe0acce9ae3bf82f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 23:25:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6fd32c23a4901a313f0e1e05b99d78cf94319fbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 02:20:12.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1335c579d287457995036b694bd05340 : SUCCESS in 10m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/903838095fa8461583800700f596b503 : SUCCESS in 1h 07m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/496d0c45a13c414488b810b3023e655e : SUCCESS in 1h 44m 20s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/769ca16af09e4351adb5675d372bb6e9 : SUCCESS in 1h 25m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa5bb6181be04f6dba26c8d591048988 : SUCCESS in 1h 17m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/24e6af3726d142df919fc278bcab185a : SUCCESS in 1h 23m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84ced80ca45b4acf8f441eda4fb3fa9e : FAILURE in 54m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817344d4250d4a19987b816d210b8700 : SUCCESS in 5m 03s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/27fad44ffddf424695156f00137c0f9b : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":3},{"id":"a09bc6369db05923088c22b90cf9c1aa88ba44bc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-22 19:14:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1bb9f24a0b763f3bc71a81d7de78f1c0fba85299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 22:11:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a7df3cba684b4d6bb462fe49eb75411d : SUCCESS in 6m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3e4376fab2304cd397dbcb7438754e4f : SUCCESS in 1h 05m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d1f4a0457c71467a959b243232739044 : SUCCESS in 1h 39m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/697d504e8c07434192ee08f177b721f5 : SUCCESS in 1h 21m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc25c9a7303a4e97b1f48c5591a592f8 : FAILURE in 56m 04s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9863eaae13034c6fac76581fa1895624 : SUCCESS in 2h 49m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b1b3d9e83f748c087deaa996e69b80d : SUCCESS in 2h 03m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6543e13e229e44a688f6c7e32a9f94c9 : SUCCESS in 1h 15m 56s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a73acd13ebab40a5b67e49915c2d993d : FAILURE in 37m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c335caab25249108049f66668186b2e : SUCCESS in 56m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bad17dbaa654935976e376ee4727481 : SUCCESS in 5m 41s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/b1cb6b66395543969adb65b26a8a53cf : SUCCESS in 39m 37s","accounts_in_message":[],"_revision_number":3},{"id":"04e2d29c860943d12967108e9c601beee120716a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-27 10:54:15.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"aacc2d20df285ec388cea0941caa2f5e4517fc04","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-27 10:54:18.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"69bbb93011f1ead5c9dbf69886be3432ea484e70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 13:28:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7fafe400f2a543a090b15860193b55b3 : SUCCESS in 7m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4a199af11d6f4122935abc537b1c19b1 : SUCCESS in 1h 05m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b2c14938ee134f078dcd78b8c380b74a : SUCCESS in 1h 32m 59s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/313b6056e64b49a0bdbd8a54da723d05 : SUCCESS in 1h 22m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff68e9eb4c2c470d9263d746dec46843 : SUCCESS in 1h 15m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ce19d84cb4104d3ca3df0d990845b41c : SUCCESS in 2h 28m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d064b05ba9724cf5b526a04ba5fd75c9 : SUCCESS in 2h 00m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cc17e880e9a2487c9c51185acbf99c54 : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4516c5319fca40f399a2349e251e615f : FAILURE in 35m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da6560c8c61f4953946f67b9fda6e4db : SUCCESS in 55m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54e6105fd4b4381ba829cd90df640f9 : SUCCESS in 5m 02s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/6eefc550647f482c96483361f881af1e : SUCCESS in 34m 46s","accounts_in_message":[],"_revision_number":4},{"id":"2e5db2e8b37b7faf0d35931b0c7f4d015dd52e58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 13:31:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7f5986d2401b148b5635fb7bf4f6338e98686b7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:29:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6071eab65a744e8cbd9f1343786f248f : SUCCESS in 9m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5db7676ca843472ba5631f649dd38ef1 : SUCCESS in 1h 26m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8b97829a26b243b48b13fc406239352a : SUCCESS in 1h 47m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ca6571eeb317480e9552aad4a5962004 : SUCCESS in 1h 38m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/473208c7feb8489ba8884772a5fd5419 : SUCCESS in 1h 20m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d4fbe7252365432081473375f9e6c921 : SUCCESS in 1h 15m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e8bb327a9ff4a308934a38a3be705d1 : SUCCESS in 1h 08m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df71736c54244c08849e45bf311332da : SUCCESS in 7m 40s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/6d2f3fadeca648888ee4b4238da0c02d : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":4},{"id":"2a90ac9c7ce1a772c5d8c746457d4222a6eaaead","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:29:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"661fe9b910c8f5d9beda53800fbd132eeb93d4bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:32:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efbb6c64a24349d5b82b1a33926aa3f6 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":4},{"id":"42e0ac80601a63c093f8e73492a9942a62783c9b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-22 13:50:32.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 68f9ccab1cad0ac3422b7294598e09a4d05a2edf","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fbe7177e7d7cd09ae602941bfb1be86a6a14ae99","revisions":{"73ab968afafa05169f914273c7ef526cdd28ee09":{"kind":"REWORK","_number":1,"created":"2020-04-08 18:00:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/718530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/30/718530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/1"}}},"commit":{"parents":[{"commit":"0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:59:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:59:57.000000000","tz":180},"subject":"Make possible to run glance-api not with uwsgi","message":"Make possible to run glance-api not with uwsgi\n\nWhile running as uwsgi glance has malfunctioning interoperable image\nimport feature. So we add new variable `glance_use_uwsgi` based on which\nglance will be either started via uwsgi or as a regular service.\nAlso once glance_use_uwsgi is true, enable_image_import will be disabled\n\nChange-Id: Icf572c656c24b646110ce3fd90727205c22eff15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/73ab968afafa05169f914273c7ef526cdd28ee09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/73ab968afafa05169f914273c7ef526cdd28ee09"}]},"branch":"refs/heads/master"},"a6134f7ebce21985212ed6eebb5e5b6639d0611d":{"kind":"REWORK","_number":2,"created":"2020-04-08 18:43:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/718530/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/30/718530/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/2"}}},"commit":{"parents":[{"commit":"0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:59:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-08 18:43:49.000000000","tz":0},"subject":"Make possible to run glance-api not with uwsgi","message":"Make possible to run glance-api not with uwsgi\n\nWhile running as uwsgi glance has malfunctioning interoperable image\nimport feature. So we add new variable `glance_use_uwsgi` based on which\nglance will be either started via uwsgi or as a regular service.\nAlso once glance_use_uwsgi is true, enable_image_import will be disabled\n\nChange-Id: Icf572c656c24b646110ce3fd90727205c22eff15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a6134f7ebce21985212ed6eebb5e5b6639d0611d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a6134f7ebce21985212ed6eebb5e5b6639d0611d"}]},"branch":"refs/heads/master"},"f070b05596c3e752497b47d6b1f053721e3fe344":{"kind":"REWORK","_number":3,"created":"2020-04-13 20:23:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/718530/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/30/718530/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/3"}}},"commit":{"parents":[{"commit":"0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0b9bcc3e5a4fde0b44f50d63c3b0a1bd3f1672d9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:59:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-13 20:23:21.000000000","tz":0},"subject":"Make possible to run glance-api not with uwsgi","message":"Make possible to run glance-api not with uwsgi\n\nWhile running as uwsgi glance has malfunctioning interoperable image\nimport feature. So we add new variable `glance_use_uwsgi` based on which\nglance will be either started via uwsgi or as a regular service.\nAlso once glance_use_uwsgi is true, enable_image_import will be disabled\n\nChange-Id: Icf572c656c24b646110ce3fd90727205c22eff15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f070b05596c3e752497b47d6b1f053721e3fe344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f070b05596c3e752497b47d6b1f053721e3fe344"}]},"branch":"refs/heads/master"},"fbe7177e7d7cd09ae602941bfb1be86a6a14ae99":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-04-27 10:54:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/718530/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/30/718530/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/718530/4"}}},"commit":{"parents":[{"commit":"6495419a9f6b617428e6dd25eb3102309fdcd146","subject":"Fix glance stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6495419a9f6b617428e6dd25eb3102309fdcd146"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:59:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-27 10:54:15.000000000","tz":0},"subject":"Make possible to run glance-api not with uwsgi","message":"Make possible to run glance-api not with uwsgi\n\nWhile running as uwsgi glance has malfunctioning interoperable image\nimport feature. So we add new variable `glance_use_uwsgi` based on which\nglance will be either started via uwsgi or as a regular service.\nAlso once glance_use_uwsgi is true, enable_image_import will be disabled\n\nChange-Id: Icf572c656c24b646110ce3fd90727205c22eff15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/fbe7177e7d7cd09ae602941bfb1be86a6a14ae99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/fbe7177e7d7cd09ae602941bfb1be86a6a14ae99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
