)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~732262","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fussuri~I6fc3a31f91302e9be3369d6c4cb1e544acf6297b","project":"openstack/openstack-ansible-os_glance","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I6fc3a31f91302e9be3369d6c4cb1e544acf6297b","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-06-01 13:13:48.000000000","updated":"2020-06-17 23:24:09.000000000","submitted":"2020-06-17 23:21:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732262-1592436077445-87c47c09","meta_rev_id":"546c62280a3630ad081f65c45ac89f958a8ff8de","_number":732262,"virtual_id_number":732262,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 23:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-17 22:28:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-17 22:28:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 22:28:17.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-17 23:21:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d89630aff7e054e20b26f5391d25d9da56c6bc92","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-06-01 13:13:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b68498d0986a57a71cff21c3c720e57ff41aee99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 13:54:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c0df2a4ba74443f989f5bd5bbcd5bbaa : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76a921017d434be49182cb7c9e41232d : SUCCESS in 13m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/433fbbe6aa224f17ae89c487b8350702 : SUCCESS in 5m 03s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/fdd3e977214f4e52978231a208edaaec : FAILURE in 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cdf6440e3ac0fc1b2f1f6f8fdc654980d8e1fcc6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 19:49:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e4789eb69ea5112b0eb7dce15eb9c4e4a8a31faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 21:20:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f18379bf629840e187add0e5be967446 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93644be0e9ae418baa09e4f58d46fa33 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bad7c76446a41d6b6b4227e98c2f810 : SUCCESS in 8m 10s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/134bf71de96949ecaff6f1e54db4d291 : SUCCESS in 46m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0ccc7349b74808803e4a22febf8239e95bbf8663","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 22:28:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2aa45594302df9f00735006a214fe207c4284aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 22:31:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5a8ec1af77a1cae38fe1383c21e12bb5bac9d98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:21:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d7ebe84b25dc4897b7aea71bf61f2fa1 : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc83673d21674afabc81a6a72732797d : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/158462284fce416a864469fb707241f8 : SUCCESS in 6m 57s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/69cddb1f93004bee8507237664d8c971 : SUCCESS in 37m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6db90e04d82e2bd93386d8a52b764170880b0c59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:21:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7e959771512d430af6011e02d6103ca137e8bb29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:24:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9fb0af5ad0947a4928d9ac589e7820d : SUCCESS in 2m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63084094d97c49f688e297a64a8f9cf8 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b213a52b45869e7d7d7cd327bd8551902201e45","revisions":{"6b213a52b45869e7d7d7cd327bd8551902201e45":{"kind":"REWORK","_number":1,"created":"2020-06-01 13:13:48.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/62/732262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/62/732262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/732262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/732262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/732262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/732262/1"}}},"commit":{"parents":[{"commit":"e755177bc1515ce04338bf9ad24781b74c1bc112","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e755177bc1515ce04338bf9ad24781b74c1bc112"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-01 13:13:46.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-01 13:13:46.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I6fc3a31f91302e9be3369d6c4cb1e544acf6297b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6b213a52b45869e7d7d7cd327bd8551902201e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6b213a52b45869e7d7d7cd327bd8551902201e45"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
