)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~735926","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I33297d21fbd8618e167e1f332ec2c838c5086b0e","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"I33297d21fbd8618e167e1f332ec2c838c5086b0e","subject":"Add Centos-8 support","status":"MERGED","created":"2020-06-16 14:38:34.000000000","updated":"2020-07-09 10:30:52.000000000","submitted":"2020-07-08 13:40:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735926-1594215628240-f3086ce0","meta_rev_id":"b6e55e77dccac4eb92642200a1f4472a665288ea","_number":735926,"virtual_id_number":735926,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 13:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-07-07 04:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2020-07-07 04:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2020-07-07 04:30:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 16:30:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-01 13:33:16.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-07-05 17:43:28.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-07-08 13:40:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd18a7b6b66c8243953e4cf0b7ab6ca124af24dd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-16 14:38:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db89378079e7952baa2ee0b3525037770e18a2e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 16:37:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/57b2c67298c1476984c66b789e2380b0 : SUCCESS in 11m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f189851ae8e6400f8919037312f142b6 : SUCCESS in 1h 14m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fd786515aea543df8b00e603ca2d0490 : SUCCESS in 1h 39m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ffd9108d64446d99ff2fbd6d2489a32 : SUCCESS in 1h 25m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bacd2546771d412ca17afcf1e08b9e7e : FAILURE in 1h 34m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09ba1ba23c3149a2b7ca90dfa75e54e3 : SUCCESS in 1h 39m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/71df1f826bc349dfb851ce8de25ba7a6 : SUCCESS in 1h 19m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db35f42493d042a29d11a0d034f4fb77 : SUCCESS in 1h 13m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b95ff6a390f490f91b9ecd2d3a57dd1 : SUCCESS in 10m 49s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/ce2e09951bba47808fae9c3ef7d63b9c : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d0f631097dce6fd089acad96d8b8102b8950a234","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-17 06:58:39.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"95c0fcc3a2ed247e5bcc9ed6592ab8d2cd57bf2b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-19 19:43:47.000000000","message":"Removed Workflow-1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"eda45506fbe477db9425bd2ba0d78e82163f90ab","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 16:30:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8929e2aa10e011f99cba6f557f234d7fdad2e73b","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-07-01 13:33:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0505ade511904df82fecc222d25f2d0732cecb2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 13:33:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94de84e21de582124242f1aab5965c3183ab2893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 15:17:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/23ac0bd1274a40d583f6a325c73f50a8 : SUCCESS in 8m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/86b4716901d844b8b59ecdf6f67810e1 : FAILURE in 24m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/adb38a40064b4837b1327c4c39eaefe1 : SUCCESS in 1h 37m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4e1b81b00c240eca989d20389005a57 : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6997f58ef5144eb29363330471c6aeab : SUCCESS in 1h 06m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9b769bd71a90468a8c2ec1a3a7fc47b2 : SUCCESS in 1h 12m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1ac0a640e124df5b6f4516e9f35cf4f : SUCCESS in 1h 02m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57fb817c5fa24029a5955a8910273fee : SUCCESS in 8m 28s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/8af9d5335149469486bd714acdbe2dbe : SUCCESS in 41m 54s","accounts_in_message":[],"_revision_number":1},{"id":"42601890402a4076d76efa24fd4dfb4c88eb168e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-01 15:20:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6378a17edf4a3395dd433908efa325d308d39f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 17:28:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ac079eabb0f3435f9e9b7f90887e17c5 : SUCCESS in 11m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/84ddbceab992421dbf1f54cbf88f0f87 : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1e6a58ee72254940928a70c2a3cb6d99 : SUCCESS in 1h 33m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35ce69a164c04b608f4e5a020cbdc6f7 : SUCCESS in 1h 19m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98993192532e471a8532ff47188c8b35 : SUCCESS in 1h 10m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a192df2bb211428088e1bb10ffd04386 : FAILURE in 1h 48m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b21b05617e294257a4f2f7fa6cabb370 : SUCCESS in 2h 02m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eb00ac5bf8934b888f3b9f4daa707848 : SUCCESS in 1h 23m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55a339bb3c5b46899447cd1be65c0263 : SUCCESS in 1h 20m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79f077d1116d45f485f1ab401ba9ee86 : SUCCESS in 10m 23s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/7e741136ee8047e0bb166d3ce16a766e : SUCCESS in 46m 50s","accounts_in_message":[],"_revision_number":1},{"id":"94b963b9608f03334fd936642ca989f4e6c55684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 17:29:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a76b4c014e95e6bdd26cb9d6db4ad34178959fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 19:07:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/782ce176200a45a5bc4a7ed33bfc96cc : FAILURE in 8m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a274206b0f81459eb109ebe93b9fb862 : SUCCESS in 44m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/78a6ef9374764734baf3924ef1bdd92b : SUCCESS in 57m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90cc80ea94eb41d090de0c906edc6333 : SUCCESS in 1h 21m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a039f11b1fe43cfac504a0eee816d45 : SUCCESS in 1h 08m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/792d735ff24449c193b6a3aac8f186bb : SUCCESS in 1h 11m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1907eb40056a433989eaa27e7dcd2173 : SUCCESS in 1h 27m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b6c5b4cf5c43ce98c9e95a00e70f37 : SUCCESS in 7m 40s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/9849754e2f5e4c46a23584463c39a6e7 : SUCCESS in 37m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0d97d743c8c1c74992d40cf6650bb990f7046665","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-01 20:30:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"26f6b5e2c0289e080a3bd9ae095967e167b235c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 22:45:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c89ed11dd06e45da80e90f3bf676ace8 : SUCCESS in 10m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e9617c947b254fa89c5ff81aa1606e1e : SUCCESS in 1h 57m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/72a6d8d3faa24ec084d83e9b220ad0eb : SUCCESS in 1h 46m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c61f776574fe49df9f29867bb8d27623 : SUCCESS in 1h 31m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ead5cf25fb2c4b5bb57366bbf3380d02 : SUCCESS in 2h 00m 27s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dae30688e06e4c9082e21fad09eb37f4 : FAILURE in 2h 07m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d57a1a65e234f0681b060bf6a358f21 : SUCCESS in 1h 49m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7a7e665ec6804f41b3ef74e5e67918a6 : SUCCESS in 1h 28m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e6c99302e2542ceadbe9d78c6c05aa4 : SUCCESS in 58m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af4b1b0fe6374c5eadea35a6c56d3ca9 : SUCCESS in 9m 01s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/9fb41faf347c4d43a5b24f27088c5950 : SUCCESS in 36m 54s","accounts_in_message":[],"_revision_number":1},{"id":"af5e5934bda39511f92598642607bac1a615e605","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 22:45:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09a8342e6d34c50fb778dde910dfb545c608363b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 01:51:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/19a502c9bfcb479ab629b5916f153f38 : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/342ac493c5644a51b0164db7b34b7195 : SUCCESS in 1h 13m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d09adde9b9404fe29928ecc5401168fb : SUCCESS in 1h 26m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b22b5bea69ab4756b54ef2107cbdbd79 : SUCCESS in 1h 14m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c20815f9c7e49b3a90d04aeb026664c : TIMED_OUT in 3h 02m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/10db8e1bc6534066a2cc82f90f7c8e1c : SUCCESS in 1h 11m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f19a9b6038434882b7eedc02f5db6722 : SUCCESS in 1h 17m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d73b12b6fd47528794e52389113bc0 : SUCCESS in 8m 32s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/16525bad45a0436a895dc6438e33435a : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":1},{"id":"189c751fed38f9c45fa188fa055c09c3950de4c8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-02 06:22:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f6294b41b73231cf8a9bd86616449652679b3574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 08:39:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fa7fa408e63140deb177c462dc851519 : SUCCESS in 8m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fba8d26a8cfc4a51880c764f09f801f6 : SUCCESS in 1h 12m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dcaddaedddaa402cbdd5c95136e883a9 : SUCCESS in 1h 00m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0923ded3e01f45faa5c7d18af1945477 : SUCCESS in 1h 19m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d97cb156ee74bc987b705a1606052e9 : SUCCESS in 1h 14m 31s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cc27a802b4a84065977fd97808a9285a : FAILURE in 1h 53m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9189272d9914d42b17e06e92b4d0e70 : SUCCESS in 2h 11m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eeb0257979ec4519b078be23d12a232a : SUCCESS in 1h 10m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc222967e59047c490b8499bba2f823f : SUCCESS in 1h 02m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ebb3848a04b4b9f8a2e7f01fcdd62dd : SUCCESS in 9m 07s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/45a704e865204ae98437774113171cb7 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5aaedce4cb97201acb4a6fe31cd8617bb1fae812","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 08:39:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fef98e1acbdc941864dbed8b9e630fd0f9d10e82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 11:48:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b21552b7d2e64b58b6280e723048f1a6 : SUCCESS in 7m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/65d72ba6e0dc49ae898d3bc25db8d15e : SUCCESS in 1h 05m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/76c89ab8e31148f0bde323fa1184981a : SUCCESS in 1h 31m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18ea06bb9c564f78bd361e7610ad123c : SUCCESS in 1h 10m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb5fe1e102534f4e8d83e5b1d4b02a32 : SUCCESS in 1h 05m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7962a2c3a6574744a34b3a3d4ef58263 : SUCCESS in 1h 04m 09s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a7a8d91f6f8494ea1479be7d3de01cf : TIMED_OUT in 3h 03m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc0c08a2169743f78be4568973b9916e : SUCCESS in 5m 30s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/baf7e6ef28b446be9780d6700060e159 : FAILURE in 28m 54s","accounts_in_message":[],"_revision_number":1},{"id":"24898d51f1decd788d73d75dee6d48dd1e167bb4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-02 14:07:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ba1819fdbd0d122ec9b35b9708fc95fcc642bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 16:24:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2fe9fc84d37f4a37bd6b96ab86b2d360 : SUCCESS in 8m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0de157341581436680738a8cb4290c92 : SUCCESS in 1h 12m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dacbc4ea41a64130bd5211f75c9bb454 : SUCCESS in 1h 30m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d462fb854404029b046e8bd9740b12f : SUCCESS in 1h 13m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4295c2aa831543a08e40f994ce59f65b : SUCCESS in 1h 08m 35s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a01b53d0b7ce41d0a69ad6eaec32612e : FAILURE in 1h 36m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfd122b96e524ffdbfde1d8ac06e5f67 : SUCCESS in 2h 05m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8695c2e38cdc4a078680a8e0408bd911 : SUCCESS in 1h 09m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1410a0c1cc9641a78a82ba2e62b22a8c : SUCCESS in 1h 01m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004410b4b3704b07a9c01a2bafd61a21 : SUCCESS in 7m 23s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/e031a4ea37fd4138a09482e0c07180a2 : FAILURE in 27m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9840a63afdb98105267edfa0715584127c06840e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-02 17:09:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7925ebca0a91af3658a174d5a4f6f9eac63c2f85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 19:02:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dc4d91a1ebe148a2b6524c28c12826a6 : SUCCESS in 7m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/233495e6af0748df9df9b4aad2fdda04 : SUCCESS in 1h 09m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d8da62dc3c354952baa6c3099caed327 : SUCCESS in 1h 33m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13a26a616205436684af6463358e3b75 : SUCCESS in 1h 13m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36fa15f1384c49f389d8b793c62364a7 : SUCCESS in 1h 09m 34s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b13259e879dc4b9f914b225a1fd933c1 : FAILURE in 1h 43m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfaaeb1e39e043f988fdcb0bc15232e5 : SUCCESS in 1h 36m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1150be2465b148f280247ccfebc7c149 : SUCCESS in 1h 35m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23eded47948644c38869b06860149166 : SUCCESS in 1h 31m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b200348fc7ac4616ac33fec8c58fd72b : SUCCESS in 6m 32s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/d4016eb5c09e46789cf4fe466cb45fb2 : FAILURE in 38m 00s","accounts_in_message":[],"_revision_number":1},{"id":"aeca11e0364dc801d727202432abd95952fd50d0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-03 16:16:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3776e931ca847de36a2216bed923370949959035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 18:48:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/24e0b1fb7294466092ef182113885760 : SUCCESS in 9m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9793d9af96ad41608a36bcae52112950 : SUCCESS in 1h 07m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0ce00feee0e44ece89662f42107db2dc : SUCCESS in 1h 27m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/707e14ab1a9d42769b1c056d08c73203 : SUCCESS in 1h 17m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cba9c09efd0406a94d1788d59ac2b33 : SUCCESS in 1h 09m 21s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9d661d84630c49a7b82ef302387cf2b1 : SUCCESS in 2h 26m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6567cc1ef4a547b9936591f904c63cda : SUCCESS in 1h 52m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a25b3ce4cb754d1f81f469f320ff7a77 : SUCCESS in 1h 08m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b8083098b4f44b19b118deedc66cf20 : SUCCESS in 58m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a191918cfff24b02a10533506e836e19 : SUCCESS in 11m 23s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/9d0c81d42a1e4ecca977fd94934add30 : FAILURE in 35m 17s","accounts_in_message":[],"_revision_number":1},{"id":"13a0379e48bc4310b96819d9a148ec47e2aa2aac","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-04 06:55:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"81642958dc9633dd5d70c3dadfc4e7803df0df72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 09:36:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c0a246574e8e4cdba40cd0c2d6d4a13c : SUCCESS in 8m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0775bb8cd29d4567ac89512281425dfa : SUCCESS in 1h 06m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d33f794bf5d0434fa9fd8cf2bb291047 : SUCCESS in 1h 29m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdc90d5a7fcb4f47a005766d4425140b : SUCCESS in 1h 14m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42fceea9a2c445239e5d27f970bdf8e9 : SUCCESS in 1h 54m 25s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/019374f38399464d9c2cb2dbbfcff40a : SUCCESS in 2h 34m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c723d827756e46508be7b2be3a150e0e : SUCCESS in 1h 23m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/36947d0a0f5f4e39ac71d60c5a66b7c9 : SUCCESS in 45m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85e603dcc958435ca3bb2f41b40e7708 : SUCCESS in 1h 01m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0052e706955d461bba2768a23110a955 : SUCCESS in 8m 28s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/1d8b6fb6640249b89db484f48a9a94ae : FAILURE in 26m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ed55289c0f6c94578a4c82a532b472b34f2f91fc","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-05 17:43:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"674d5a49caa14a65d296b37653ad80f66e2bc9b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 19:42:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/607d903768d8467e9b0807ad22d8ea3f : SUCCESS in 6m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/36c0455dc4334e719b256a072483aa48 : SUCCESS in 45m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8150c7ad537545c8ad638ea677759274 : SUCCESS in 1h 29m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/914811ff31164519b2de8f22a9a9691f : SUCCESS in 51m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae39125a23f841eda52e9ad88c2e0597 : SUCCESS in 45m 13s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/55375c007c8f46ef991bfc46627f0334 : SUCCESS in 1h 33m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6474a74c361048a78e2181a91a563ca8 : SUCCESS in 1h 52m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/837b32e2b2cf40999d54e547271fb493 : SUCCESS in 44m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6dd42aa030741dba2af8362bb874dfb : SUCCESS in 41m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5d315e6a0ce4edda53159e0b02e45a8 : SUCCESS in 4m 45s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/484840d75c8348559668449e10aad90c : FAILURE in 22m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bca5aef1e7e46c089fcfe3ae97d997886ff6322d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-05 20:19:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"371100cc341d2e7ace038f58152896797c44b84d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 23:01:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bb5b8950718148aab3b3d56c674a8c19 : SUCCESS in 7m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/812874ab926f4187bc943c7eb76a8b70 : SUCCESS in 1h 02m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3219e52fb49e4fd08e1ba99f28f473fc : SUCCESS in 1h 23m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57069af9dbe040b88c61e366970eb71d : SUCCESS in 1h 34m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e49269b284964598a755b6f6cf973f0d : SUCCESS in 1h 01m 13s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f59986e2828b47659005424581923156 : SUCCESS in 2h 36m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f7fccd5618d474aa7df2519b2da2d26 : SUCCESS in 1h 56m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b8b49cb3bba24953af4009e0c66b2ac8 : SUCCESS in 1h 03m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b1279ac605047aead2ff68f332f9ee3 : SUCCESS in 54m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3baa09e75fa54b12960bf3d7493bd68c : SUCCESS in 6m 16s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/2cf1198ee96a450380a11aa443d00220 : FAILURE in 26m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7e0d08b2e809402818f3c03e0ea0accc78bbcd84","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-07 04:30:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"220f728507f3d08773161c443eabebbcc5856861","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-07 04:30:48.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"23b516e3298e9e76fcc681bc02b448e089d46e86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 07:39:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fe0be9b56ad64e4889668197233cfea1 : SUCCESS in 5m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/90fc63476f7a47d092e28c4df29ab5b7 : TIMED_OUT in 3h 04m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b68b93a37bcc4d9c9132a03d1b23dfa5 : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7aa0f0e5da0e460f9a3e5b4bb8ffe05a : SUCCESS in 1h 05m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a0a1159ee384c08a4a8a4c94be5108d : SUCCESS in 1h 21m 47s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7a9c7fe39237446b8e793c83cc75a7eb : SUCCESS in 2h 30m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f1c8294b611466aa35614faa0ae9b6f : SUCCESS in 1h 52m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/87b8190162524f17a1fadc6728c0776c : SUCCESS in 1h 11m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18aa5d8ace8d4e62ba9c3f2b12d95124 : SUCCESS in 54m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfdedb836f7b40c9ac7eee538725ba69 : SUCCESS in 4m 40s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/b50f8905c501435ca03275fa6ec8706c : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":2},{"id":"dd6a9977d74f94d6f0195bded673ae87c7885f76","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-07 10:15:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"209a74fce65d11597505c367ff56036d006407b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 12:36:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e029f4bb44ba4700a5f3040a59f299b5 : SUCCESS in 9m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3cf94aa9bd73419a91bd2b86626b25f9 : SUCCESS in 1h 09m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/93667fd28e0b4c91a7289d9027c7dd94 : SUCCESS in 1h 27m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2bc7653222d4a2fa0077052ada71f64 : SUCCESS in 1h 14m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/517621395649471ea31d49589177d4ac : SUCCESS in 1h 10m 49s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ccf72a86a6c545c8a9930427008b23c8 : SUCCESS in 2h 17m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30d55a685f3e4b178b780b3f37014f52 : SUCCESS in 1h 55m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a5c5f527c09f42b1a9c988a5e622e1dc : SUCCESS in 59m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0920a3377a8c452cbc5e67ef84ff1726 : SUCCESS in 1h 02m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7deaa89aca3545b09af0d5c1b9a9e23f : SUCCESS in 7m 22s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/5831e2113d374f9c9471b5d36327154e : FAILURE in 12m 37s","accounts_in_message":[],"_revision_number":2},{"id":"e0f6b3d469d95da83afabd2a056a0fdbc1b299be","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-07 15:04:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"18a3ad7c43345bf976f31745c9c1c6da3d564640","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-07 16:49:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"25cfb8028bd6b7054e0802c11f30e92513c5e6f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 17:59:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cb469cd9f932489bb188836708469672 : SUCCESS in 9m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4dc8db89fbc547ce8c625464ef3359e3 : SUCCESS in 1h 10m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1f0939c273a24ae39c551ded5f117c81 : SUCCESS in 1h 28m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1c6e1c7be8a49ee86a2fbd018c3a1b3 : SUCCESS in 1h 15m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdded3870f8748bdb4900736ebf66106 : SUCCESS in 1h 10m 27s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0a672fd9e10942ecaad7c21a71142301 : SUCCESS in 2h 45m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd73977bf91e4c47b7baea3d58fc4f68 : SUCCESS in 2h 02m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a4106edc3cbe49ed88cf2a74301764ab : SUCCESS in 1h 11m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f8978feab6042e0bd3b494b48bdceab : SUCCESS in 1h 04m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6216515d431423381493420ad50ae1a : SUCCESS in 7m 18s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/4185cdf3998947d2a797442d56c7e4fb : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":2},{"id":"b90e4b6e27100a8cdba84893acc0949148d475c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 12:03:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9c21eb3e61319fd52cedf016d500e083c06a9798","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 13:40:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c3f7b511ad5e43acb76e6fadb197492d : SUCCESS in 8m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ac0c09c1ee444719b97bfa50d008f83a : SUCCESS in 1h 12m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a6d98fb2c73d4c3a8cec0b67ff7e9219 : SUCCESS in 1h 22m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7ff4a2bafe545719801d4eda4dbc27a : SUCCESS in 59m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dc6d94ab2604975b2831e27a606da84 : SUCCESS in 49m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cfb3dbfc579c4b6c820ffb7bd7f70d11 : SUCCESS in 1h 04m 16s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03c85b62ce7a4e9d98d113fa5b42b864 : SUCCESS in 1h 07m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bfcee3b349341638085c66f1949c8e7 : SUCCESS in 6m 38s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/20befd5d759b426a8ab84687af72de4a : SUCCESS in 41m 56s","accounts_in_message":[],"_revision_number":2},{"id":"7932ba2e29aabc7ca9207c5f1e84f60d4dfdc92d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 13:40:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"55839fce463bee485b30cd60cb69ec7f29bcdb27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 13:47:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dcc556d161641f49f7199b7219be9c0 : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1c7afc3471be19b8bf1115dab21467ac48e1f290","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-09 10:30:52.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 293abff24be1f302638b7d87f69a81d752d5fafb","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5245eafd146ab8c886d432f38a161fa44880beba","revisions":{"5471dd56425d2c088fdb28486677fe99c5933d09":{"kind":"REWORK","_number":1,"created":"2020-06-16 14:38:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/26/735926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/26/735926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/26/735926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/26/735926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/26/735926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/26/735926/1"}}},"commit":{"parents":[{"commit":"2544fa50dfd748360583e3d7c2b5c16a1b4be649","subject":"Merge \"Use newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2544fa50dfd748360583e3d7c2b5c16a1b4be649"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-16 14:38:24.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-16 14:38:24.000000000","tz":180},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\nChange-Id: I33297d21fbd8618e167e1f332ec2c838c5086b0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5471dd56425d2c088fdb28486677fe99c5933d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5471dd56425d2c088fdb28486677fe99c5933d09"}]},"branch":"refs/heads/master"},"5245eafd146ab8c886d432f38a161fa44880beba":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-07 04:30:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/26/735926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/26/735926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/26/735926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/26/735926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/26/735926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/26/735926/2"}}},"commit":{"parents":[{"commit":"ed736fd783a60d583caf4727e15fe813e8d78757","subject":"Cleanup glance-registry service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ed736fd783a60d583caf4727e15fe813e8d78757"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-16 14:38:24.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-07-07 04:30:44.000000000","tz":0},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\nChange-Id: I33297d21fbd8618e167e1f332ec2c838c5086b0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5245eafd146ab8c886d432f38a161fa44880beba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5245eafd146ab8c886d432f38a161fa44880beba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
