)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~739795","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Id177ff37c9254d24d2cd2edd0a534055f0248de5","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"Id177ff37c9254d24d2cd2edd0a534055f0248de5","subject":"Remove glance-registry from glance roles","status":"ABANDONED","created":"2020-07-07 15:22:49.000000000","updated":"2021-03-16 13:30:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0ab8652b1515a89f6e50dfb9d761506c85395e15","_number":739795,"virtual_id_number":739795,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-07 17:55:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-07-07 16:00:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-07 16:00:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-07 17:55:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a214a33f5e62d5965ec7eb2db5b0523c924f274f","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-07-07 15:22:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ce0d654fa671afb52408c967696d8c2c91fc3e0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-07 16:00:53.000000000","message":"Patch Set 1: Code-Review-1\n\nduplicate of https://review.opendev.org/#/c/739656/ and missing a lot of parts","accounts_in_message":[],"_revision_number":1},{"id":"e7b4beac3d1b3972635eb122f997fc31fb2f0b16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 17:55:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ac4aafac3ae042f2b61ed941a477559a : SUCCESS in 7m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/805659bc20724158bedbccf4d08bd90c : SUCCESS in 1h 08m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/85d69a7716b94706a285a0fec719e0ca : SUCCESS in 1h 01m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a4619ea7d2f45a4bf70b3353e030afe : SUCCESS in 58m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b3192bc7f584d369e9f36d0f6356061 : SUCCESS in 1h 25m 59s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4172563c7a5443bd8add505454f1f3a8 : SUCCESS in 2h 23m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f13441c761642b5962173b50fba6d56 : SUCCESS in 2h 13m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cbcfd68943694a0c85f4501a41151204 : SUCCESS in 1h 33m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12eac5c3534a44e28ceb1a78d1843254 : FAILURE in 1h 02m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94bba173f3bb4090a6fa81b1ebfb43d6 : SUCCESS in 7m 09s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/ed99933bbbf84e60837a2a8adc544a1d : SUCCESS in 39m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0ab8652b1515a89f6e50dfb9d761506c85395e15","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-16 13:30:42.000000000","message":"Abandoned\n\nalready implemented","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc5a4bdb1e0e98666c12f2491dc333373ff0f5bd","revisions":{"fc5a4bdb1e0e98666c12f2491dc333373ff0f5bd":{"kind":"REWORK","_number":1,"created":"2020-07-07 15:22:49.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/95/739795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/95/739795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/95/739795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/95/739795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/95/739795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/95/739795/1"}}},"commit":{"parents":[{"commit":"912a8346047207a4f3c3473e3760baf4c987158e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/912a8346047207a4f3c3473e3760baf4c987158e"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-07-07 15:21:32.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-07-07 15:22:39.000000000","tz":480},"subject":"Remove glance-registry from glance roles","message":"Remove glance-registry from glance roles\n\nglance-registry service was deprecated in Queens release and has been\nremoved in Victoria [1].\n\n[1] https://review.opendev.org/#/c/738671/\n\nChange-Id: Id177ff37c9254d24d2cd2edd0a534055f0248de5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/fc5a4bdb1e0e98666c12f2491dc333373ff0f5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/fc5a4bdb1e0e98666c12f2491dc333373ff0f5bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
