)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~740232","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fussuri~I33297d21fbd8618e167e1f332ec2c838c5086b0e","project":"openstack/openstack-ansible-os_glance","branch":"stable/ussuri","topic":"osa/el8-stable/ussuri","hashtags":[],"change_id":"I33297d21fbd8618e167e1f332ec2c838c5086b0e","subject":"Add Centos-8 support","status":"MERGED","created":"2020-07-09 10:30:52.000000000","updated":"2020-07-10 15:26:06.000000000","submitted":"2020-07-10 15:23:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740232-1594394635229-a252ec55","meta_rev_id":"dfc4d90c86f41b27f087f2c47e27311dd7e1b0fd","_number":740232,"virtual_id_number":740232,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 15:23:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-07-10 13:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-09 14:55:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-07-10 13:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 14:55:34.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-10 15:23:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0970eeee25eb402ed5285b3a392d53bc6b7247c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-09 10:30:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"95c4ba667c91f6d5ba2924550d7c63999be73fd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 12:57:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/218bdcb84f0a46508b1b158ec7efd34b : SUCCESS in 10m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/faa40a4a5d57486b801ebc1fe90bcf5d : SUCCESS in 1h 06m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a6fd5bcd697443939d2f482ada14042a : SUCCESS in 1h 27m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c879546a3f934c90ae62ace937d3be1d : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5a19ff686024e42bed468d7a8d031a4 : SUCCESS in 1h 36m 56s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/78d20f234624441bbdb8b9f304ff6b16 : FAILURE in 1h 18m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91f934785e704bbb8d3c35f291e6b305 : SUCCESS in 2h 19m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/96868e95478140d4b98ba5de6ec051b8 : SUCCESS in 1h 03m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/deea87bc549e492db2777a76ffece784 : SUCCESS in 59m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e663400fe4e4da1859a4c0b0797df5b : SUCCESS in 6m 53s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/d7d6823f62704422b3ff1605b6789dee : SUCCESS in 34m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f80b21c39a8ac65949965e093ab9e25594594018","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-09 14:55:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d1c54dc5525606517701027019f82f339b24f536","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-10 13:52:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa93abb882b52979028213bba25f2c46c3e5a692","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 13:52:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4215ccba529328908182bf0c394ebc28c4a0b0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:23:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c2f15ff690f84fc7a4a58308a60ee6f9 : SUCCESS in 8m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/71b94ca0b4a247eb893d58a9c412d56f : SUCCESS in 1h 10m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/84fb15ddad234382a24da73257ea9da6 : SUCCESS in 1h 23m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4863619aaa76408cb509b8bfc13170d8 : SUCCESS in 1h 09m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbb60718a869468ea2b18264ec4d5a9a : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fcbead488ded4317aa0d5e553f770f01 : SUCCESS in 1h 05m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccbe73ca6c1c4c529e2fc7ed53699830 : SUCCESS in 57m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a04784606e4cd78328abb2b9c4f49a : SUCCESS in 5m 49s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/bf95552cf9cb478b95a1c04c116be9d3 : SUCCESS in 36m 56s","accounts_in_message":[],"_revision_number":1},{"id":"557df39eb99dbeaa12e09168ede3b489dc2a9965","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:23:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8700c501a1d7d551f8bde9eb8ba4061643f78749","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:26:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a449f308818c4e6184969b3670a33914 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"293abff24be1f302638b7d87f69a81d752d5fafb","revisions":{"293abff24be1f302638b7d87f69a81d752d5fafb":{"kind":"REWORK","_number":1,"created":"2020-07-09 10:30:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/740232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/32/740232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/740232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/740232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/740232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/740232/1"}}},"commit":{"parents":[{"commit":"b99d4c00192ba94bc0d37179a1c23f6caf7cda11","subject":"Merge \"Updated from OpenStack Ansible Tests\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b99d4c00192ba94bc0d37179a1c23f6caf7cda11"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-16 14:38:24.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-07-09 10:30:52.000000000","tz":0},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\nChange-Id: I33297d21fbd8618e167e1f332ec2c838c5086b0e\n(cherry picked from commit 5245eafd146ab8c886d432f38a161fa44880beba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/293abff24be1f302638b7d87f69a81d752d5fafb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/293abff24be1f302638b7d87f69a81d752d5fafb"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
