)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~742433","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Ftrain~Icf572c656c24b646110ce3fd90727205c22eff15","project":"openstack/openstack-ansible-os_glance","branch":"stable/train","hashtags":[],"change_id":"Icf572c656c24b646110ce3fd90727205c22eff15","subject":"Make possible to run glance-api not with uwsgi","status":"MERGED","created":"2020-07-22 13:50:32.000000000","updated":"2020-07-27 12:06:35.000000000","submitted":"2020-07-27 12:04:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742433-1595851481629-7b2e085e","meta_rev_id":"5c2b0e83bfcc4bef96d5b0cdde27f0f9e5f7a37b","_number":742433,"virtual_id_number":742433,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-27 12:04:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-07-27 09:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":1,"date":"2020-07-27 08:42:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-24 14:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-07-27 09:52:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 14:31:50.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-27 08:42:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-27 09:52:39.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-07-27 12:04:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6969618e45b3b43f446b531f1354c63718408948","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-22 13:50:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d3a3be6f678b849ea80b62ec2b146eca84ffbc87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 16:14:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/855f2792625b4fb4aca80352849ddb89 : SUCCESS in 8m 17s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/76e1db316d61483d9b87e52826028f0b : SUCCESS in 1h 25m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2fa375d944604fb29bbe9f41437cd64d : SUCCESS in 1h 17m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2cc31a37524d4175989c08455df6d9c6 : SUCCESS in 1h 28m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/544a5ce5d794427780a0f1fa2cabac47 : RETRY_LIMIT in 20m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3944afdd0ed4c719962134a13cb87da : SUCCESS in 1h 07m 21s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bf0b647b06574feaaf86f02844fe6c47 : SUCCESS in 2h 13m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/811e967398e147af99ec4aad09ce8993 : SUCCESS in 1h 17m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a1c23671d2ed4edaa720b60e74418606 : SUCCESS in 52m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3164aa1b971d402a9ad0ac4f95726686 : RETRY_LIMIT in 16m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb114fae9dfe46fc99e10d9419680226 : FAILURE in 27m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48faa7af1bdf4fda957cf4a01d485d5b : SUCCESS in 9m 09s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/058123c93b4547fd8b1f19e2c4b52119 : SUCCESS in 39m 01s","accounts_in_message":[],"_revision_number":1},{"id":"038d194ebab15ee2776824d0ed4962cacd0ad19c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-24 07:04:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7bad6b831348c941228eb5ec7eb3c741dd651364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 09:16:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/66bb0754d7b54bc8996cb2825b79e8f5 : SUCCESS in 7m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c1319f00697144b99fb6a8825493e257 : SUCCESS in 59m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b0838ed6abcd4cbc81d18b1fc37fdc38 : SUCCESS in 1h 22m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5e077b07cb454408b92b511ebe7935b1 : SUCCESS in 1h 39m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7862eb3439be477898c39e22df2d730b : RETRY_LIMIT in 21m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/188cc2a05e3c41f1924b9a2efc39fecc : SUCCESS in 1h 12m 05s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e94ce5318df045f1843ad5d00bd38bc8 : SUCCESS in 1h 58m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d73b0e096354ad18a0642fb78ff4fe4 : SUCCESS in 1h 15m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5e7e399481474136a91c24545655d8a8 : SUCCESS in 48m 22s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0f4dfcc8a9f3468d8de8b09f36177c53 : RETRY_LIMIT in 20m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12afe316f82c4e598542cc0da87e1e1f : SUCCESS in 1h 02m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f681d846d44fb59d9fc40b9ff83c5b : SUCCESS in 9m 13s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/821c7522196b469eb564fe77b98d71d5 : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":1},{"id":"66e45d71ac6500a64e6d36b43c542cac57d193e9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-24 14:31:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"197ec7c7264e5b7ab698dd00962a9b652b7feee5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-27 08:42:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"36c9da83c66693832c420e3b8c3e1b76348375c3","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-07-27 09:52:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba4a579bc778f1c227b450a422c4c11e1ab317ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 09:52:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eba73b194d8837cd8f24080719fed344c8f3cb05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 12:04:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d021217b6bc84a3d86a13c1b39353abc : SUCCESS in 9m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/01b7930d6e6e4dcf8699d527832a7bba : SUCCESS in 1h 10m 53s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/92febbb0097241f0846ef50f38fa4e9d : SUCCESS in 1h 27m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/24b6121396ce488196b58c1450637edf : SUCCESS in 1h 31m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7cda2e3763d74bee81d6c8c3e72ef0dc : RETRY_LIMIT in 17m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0fdb3f4633b4a27a2c5dc7fbe23b78e : SUCCESS in 1h 15m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1476ae86ae6248f88cc284a43d050751 : SUCCESS in 2h 04m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bd9f7e96d80b4569842365e041f625f3 : SUCCESS in 1h 16m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/490a1b3390184c8a9c706ccd6c6252d7 : SUCCESS in 55m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1214d866a1b4974ad4885637e9114a5 : SUCCESS in 6m 15s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/739bf8884e8a42399fa9f2357e098209 : SUCCESS in 40m 50s","accounts_in_message":[],"_revision_number":1},{"id":"635818bf622d7bcaa4355899806e4b606c8eb5c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 12:04:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5c2b0e83bfcc4bef96d5b0cdde27f0f9e5f7a37b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 12:06:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b03fa4f667d481f9d80b5d3f8b26254 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68f9ccab1cad0ac3422b7294598e09a4d05a2edf","revisions":{"68f9ccab1cad0ac3422b7294598e09a4d05a2edf":{"kind":"REWORK","_number":1,"created":"2020-07-22 13:50:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/33/742433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/33/742433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/33/742433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/33/742433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/33/742433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/33/742433/1"}}},"commit":{"parents":[{"commit":"80a510c2439939a9f45bbdaefa88c1953be0f2c9","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/80a510c2439939a9f45bbdaefa88c1953be0f2c9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:59:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-07-22 13:50:32.000000000","tz":0},"subject":"Make possible to run glance-api not with uwsgi","message":"Make possible to run glance-api not with uwsgi\n\nWhile running as uwsgi glance has malfunctioning interoperable image\nimport feature. So we add new variable `glance_use_uwsgi` based on which\nglance will be either started via uwsgi or as a regular service.\nAlso once glance_use_uwsgi is true, enable_image_import will be disabled\n\nChange-Id: Icf572c656c24b646110ce3fd90727205c22eff15\n(cherry picked from commit fbe7177e7d7cd09ae602941bfb1be86a6a14ae99)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/68f9ccab1cad0ac3422b7294598e09a4d05a2edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/68f9ccab1cad0ac3422b7294598e09a4d05a2edf"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
