)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~745846","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Id1dcf4e0ca5c3e4e157a4814c2732b8e91ab7cfa","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Id1dcf4e0ca5c3e4e157a4814c2732b8e91ab7cfa","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-08-12 11:32:51.000000000","updated":"2020-08-15 23:01:46.000000000","submitted":"2020-08-15 23:00:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745846-1597532441013-69ff8aaf","meta_rev_id":"0178ea9bcd09f207b403b0eee9e706eb6efd77f4","_number":745846,"virtual_id_number":745846,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-14 22:42:16.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-15 23:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-13 05:41:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-13 05:41:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 05:41:01.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-14 22:42:16.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-08-15 23:00:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d268b4705c87347755a12966613eee16d6e9f2d1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-12 11:32:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a88a96468a114a331f586c827503be916635bc9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 14:13:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/95738e2a93074cbebfbc499376cfcb94 : SUCCESS in 6m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0a5adc605eb2416bbaaa8c98f8f81c37 : SUCCESS in 1h 09m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d2d4838ed3614352baa72d5784f3c7d7 : SUCCESS in 1h 26m 49s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3c1e7d05cf6a4afe8725fcff9b700824 : SUCCESS in 1h 17m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c121a4b3ce434375bad409dd3ca0ae4f : SUCCESS in 1h 09m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e279d776d1864439939b68c11713e3e8 : SUCCESS in 1h 01m 57s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4604dd1ac1b041e89d52cb314761290c : SUCCESS in 2h 19m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c42e5ba434ae428d9ed4f49adce4cd53 : SUCCESS in 1h 46m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8511b94653824e7cb5b3e3c8b5a5bbd5 : SUCCESS in 1h 07m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1fb781c6f24d4ee880a724a2f01e124a : SUCCESS in 59m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cc46a7169454dec928d56a7057defbc : SUCCESS in 56m 14s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7c94ff258604e87b2aba6efe328f06d : SUCCESS in 47m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e333f7d15ca848a79f185577c47a08e6 : SUCCESS in 5m 05s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/d35d8b25aee042a8a320f1af2d203be8 : SUCCESS in 35m 15s","accounts_in_message":[],"_revision_number":1},{"id":"cfde8c7c9bfd80fdaeabbd56afd2637f7448c9b1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-13 05:41:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8475d1d841babe15aa00724b149e896fb9d80534","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 05:41:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06aa98cb2aa3ea1dfec8adeb61dce17ec06b8a83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 08:07:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/05ca9db3ede248f8ae3d23f0d5d4c27c : SUCCESS in 6m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/03884fc5aea24b6fa75559a906e20022 : SUCCESS in 1h 02m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/252e6f6fb40146d98300df8e8c261a99 : SUCCESS in 1h 18m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/527b4e253bc64c5aba6c8e9335f70475 : FAILURE in 26m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2732c984c2fb420bbe79239c77a41ab8 : RETRY_LIMIT in 15m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e90604834744777ad69dff2da1133d6 : RETRY_LIMIT in 10m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d1ebb71ca5834000b19aa60b2fa8fba0 : SUCCESS in 57m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/35ab18d07fe44a90a32ac640e79dbcb3 : SUCCESS in 1h 01m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97ddf584ccb14250a30e888de2b9e196 : SUCCESS in 58m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eec10536af99419cbadaea187387c5b1 : SUCCESS in 58m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e734c1a6b7d41799996c2a665852925 : SUCCESS in 5m 16s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/97275bd740554fa491e052e81efe24b8 : FAILURE in 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"dc9c10a738514c814828efe05aed28cd7e141b88","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-14 21:10:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"571505fb30fa3204e897a1728b35ddbe91ec4abc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-14 22:42:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7cfa25f5f2e527a9a0ae1a85209136826d0e83ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 23:50:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/420c7e805e364c0287724b01683b0ec2 : SUCCESS in 6m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1136f4b782bb421eba9cc260d217ffb7 : SUCCESS in 1h 06m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eb01bd7c668f46559bcb7e024f54c20d : FAILURE in 22m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/552b2aab05214c73a4ff52a91cb122b2 : SUCCESS in 53m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ebf545425244c2ca73d75c1a113f156 : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4e7714cfc4a4b04bf2f6ce57942dfa6 : SUCCESS in 56m 48s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/202436fb45434892b5e04e0ec898f654 : SUCCESS in 2h 25m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c608a70666d4002bddb608aa55d6823 : SUCCESS in 1h 16m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b9c73af9291b4a99913159fc91ed67d7 : FAILURE in 1h 03m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d5fac64cc50941d2867c9ad10727f4e3 : SUCCESS in 52m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae97d59f9ed5463eaad971a48c2442e8 : SUCCESS in 51m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1fafd231c94400090ee6d163cafdee9 : SUCCESS in 48m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14d0f3ec565f414caa1324ca082e244c : SUCCESS in 5m 19s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/bb117c53410f4c57918e817b696889a7 : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8c70603ad8acce7743c3c688a3876d38c0b46467","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-15 14:19:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"166ae59b8b82ad01ae21f4bd267039f31b730ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 17:45:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a5cb82a5a207430fa5dd86775d5feb2c : SUCCESS in 7m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/641eab53ca304485ae48ddb6fedcef39 : SUCCESS in 1h 02m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/452c23c1f6714295ad56faf6e9f3a1cf : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/764beef9293145978d725a836d4a5e67 : TIMED_OUT in 3h 18m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b420dda5e3b4b879bf1f5a6b3210d27 : SUCCESS in 1h 07m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/792d7336f5194a8cb5ddeb737fa38ba4 : SUCCESS in 1h 42m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4ac163f376dd4b318fa4a8e324fbe8c1 : SUCCESS in 1h 58m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9446bbb60eea42fa94722ab7a5c16087 : SUCCESS in 2h 21m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e7d4c62d744f48c3b89754a0e567aa70 : SUCCESS in 57m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cf9c7d589e244a5abb0c8e59bed6956b : SUCCESS in 1h 35m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c37811e566674304b5aa2d99cbb8813e : SUCCESS in 55m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/610a016b2cdb45899fd2513743767a15 : SUCCESS in 1h 46m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb99dffc6694abda37ebd1b7603b688 : SUCCESS in 4m 43s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/e27857faa12b4937bc9fdcfdf470a237 : SUCCESS in 47m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2ac7bcfadb46a5864afe98b574b2e531a2eaf4f5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-15 19:06:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4aee1387ea4c86fe32409a656c7f3a5f75fcc968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 21:30:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8ff6228052444d5fa8ac23ae156246ff : SUCCESS in 6m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f42a6b6f36e543fd9bf223ae5ada84cd : SUCCESS in 1h 09m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bc28d3894a3c40d38f0dfde0434e4ac7 : SUCCESS in 1h 15m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5e5f2c58b746499eb9b7cb2bb6fd8285 : SUCCESS in 1h 12m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d76092d15ec408c8050228f12cb4e51 : SUCCESS in 1h 00m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1774b3f32e3746388c31ca0dc3b5c069 : SUCCESS in 1h 00m 22s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5531faeb99124cedbd3eb58cacaae1b0 : SUCCESS in 2h 15m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3534d839407c4f8cb414b81372787889 : SUCCESS in 1h 46m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8008a8768ab74fa084e716c4581dbce4 : SUCCESS in 57m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/57828051793d4c4baf1ec68269cd84fc : SUCCESS in 53m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97343f1d64ec48b2a7fd15d4f3fe1058 : SUCCESS in 51m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5893aa1303664378a59119037d257140 : SUCCESS in 44m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99fd0f93fdd24af5a3e79cc1b9d664a6 : SUCCESS in 4m 50s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/db8a6bc6eec14997a52654cd3fd45971 : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ae0bd4aebf56f6f78025aea25aa292033e65c4c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 21:30:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e46fe5f3673d9eed01638f9443caca8a7aab91d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 23:00:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/67f73d2201c34746aa7efd2d4fea6bd2 : SUCCESS in 9m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7953957ffafe478fb9b587fe4e346d82 : SUCCESS in 1h 11m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c37c2ab9f8e349019c46361988698c26 : SUCCESS in 1h 22m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/66f6e841c8364e518cb2ae49a05f1e7f : SUCCESS in 1h 23m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59ffb274637a4230b7b70accf65fc251 : SUCCESS in 1h 06m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fee6653ddc9d49cdb934a6c39ae5d64b : SUCCESS in 1h 00m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bafc11adc4154e8aaf1455a02519bfc8 : SUCCESS in 58m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/960a092fba574d0e9f6c74fc1ffbf191 : SUCCESS in 59m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94ff61d8e7544591bc152a6a0e95febb : SUCCESS in 58m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c89c87e603942e68de3efae9ebc77db : SUCCESS in 48m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4eb3a294bf4f69b34b812820c50ffc : SUCCESS in 4m 45s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/5953ae0f6047454a9c5e9a1088e4d5d9 : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a1f208ce8f74d95a7ce6f754053de99a930b1c2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 23:00:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c6c20b28ea2bae463b4fa27d1fd0a809fc616a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 23:01:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e823a0162a439c865665d5fb733d73 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a489cc86fa2f48754317d6886c9552d50c99bf87","revisions":{"a489cc86fa2f48754317d6886c9552d50c99bf87":{"kind":"REWORK","_number":1,"created":"2020-08-12 11:32:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/745846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/46/745846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/745846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/745846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/745846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/745846/1"}}},"commit":{"parents":[{"commit":"d649d1a362eec9502568be219eb3c4b3fffb5814","subject":"Merge \"Add Centos-8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d649d1a362eec9502568be219eb3c4b3fffb5814"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-12 11:32:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-12 11:32:49.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Id1dcf4e0ca5c3e4e157a4814c2732b8e91ab7cfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a489cc86fa2f48754317d6886c9552d50c99bf87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a489cc86fa2f48754317d6886c9552d50c99bf87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
