)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~746291","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fussuri~I8eb92486583d99c09e42b6480ca317b9c6381b6a","project":"openstack/openstack-ansible-os_glance","branch":"stable/ussuri","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I8eb92486583d99c09e42b6480ca317b9c6381b6a","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-08-14 07:38:53.000000000","updated":"2020-08-14 11:41:02.000000000","submitted":"2020-08-14 11:39:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746291-1597405196652-ff0595dd","meta_rev_id":"533a8b1a6693dd7bc45028e7c9a3d42261ad9f68","_number":746291,"virtual_id_number":746291,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-14 11:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-14 10:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-14 10:03:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 10:03:08.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-14 11:39:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c4319b1fe6dbfcd403efc02dc2fa41f3d9d5ccb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-14 07:38:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbeb1826b031fa3ba361e55687012c49518102b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 10:01:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/977393a747c54776bc09ce11f1ae38d0 : SUCCESS in 13m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7bdb868c1d2b461ba56a8be0f42ab7d1 : SUCCESS in 1h 02m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9e6232b135a043b7b0918a22a1f279fb : SUCCESS in 1h 49m 41s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c287910bd131400f92d1c289a91726b2 : SUCCESS in 1h 17m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a94d3e7b49fc41ac8b48934576cc5704 : SUCCESS in 1h 07m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8553a2a680904454a0540c1ecbbda0a9 : SUCCESS in 1h 34m 29s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/46bede247fb449b7b1f557fc7b8bae33 : SUCCESS in 2h 08m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a35cb1a09dd40ed87435feef2921776 : SUCCESS in 2h 11m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a9daefcfcfa34d2ca9de178acca246d3 : SUCCESS in 58m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b23a90843e57409ea6b9bbb4467eb599 : SUCCESS in 56m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae1c82c887544496986945ae12e284ec : SUCCESS in 1h 26m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0884a546c5cf47e4a4b146c505240bfe : SUCCESS in 50m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c80924a4a14f9697105f4988e3624e : SUCCESS in 5m 23s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/09c8e47d81a442b2bdbd62b1d428cbdb : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fdbb4dd27d0847c72c07b4bddeeb1b5f384b3c40","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-14 10:03:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f6fae998e0655be1018ced9a8cadb322fd2b9c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 10:03:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"487ee7d6de4baa1146957a3f76fceca6b4f55fb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 11:39:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9c63791c1a6f4d53972764f8d9f48a84 : SUCCESS in 7m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/452c3773e38742c4ba6e7437b50622f3 : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/580c9f4731564419a206ba289355bc95 : SUCCESS in 1h 31m 49s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ea0dde00c3204983b5a9b26a4136c2ca : SUCCESS in 1h 27m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1941a02d28144d8a68d78da6ae573ef : SUCCESS in 1h 22m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/014808cd2a344a1989d8eaa01930a1e9 : SUCCESS in 1h 16m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/66ddaa1924c941e6aaed28c98b09c186 : SUCCESS in 1h 10m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/84c10cae11034199839f35f7b5b90cbc : SUCCESS in 1h 11m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5e8218745d74f839889d26771055519 : SUCCESS in 1h 05m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/533a0599f7f5447098c14b415f196cd7 : SUCCESS in 1h 03m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45c4a1929474d4182f55c7a662b68fb : SUCCESS in 6m 20s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/60d31326a2544260923822ddd0a16e22 : SUCCESS in 44m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d63a615ef11b4f7a9ee4a35a9ad3f57200bd1011","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 11:39:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b9fe9359236a03c2e96ef2916e1735c26d3dac89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 11:41:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e545e9578324ed681b0393d646eec36 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc207fff74073d45b872cc4ad4aba15169eb4a62","revisions":{"dc207fff74073d45b872cc4ad4aba15169eb4a62":{"kind":"REWORK","_number":1,"created":"2020-08-14 07:38:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/746291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/91/746291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/91/746291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/91/746291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/91/746291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/91/746291/1"}}},"commit":{"parents":[{"commit":"293abff24be1f302638b7d87f69a81d752d5fafb","subject":"Add Centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/293abff24be1f302638b7d87f69a81d752d5fafb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-14 07:38:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-14 07:38:51.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I8eb92486583d99c09e42b6480ca317b9c6381b6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/dc207fff74073d45b872cc4ad4aba15169eb4a62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/dc207fff74073d45b872cc4ad4aba15169eb4a62"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
